)]}'
{"id":"openstack%2Frequirements~873396","triplet_id":"openstack%2Frequirements~master~Ibc89f1b609f8bb7999c235b8808ac3f496a13c23","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibc89f1b609f8bb7999c235b8808ac3f496a13c23","subject":"update constraint for oslo.policy to new release 4.1.0","status":"MERGED","created":"2023-02-10 16:31:52.000000000","updated":"2023-02-17 02:34:50.000000000","submitted":"2023-02-17 02:34:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873396-new-release","meta_rev_id":"07c28321338935bace8184f135a1c9d0cf160ab8","_number":873396,"virtual_id_number":873396,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 02:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 00:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 00:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:10:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-17 00:56:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fb8e4540d95162cdf39dedacb42a8161ea257098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:31:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a09489eb62024c7a6f4598205119789627e88279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4d45f978b64a0e84b85616c647344e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f37267c9ba5b479ab581fac51bd2c7e3 : FAILURE in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"96c888be32bf483bb2d089a019d0bc0367f0094e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:15:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/065b6252d5704963b553acc2dad04480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7565ec916f7a495bb20e5c13a08195d6 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e2c12a21d445eca2b3794721c99d50 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a1746cb14794eb5972240f48b9c980e : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/615096b77a7e4b0bb85b1a66980f8894 : FAILURE in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8cf2440f1384bdca82d9a2360772363 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10e012abd67149ceabb5a644bae21f3a : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ed1f07b77d344934b9015d7c64419bf1 : FAILURE in 8m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/934d035336fc4c76911029e325803e67 : FAILURE in 12m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/236f48cbc621433b90d95c208ecf3e43 : SUCCESS in 7m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a76e8b56a7684b49b8ab81752c4aa4a9 : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8c0112ba9e3d4bd68c6ddacc79c5a101 : SUCCESS in 6m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e68cd775cd8e48d6bd4df47fc8c91543 : SUCCESS in 13m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/08822fd203104cf492c0cdaf0c12d780 : SUCCESS in 5m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2fc70a2cb059407584b3074bf741b89d : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/80bade3c33924924b78e783801a8b6f7 : SUCCESS in 9m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1df31e9c0987415a8cf02fd922c8af39 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c90a2bb65554337b2fc98aad5042928 : SUCCESS in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/25c51d4a5b734861a1839f6c23aaf3b5 : SUCCESS in 9m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/144e6dc3ca454b8a96f4e1586548f82d : SUCCESS in 26m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8546626353304c3b99c26daac30f9e0f : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/97a03b91d4d24675b9f29624f64a790f : SUCCESS in 8m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f1001334f51e41fab196cfe822f7ae01 : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3e2158289e7c41e98780ffb485eba1f2 : SUCCESS in 10m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7a56aae9f6294cf1a1ea498d3f036c79 : SUCCESS in 30m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e55bf69920874680b03b10614e6195ec : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d4a488ec24d14fdaaf72fae3ce556607 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6cfcefae5f074ee58ae1cfa532bad541 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb3fd9b79c7740589834c26caa54cd43 : SUCCESS in 21m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b13dfd5882794989a270c2f4831996c1 : SUCCESS in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/65798b91404243babbf081afbd55c733 : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/647f69db60ca484aaaedc691c8ffbf63 : SUCCESS in 8m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a3c4aa8d50104903b887a0474fb0827b : SUCCESS in 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be6421025e374cd9a025c68a5ed498da : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97c858de132947088dfc7d8a085d5528 : SUCCESS in 1h 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f7ed4c6e1cf69261fb9a6d7281275c9b168e4093","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e73843628f4defae2a639dc5eb57d58c59f1bf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d126b7d4754abc8d1df9fa99b9b42a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4f9d5ad003564ed38fdb8e63a1dcf141 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"57de779c024cf8dc762a0a882eff9715fe49fba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:49:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27fb939bdbe6436ca2852e04d9b94001\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5f323656da4146afaf4d634575c66d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299a77ae80e74e309441293716a47833 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b4da2a42894bbc9b3c643ba3d37679 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e195819f4914ba1b5701419cb329ad2 : FAILURE in 2m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac0b9d48b5514a8399d6e85a515bf3dc : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71a034a33b6d41db89109a67eeaa4592 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/964e595352844af4bd5a9c1f5372e408 : FAILURE in 8m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d58ae7b99a0d42d19f4f48f7a68b6258 : FAILURE in 13m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a4cbb51469674eb8a8b6eea58c9303c9 : SUCCESS in 6m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7ce6c3d2407c4124bb9be0b974623a2a : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/591e7abe1ab2461bb054c4af7722365e : SUCCESS in 5m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5ec0fc49d7f1435eb432a85ca363538e : SUCCESS in 10m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5a7b0578c9984651b76a49ee31329b4d : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/82df5562437b4343a5a9934d0c2ffa09 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c3b35c885bfd4865b599393cfcada3a8 : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f3a82af524524ab388d20469be4cc848 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9df7be070994cc4856b65cebc028b1e : SUCCESS in 19m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/26fddb643a7f4eaaa7918905f551ddde : SUCCESS in 8m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/264c965518c84c3a913d1fd49e93bed2 : SUCCESS in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eb404fbeb8234736b4adb40820981e32 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c35a9445fb39449eb1a80b933914a98d : SUCCESS in 8m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b8b4c79af4894720ab094c7706324eae : SUCCESS in 4m 48s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f8bc4f6b04284297a51288587ac06210 : SUCCESS in 8m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c63367a112f7408faf213393882d2f15 : SUCCESS in 21m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/04378f3e0e3e4460af4138fef4a1b6e0 : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2b444a0c16794b40b52fa62a36d1ae56 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89f55353d03e4a6694f4c079dbea87cc : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/050de06f66154e5e978dced6a733db12 : SUCCESS in 28m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a61667ebbf1142a89521bd5e18501da5 : SUCCESS in 6m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cddcb1df12fd40dfbed4dab0924298a4 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f565704b9579495886ecdb41d5ce9940 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0b8ef9e9c4c14fa788255dbbac6efa83 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b21b418fcd96409cb308c5d142299457 : SUCCESS in 48m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/537daab78cd5434291023ff6d12bc50a : SUCCESS in 1h 48m 02s","accounts_in_message":[],"_revision_number":1},{"id":"08856f4bb5957e01a63f0fa86c8cc41978bae640","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9540f065058480cf631e0ba4c78894d0928f0c77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:57:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18115a54a57a4e5d803993f97c10dc3a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/519e7e77879c40b39aa53f31a1506b27 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"72f6ad8a83093af4ec91da7eb31544cc132e11d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:35:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/169a8b37cc6447e59036dac593897480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee03d07e22e4099bbe1fa49fc31b577 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcdca107a6f4446ad367fa87d0ba64e : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/078b4c8c31894fe39bcbc046d3d6e2ef : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cba652e519ec4ffcb06005f7aefa6b11 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/357b8b95de22492dabfaea9c738cce3b : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a68429c1d259431d948c5e1f46cc0287 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/882878c870f342288361948c5e7c9dc8 : FAILURE in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a8087751fc7b47fe8e18e1c2105a315c : FAILURE in 12m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/97e7fc5af34049be925334413dc79299 : SUCCESS in 8m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7c2e2c8f948545d7a70e6eb4472f2864 : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bfdf990d5f274153ac5ccb11702425f2 : SUCCESS in 6m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6d518d5800904f43a2d3efbe26c14311 : SUCCESS in 11m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d5f08c2863284a4f8f72b6c441fd5220 : SUCCESS in 3m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f183493195924da4a2505563d11459b0 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/141a866fba0f400ea65ad625502f4368 : SUCCESS in 9m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e2c0831a4ef445d6901791a0d48961cb : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2f5b82d9a8b447d8714e19d45a42f5b : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e9d7af494b1e47d6aa26bc238315655c : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ccfaa353530841b28eceacd7702cf6e2 : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e042a2f679f94366b499313e266d62a2 : SUCCESS in 3m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6c08ab6040564ec8a8dfd29ee0a488fc : SUCCESS in 8m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d005873644ea4f34a06f9162525d4086 : SUCCESS in 5m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cd3844b0577d4e318d9cdd9a557b079e : SUCCESS in 8m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e652aa9146fe4009a3b09eb45a5b80d6 : SUCCESS in 29m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f1e3a5e4d784dc6aacc628571dc0574 : SUCCESS in 16m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/63c953ec20904956b67b5a7184e43fb4 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b791a7dbef4401f8e13637953d8c7fc : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e16a5d4fef64eccba9c4470aef26ebb : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6f7a2835f0584b08bb83b6b5a13b5828 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/55635f9d51ec41809d84f4741f25f2d6 : SUCCESS in 8m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39b09c1c7d1e464eb31094f82e389a82 : SUCCESS in 7m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/710437998a034061b07380392f3c9948 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5fcffdcb01149b0936d2621d047674d : SUCCESS in 48m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66d60671143e498ab75e7b685cc38b34 : SUCCESS in 1h 33m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5f0b33c788cc749687b168ddb0887b0dc4a42e27","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:10:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5adf6685213b3abd98b2d967b874f75ac3618e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:23:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d0658aab5f453ea710abd2de591053\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/81a16a7c75e341ef904881a2b05fd2be : FAILURE in 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0bae07df885b749ba8536e1eca9baf99d4fec88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:48:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80aa58cc77464222a4ab8dbf17823f57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ff1820d8fc491d8fd61e4234878279 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f38a636b884f059af113407d81189b : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2ebe95e059e42518f40bbbdbc7284df : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4926ab598d6d4ef09612eceee8be04e2 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9095ac5b209b4e98b6f745705a89ca79 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/859362367871419e9a7f3e332c2df174 : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb7dd1d44c8c4bf597c253f2e8001c74 : SUCCESS in 10m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05408cbf6fa24f7ba02fe2ffe112d003 : SUCCESS in 11m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/15f28cdd09914be99455191564624de4 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/248dc6b8618244039907bde8cd20eff6 : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/220ec36ff51f4690a6babc9423ea2cbf : SUCCESS in 6m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/26a44d6c51394c5e9dc81f115c277911 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0a79d3f26a8d4ff5b941e6f11660dfb0 : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ff0038a54aac4c088c9c658cc8353b7e : SUCCESS in 4m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/776cdcee0b5949c0b2f503134d49bdf7 : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/587896c1949d4aa2b8a2367ba1fd59dc : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c71b16116caa4fb7ba60c108e33d63ac : SUCCESS in 8m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/73538cfed4484c0f80304a323329a2e3 : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/99a8099d8dbb47c6b120f3939d606796 : SUCCESS in 11m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/15e830bf339a446d87f2fca1044397d0 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2ec910a6effc4ebcb5b4fe7222dd138b : FAILURE in 18m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/282b806df00e486c865e5e1727d4aa69 : SUCCESS in 3m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fe9947df22084ef5b72c0e4512b8ecf7 : SUCCESS in 10m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/70bf55929d92491cb50e84d5257ab7c6 : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a7ee09cd01c843169bedcc422568b01b : SUCCESS in 12m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/29f5ded98a5d49f693a9d26eaf196f04 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39146881b99d4225b506b56c4c2f7189 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99dc2eb5e5974647b0b950efec877313 : SUCCESS in 29m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6ce7bb7f5863441da81f8b722d5b9a30 : SUCCESS in 6m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8a72373872d541c1a4a4039abfe41482 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21ed22c27564497bbadb9f70613185f6 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/160b1f2fa7414ee4812f46c5dd2235fa : SUCCESS in 9m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/124c01b534f2443da53bb546a95ceb86 : SUCCESS in 43m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06e334083b744ac78fa4b222fd7084b7 : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8afa9eb221097a2c28e8f84f6043c3668aee7810","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0a4bf42e1340b61aaa6fc56b2b504b42f21292fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:30:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e8836a3815b40c6a703281ffc84a1d5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/52ceb05bc4f548d0a36cbaf45924ab9a : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3e5a1da324aa053b54cdb0861e741f0191cd0972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:48:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/656b604236024410b0869e455b0c195a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d033e68ed7e4cb7ab08e4c99a68ab71 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd549ea7f3640e69a379a2b44c5abd8 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe2608ed5e854883bb00e43bdf00eb58 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cfb09f4be114acf9d4b9aefde93a741 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87c89dfca3124318ade62302e4384a7a : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/910903e47519459fb828d39df04e60ec : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4e13796d7d54954a8a2a62b1b3a6e92 : SUCCESS in 8m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/efd9872a773241078e36c0ccecf5afbd : SUCCESS in 14m 19s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c52497bac5ff4b1abf8033ae3983419e : SUCCESS in 6m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c351062e0eeb4c189a2c8716c0dec0a3 : SUCCESS in 4m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c4f01c98523543f082457f89d9f3b9fc : SUCCESS in 6m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9529ba8cdef94c13a139bb526bce8c81 : SUCCESS in 10m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff1257115a81498dbf502e83a256909b : SUCCESS in 4m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8da78fde84694bd0bafa89136a8c34a2 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ea24436dc4bc4fd7bc613e6e2cfc2358 : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/02d118c1d40d42fd8e76b8e91c24b505 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bee0aa1200d44d61a4f5cabd6531fa76 : SUCCESS in 7m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6d837b4f2a1b4d45a4386b7833d362fd : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5f8666d994524125ab60adc7a389163a : SUCCESS in 15m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fed0d43d230e47869223a61e8ef3f94e : SUCCESS in 4m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9b2d49e06ac0463fb2e291a82cf7cdca : SUCCESS in 7m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/93dbc3616e7948648275ff57eb454696 : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2d7b0edf163c495aa73fa5425b31f602 : SUCCESS in 10m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/500fcd261708405ea78f7422601c66b2 : SUCCESS in 23m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0d4903fa46e847ebafc360e9277cde4e : SUCCESS in 12m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d0750262dd8947b2afbe9559b9399213 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8c1c89b7b4e4970a06a4b60d5ea03fd : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/411df704a35e4e598c2f6ebe02c280d1 : FAILURE in 29m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ab31d77acda24360990cba9cd7b205c1 : SUCCESS in 7m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6721391a36e644b796f057556a2202ef : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08ae077e5d5642ba97496f54da585830 : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/592e7fdd4ec141028d2f965fc8b5f28e : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65d616e1321402695d4da2212ad712c : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f545359a4e5b439b9dbe3f5ca24fb4dc : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9d0e00cef40b8ef365a313c46d220adece9ac188","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aec84a3af88cded22dc35db6d13090fa6471265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:33:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58deda3381cb4ebf84fab04ab9e946df\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d4e5bba0a594c94b971c4d00d406b65 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"30dd2275b77fa858f6af3698a752f2c79fa4e387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:52:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144d80e0770b4e1da2dab7dee645f9d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98868d96e23c4d83b1e71dc3ae1dc90f : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3daa57c47784f1ebe80027fe524e461 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb8b1f5028c74b2082c54cf26fc355e8 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f969c825bf84e3983e49d0bce94f4a3 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0a8f43598a24490ba9c6035433671c3 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b455c1d02514eb8904376a1d9da7952 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/799000886e204f77be9cd679b61611b1 : SUCCESS in 7m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c228de1c69df4cbaaa53267abe46706f : SUCCESS in 12m 36s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7d222b7c3b8f432c951e97f1d9db388d : SUCCESS in 11m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/564de56b7ef84544bf61d5d67048a154 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8e8881d64cbc44daa28055e3945fcb23 : SUCCESS in 9m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/688684c88afa4100b2a11d126f53b559 : SUCCESS in 13m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9dd71a1d216443f08caf35f8911bb4ae : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8e42e5c9c0314ba79dac16944f6c203f : SUCCESS in 4m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/95c2e161d5c6445b87f867ce11bb7913 : SUCCESS in 11m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/404996d68b5b4d44b292fe19c3c70047 : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27049b34687c481098add9771f8cf6f5 : SUCCESS in 7m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ac43d4af60594f9d961e83092e6e7ac0 : SUCCESS in 10m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7e7f2ceb756f464eabf17123d69c849c : SUCCESS in 11m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/49e5ea2aceb74bbab21f1ea895c775e8 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0e6571de8c7b4ac2b6525e84d02d3a69 : SUCCESS in 10m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3f86eee33f084c50a4491a9e947ee91e : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3cc68c3390d04eb195c8adcb9dd28a1f : SUCCESS in 10m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8abb2591ab7945d7a111f6ebc71b01b4 : SUCCESS in 29m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/67a8b48b7e994a2bb718937a327f0ef9 : SUCCESS in 13m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b9fa606196884713a636f910d38c64ba : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23b1ad595f2540189fb4b0de174f66cc : SUCCESS in 5m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dafd472d96c5476a8377d088d7f7e99d : SUCCESS in 27m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b31cbe39a29f4db7917ef4f49ded86a1 : SUCCESS in 5m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aeb17cd64b06453cbc22e7c4da8450bd : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41e7422560a847c58625a9c8b56cce22 : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/32684ccb97834bacbdc692fd7b762a38 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc4e65e9ebf45bfa05d2cf73f148770 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/faa0fd8c8a1147838c147c4e0f0e7863 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"10932dcca03ee866b06d90a00fb4601624ccd14b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 00:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8cb5cac1175bbbddcc236a88485668526062142d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af084168d5baf4f2111a6b9c41493e1ce9ea89b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:34:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5155151ec2c4ffebee1dc5a3fe1d4f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b68bb8bca146029f13cdc8eb9959b1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd9015125af4f4697af403c02924acc : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbe066ad00dd47be80c71c8d3e118417 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d73cf650ef3e4b229dd51c1efa8f7c60 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f80f5921d8444e3904a9738ddd7dfd1 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f592a4c4bd4465aaf0ae1d5c87be305 : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb652c84905b4e72a7da531843381e63 : SUCCESS in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/36f95192c69e4ac08227720b8908ef55 : SUCCESS in 12m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7215e7789c114052b683a94e7693bd85 : SUCCESS in 11m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/208cb5604ad44717a9f64a4863c70190 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/97ffd67b61224baba687fffe8246a08d : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/db64446f227d4643984f0c79cdb18b6d : SUCCESS in 9m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dba6cd4c20ee4f6a95f75eb411c598a3 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/592315cf8b874ab5ae9f1a646c408dd5 : SUCCESS in 8m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/90c67834b3b0407197c5690e4edcba53 : SUCCESS in 6m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5675aba76aac4f4994b6dcc647d050ef : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/37b83505384a43418908c2bb65c9c72b : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/16218c9eb1fc4dd48698d06ac309d664 : SUCCESS in 10m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/578145183b314fa39b2fb013b87af686 : SUCCESS in 4m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/81083f3ade8e4090a499aecd7704d365 : SUCCESS in 33m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/188e274d265f4dcda525c562c5fb2e31 : SUCCESS in 11m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/be03cb651f4b4388bbe6437af2234c97 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae01c4698163451983f5271444afc5df : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2eb689ff6ff6406fa6081aa8ff7405a1 : SUCCESS in 24m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d4db9b3809d44349bb91efffffffe9e1 : SUCCESS in 7m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/12d6f6218d7641f88aea2e8152eb5746 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29e1e921f13c455f87da5532c30f28fd : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/28bdd9f9cac9409abca5975922ccc469 : SUCCESS in 18m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ab5ce984d4c4065ab22f9b8034dbada : SUCCESS in 1h 14m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90f004c5f95d47e6bc594d90803100a3 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":1},{"id":"07c28321338935bace8184f135a1c9d0cf160ab8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:34:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5841f2ae44f72ce0b454130f169fa4f4d88088f1","revisions":{"5841f2ae44f72ce0b454130f169fa4f4d88088f1":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:31:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/873396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/873396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/873396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/873396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/873396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/873396/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:51.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.1.0","message":"update constraint for oslo.policy to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I41a9d8343ee9c832b52429411c05026a7d662c3f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ibc89f1b609f8bb7999c235b8808ac3f496a13c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5841f2ae44f72ce0b454130f169fa4f4d88088f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5841f2ae44f72ce0b454130f169fa4f4d88088f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
