)]}'
{"id":"openstack%2Frequirements~873399","triplet_id":"openstack%2Frequirements~master~I3518e971b349bc275cce4ee2c15082dcebb7291e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3518e971b349bc275cce4ee2c15082dcebb7291e","subject":"update constraint for metalsmith to new release 1.9.1","status":"MERGED","created":"2023-02-10 16:34:41.000000000","updated":"2023-02-16 18:33:24.000000000","submitted":"2023-02-16 18:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873399-new-release","meta_rev_id":"8b2995789b78d5a834bb2a5167b63864feeaa550","_number":873399,"virtual_id_number":873399,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 18:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4ca0e567950a3e4e384672cbb6b6066f0c92a6ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f3da4104fe447111799b047a3ebdc7e94dcb83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/830fcb10982045b38bff9bb972c0c771\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45bed1cb4aa3483f94b82241764430b4 : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0346d3ce263c5687ca4aaace9f5478e9510ec291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:56:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/051bea7895794e1c9936a04f4386670e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d81de66420b40079f9dae6313f3c7d6 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5863f0cf65b4479a8d4ffad674e8d1d6 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce481fe55e34820ac1f1b039c92b185 : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfdc3156a72a43118e6b75a8d0df2373 : FAILURE in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0948974d2724439e8c64b17e68a0030c : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84cdaa5c03814d08886971f0cd62ed2a : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5382ce3d575443ecb220e548af20db96 : FAILURE in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a2790dcfb9b543f4975dad0d3eb8a971 : FAILURE in 10m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/21b2b476bf5842158baaeac2918a6eae : SUCCESS in 6m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fd960c76437a413b9c13ae54dfbe5ed4 : SUCCESS in 5m 39s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a3fdcfcbcc634203bd495c9f6c402b4d : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/25a0e18d6fbb4a078e54b55bd962faef : SUCCESS in 10m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/259f0cf2fb494ee58e0c2b54e4906c1c : SUCCESS in 6m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/74033555af7446b4b015eb7c9b9526f6 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f14aa58a94f7414a8b8aa9329d35a8bd : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/346462e0e46345d5b2eeafab8c232742 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dac88cbaf47456eb4c82e0a96503f3c : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/85a55ee91e3d4dd6a6ed29cd95a1c8fe : SUCCESS in 9m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/94c4a1d3c3ce4d609fee9491b594173f : SUCCESS in 10m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c2615260daba46358636f27e911c476f : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f2585f2858354c3494278dc5d95fbd95 : FAILURE in 18m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6051033bbcb344fba58b3350a6fb334f : SUCCESS in 5m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/46c16aa3a5aa49d8af950736e0723250 : SUCCESS in 8m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8b071a9fa8e44b5da5ba57cca371095d : SUCCESS in 25m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/36df94de52d446b59a94f652d86c2ade : SUCCESS in 12m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/eaecc80c673048c08a60303faef1f7ae : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6712e3a677294d02a58a6a844aff080d : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27d1daf18c694268bcabc1796a23cbeb : SUCCESS in 26m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aaeb69c25ee8450e9c4298e6a57cb9d0 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6482584f000d4124a511f57562128894 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9e4dd3791354f1da73c88b76e37ef55 : SUCCESS in 5m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2cfc1587da8345a3b42d4eb190c30cb7 : SUCCESS in 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72bd5df68b3e4b1a8479b7b86f1ef357 : SUCCESS in 55m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06cf6f6e30634aed9457cf33e2afe027 : SUCCESS in 1h 23m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e1d5ff5a0b1551ea83d014ad1058576080cc90cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7961318491ad441d7cd6116858cbc6690ec76a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:02:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ca4f7d20244136847c9472d0527d90\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d6bda508d0c4a3c917c3fe089139a66 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4348f0e4ff61e9bc190abf985504d3c8e8856304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:22:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f8d83b5491241db8e3ffdc05bd9bb9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bcba9000a24991b61ad4a800670545 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d062c1a2a83845258fe07c51bc129db7 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e2419f95cb4f14a3d20eea820765b5 : FAILURE in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5eb218e729634f3e952890a210784355 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2021d64339141cf9d74e0cd52998152 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddf49649c094467f89aebee3ce9d5ce5 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b655ec88132946588fefa962aca2ed8d : FAILURE in 7m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/01b3dc01ae814741a872a47c2226d7c1 : FAILURE in 12m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8669af3ed90b43b98fcd91833f3f5169 : SUCCESS in 6m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/20472e08368c402c8632947a12e410fc : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c04f3e1bc8bd4ec69bc6e83657ad5f28 : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1a5bc6c9a46c4dc1aad008ec4487d096 : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f87b2797f8af4750a55442198ae17b18 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/25594ae53d7d451dbb7f78e18c57a949 : SUCCESS in 7m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c9ebe62d26fe47a294ccd48b8f430c19 : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e3a5afcc2a4b46e6ba097c2cfcd09e74 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0510553ac0a64ef686bfa92bad11741d : SUCCESS in 13m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f0a0eb60a77f402d8071fad32c95d7f9 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1a1be384ed634cda8b1d8309aa60d524 : SUCCESS in 14m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/69c7d6be3892419c82a56e00df16daa4 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/94c58858129a490992307cdd612c74e0 : SUCCESS in 9m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8bb89c1429694d35b118effbea45348b : SUCCESS in 6m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5cdbbec156994b44ab57bd6f1763b929 : SUCCESS in 10m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8ddd0125446f498e83721ab35552ff9b : SUCCESS in 31m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8487134b5aec464ea7c71a2920cc56d2 : SUCCESS in 12m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3191f5dd082a489d8ee0228f1151227a : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9528b2726ab64769864537794590e2a0 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/91974147db4243aca9c7bc662877f76e : SUCCESS in 26m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e78c9b95e43a49908a885275d346c8d3 : SUCCESS in 5m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/83227bdab26045c7ba7d99c33c987163 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c1458d74dbb4654b530156aad1dccb9 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5aee8c69c7be48e9b32c347e88606c77 : SUCCESS in 20m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/240185abb1af4d48a85471b386766413 : SUCCESS in 1h 33m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/982f07e3cada48bca06188948588e21d : SUCCESS in 1h 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3030459ac238662377f250644d4580242cef9470","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b10cb7c626ba4d64d2299da398454c1f5000199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:05:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/167b15d481a64326822aa94262d62b40\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0be979fea6c9457dabae8c9055a9c36c : FAILURE in 12m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3a9a2b81beb8616a07fcc694022303a357812096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:53:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b68f67303d74a2c8505801d673e481f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df363b9cafc4941a8aa042e38d96d01 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1797e8098429479d8549ff023337cb80 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/866838eaadfb4b95a14eca8237c77e97 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb9554689a2d46508e553fef3a6aeeaf : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4504fb6232145a880e09b1270a18d17 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80ecc06829584a67b59c98c5843b287c : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a38b5f4017074962a785315c4e9726ae : FAILURE in 9m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f2ea9a3962dc4a8e8d1c5c912b1a9dab : FAILURE in 12m 12s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/895d4ea866da4650b8f86a77c242b8c4 : SUCCESS in 6m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b2f30f42d69749dfb70e594f8cb8d67c : SUCCESS in 4m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4a980d80fbea4fdd9686bb0f71a1d0e0 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/219b5a4b62e8467b88fc56410259c453 : SUCCESS in 11m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/87af9bd317d04350bfc1968b4e199c05 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fd4d51ea92a74c6684df0e8d0cb8fbbb : SUCCESS in 7m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5a805e65b97c4247a6cec0fc7e5d6953 : SUCCESS in 8m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8f1ac1207cb54f8f85c8e8747dcab6b6 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f2da7e854864a82be3d90cbe5bd9409 : SUCCESS in 8m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c768ae41adb847018553c8100923956e : SUCCESS in 10m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/eaec13a603b0410d81c7c79d416075c9 : SUCCESS in 12m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9cf58b451b704974a4d00ea49791ca1c : SUCCESS in 4m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/39e5cbeb1c56400096851f799caf5e70 : SUCCESS in 10m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4dca8e0842b945b39188875b889c3c82 : SUCCESS in 5m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0d1543ac98c943ebb6bd052c0cfc99d0 : SUCCESS in 11m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/78feee68b4aa4d5d814415da0223af6c : SUCCESS in 27m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fff64276d4424b3cbb2fe9aa1b2b77f6 : SUCCESS in 11m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0f9da990cb4f44d9bd11baaeb1d4adaf : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3c01fad32ec44e1aeea3fa6a99945e6 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb6885a2cd154b77aa6434e0aaeacd87 : SUCCESS in 22m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c23ee8a6dbdb4cf59f0ba75a32ed549c : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/63489b1f139449fbb94cca89f5b17757 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/405b95f5b8324b6b925cd4dbd94d138b : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1bac8829e7c54d22b27c93ec7d1434ff : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f73609c74224bc0976a294b830f461f : SUCCESS in 1h 11m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a04d3d7b76e4ea3a5d1ac79f3d46022 : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b64f49d7aa8b7badb460884dfad32a2b04b1ecd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad778c19064fa6d9a4eff176f36498b3277d5d8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:30:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c582282e9da44442a234995cd2068f23\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb9e65a493aa485c8a51604c94dc28b0 : FAILURE in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"26088a33ad1f5f03c74526edaf46646bed22a660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:54:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292aca1fcf8b4deeaa2ac3aa45904294\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe7f11751064249915d478c65b6d024 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde498473ace485396232403f09cb8b3 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba440572f7b14536a33cf3210bf5f721 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/232facb9404e4753912dcfd62fae6169 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/892bea152f9647f595a8367eef490198 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdca45417cc54540b3be584993bac7db : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06a8ac68590149f5b7a90d8f4a3c3875 : SUCCESS in 7m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ba93fbfb895a413eb5c4e1f511b3ee9b : SUCCESS in 12m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5f196b46d8fb40beaa795142bd339724 : SUCCESS in 7m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b32934c0867348f4a4fddd8315de53d7 : SUCCESS in 5m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/69dc17ef257944349c86a7e69d07c1a6 : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/187f31b8f7504a2a85e38ab1f4076d4e : SUCCESS in 12m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bb185cd995974dc8acf52f94cd1d03b9 : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e546b32afb8a47d39ab283e005622217 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f4262ee1af5d45df95ec2f7338fe4265 : SUCCESS in 11m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/999f54336ea24149b3834662f0df411f : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cbb26399f514a1d99a8380ff2525f48 : SUCCESS in 6m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/42257292ff7c47aa94d8ec6499f0ebf5 : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/764e72f309c44774bf50fa1e7198df39 : SUCCESS in 10m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8fbb1afe245d4b6c80c30052d9a138a2 : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dd27e7b8f7954d6ba71b047e7f9565c0 : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ff0f86a2a787419cbdc494ae58613a0f : SUCCESS in 7m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/dfb89e524e614814a294198401d32c5e : SUCCESS in 11m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/eb31c4a42fb24eed8cebb9bc7b3efd2a : SUCCESS in 23m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b33ae7cab20b47b0a358ec675f2a5ead : SUCCESS in 11m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0e7a2dbed2274deaa71da3278f48442c : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e41c8141a96f4ec1a9a234ee510e00c9 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cdadbdc572341c3816f3a43cf752f4c : SUCCESS in 24m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/37abbbb3571140fda8333dd2ecb43fcb : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/13dabc2cb9b644238d86f958e848a928 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c525d1ab609049288182e27540f7890d : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be03ac3244a6449ea4829da20edbd9ba : SUCCESS in 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b7fe7449ab4fa6bfb42a8d35f49fb5 : SUCCESS in 54m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7154c857aca4c5684b4b4a73ebb9411 : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"edb7415a85e5786f20f68734fcb0c626d751ea42","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"feac7d6f0570abcda35250f614d32b20e2db3092","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:32:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3526713372b5a8eeeda2eeaf712f8fe12e623004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c144abb9b728433c8a5f7bbe253a8e94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec793d74bbdd477d8c80450a9de5362d : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0fa92630e94b8abf9758fe18c0f30a : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76b363ed1eca4517b95fcf350ccad415 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20ccd26185ac45688c7fa51aa74d94a0 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8011eee935d1438f85680e84dca098d4 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f854d10c2a9149409aae73e0f989209b : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0071d06cb8b941478b8a0c93e7eca6f4 : SUCCESS in 7m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4f92b934339a40d5b1720da3fd4170b5 : SUCCESS in 14m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5d8e1f3712eb43308c7aa7aeb19e075a : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/760bbf2a98f04370ab30bb7e9fcbea65 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6450001d533b464aaaa91ef9a299202d : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8c4a9bceefa14b37be0f6305b1a03c26 : SUCCESS in 9m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4912e8bbc0f347cdb58064c6a03f9e19 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e13ed8da019442c6b56918215a6926d3 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/62ac51e310bb4225843dd00b0b9692ed : SUCCESS in 9m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4e9803cc52ac49c1bd73f8f18345eb12 : SUCCESS in 10m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f39c39ec0aa148e3b5c17f89c3a8ddf4 : SUCCESS in 3m 20s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4da42a6695c14337886b5d7c99e18426 : SUCCESS in 8m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e6779b4ca83a4db9b39c03188d0b4311 : SUCCESS in 4m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/135b5315fe69478eaf43c9a8c54c9b9e : SUCCESS in 53m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/56c538d895604c9dabd1317b4c5a13e2 : SUCCESS in 11m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/96ac83232c7d4ce382cabbd8f8e2c121 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19c18add8bae45d18e6b8323b613f0a6 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0497467622234d848dcbd1430c093b85 : SUCCESS in 22m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8f5cb1a231354ce1aedad98576ebebf5 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c56b731f62ba4934a267adda2594b4f3 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46bbbcd8747e4a6b838c85ccbde871f4 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/62d23761654d4a9d9e91174218de0bcb : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b1e9cabe7af4077ac0f4b1918083094 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf203b8ad14b4713b47cbe1359d1a6e9 : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a8d9ba7800866e10622f122ebf09e38d12ebe2f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b2995789b78d5a834bb2a5167b63864feeaa550","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376b7d2be5bb4a6a896c80d1723734d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a540628ac24421789f2998bc84e2f81 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7744053b556c79ac93bf5ab508cd001623fa136","revisions":{"e7744053b556c79ac93bf5ab508cd001623fa136":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:34:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/873399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/873399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/873399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/873399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/873399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/873399/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:34:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:34:40.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.9.1","message":"update constraint for metalsmith to new release 1.9.1\n\nmeta: version: 1.9.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5765af5ae9e0bc1d2e06958a295af4670066b967\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I3518e971b349bc275cce4ee2c15082dcebb7291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7744053b556c79ac93bf5ab508cd001623fa136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7744053b556c79ac93bf5ab508cd001623fa136"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
