)]}'
{"id":"openstack%2Frequirements~873400","triplet_id":"openstack%2Frequirements~master~Ic9bb8bc1203e0158f0ef143a5883ddac78b3fabe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic9bb8bc1203e0158f0ef143a5883ddac78b3fabe","subject":"update constraint for oslo.concurrency to new release 5.1.0","status":"MERGED","created":"2023-02-10 16:34:59.000000000","updated":"2023-02-17 02:34:53.000000000","submitted":"2023-02-17 02:34:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873400-new-release","meta_rev_id":"461f36bef6c2d95aeb888694508f4e31e1cf6fc1","_number":873400,"virtual_id_number":873400,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 02:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 00:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 00:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:11:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-17 00:56:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b5917109c2a84505bf2ad9a4afc06217d4ff07b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9797674e7fa1ad32b03a9f81e9b32c7b1f480ddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5037563fff5a46538e67549db3afb4b2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e67b3296e1854541a7b74a85b955d2d0 : FAILURE in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3c21bda6375eb6365ca90477770e8b6e190387a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:11:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1539c4e43fc3435c950cb5c796953931\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64cf4ffd2384aa998d7a060ade8f321 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5357d3343dec4a8fb21a42cc6874de35 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f888537b31c04656ac04dd3bf204753c : FAILURE in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c737a8a029d429ab653165c80ab6d7f : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f753326eb9da49b08ac50317922d719f : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78433c08037d442e8094290d22a7c686 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b27df52e608a4821aaa175797c094b88 : FAILURE in 10m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0e2feafd4c4744f0b91650d5b0e8df02 : FAILURE in 9m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/71952fc8dcc746238eeead3cd89d072c : SUCCESS in 12m 17s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/59ee8741cfa44525ab55e83b46f8d101 : SUCCESS in 6m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5189db66272e4e3c971109f8445c379f : SUCCESS in 7m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/61f5f04fcee749c4956765a77930f0b0 : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dac9af31762948b7b0cfbe0b9eedd1dd : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/05cbcb1cb63045de8399df90cf4a28e0 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d96124db7d954899a527fed33c01f210 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9ce8eeeaccf46cf8d08a2eb9d778153 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67bc88e17eab4541b6037d52553c30f7 : SUCCESS in 9m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/57e8e72b305d40138b569cbb5d48561c : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1edfda8effae426cac21f44037440947 : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eeff532025814c74bc852e33d2355599 : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/688df071e7d444fe93b03ccf12e18fa0 : SUCCESS in 11m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/da6864ec95014165a45c6eee5692fbec : SUCCESS in 5m 30s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ae910808af2844c59731c1f370f56b4a : SUCCESS in 8m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1e69bf3c57204709bd7dd0b4882bf4df : SUCCESS in 31m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/823afaf9248941009c52347d01bc9f9c : SUCCESS in 13m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e1876647afb44b708cf5f91f72fa2ad5 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8ac7112d7064e879f0b142e774267e5 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a48b4e57e42e4d0cbca2edfd9da93f75 : SUCCESS in 20m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f938f9b9091c4369a6c0250fdf3cd050 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5cc12c9de329417bb07705d893e43a08 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8e8eab5a5ee486b9acfd1235c514e24 : SUCCESS in 10m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cea4e2fba55a4f21970136be27a31dc5 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83226ef98bd041159f06be205ced7120 : SUCCESS in 1h 02m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97d7ef809b6743418458ab3d0387e07c : SUCCESS in 1h 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f12b37142f50eef52dc42acca97e68262d8a53b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1def91aad0e697a182994c4461d9825306443cc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:10:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e57ecbead03942d2868ae748eb49124e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a4cbcc8f9d54183b7d593b879c4864e : FAILURE in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0f24162576a1d4de50474bcb77eaaf25ad796708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:00:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99e6170dd3234651ae0139553a094124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c30a735fb147a2bdef8ea850f567da : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4e3e9ad30b4a728b245e6a9fd2bc1b : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3365ecd90e084a06acd4ee396e85a31d : FAILURE in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55ba43a01e344e7a8cf9f0da02371ba3 : FAILURE in 2m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9363504954904ed6b1b5f9a373e4a49d : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f955632ec7c44f7b82aa8b33009d751c : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/984c9d6b8463405ea3df71279c3f094b : FAILURE in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/255a8b268db0426a82daddfbfeecee7a : FAILURE in 12m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/14fbc8d68a614ec495066ac7db677b15 : SUCCESS in 6m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/096d9952ca2947c583327f7ae9877455 : SUCCESS in 6m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a69fee42ca0546d099f1ca180f0968d0 : SUCCESS in 4m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d892b96e08e441fbbf9e3b157dd05f6d : SUCCESS in 10m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c22b7b0f579f41da9c6fcca4c66da5c9 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/06d4786a34e347f79742019d28db6125 : SUCCESS in 4m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/22a2558989d14f99a47a0868de2bc531 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/09ab5b5cb6c1479aa02c331cd634c438 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e3f9af90cb54c5da13b7e578a1f2781 : SUCCESS in 12m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4186e69d0d774c38867a08ecadfa4e50 : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/503c7100261c43b2bd9759e7caad14ec : SUCCESS in 21m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c9cdf16fbf60483c96601645aab7389f : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2dcfb19962e64167a5788d1c0c736423 : SUCCESS in 9m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/072c9972fb1c4e2387a1694078990dd6 : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3b3f3532eae74efda64dfa4835a37715 : SUCCESS in 9m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/91010705a6ab4c22a99aa044386e9aa0 : SUCCESS in 27m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3a2ebd0f188748d1af9784455497f3df : SUCCESS in 13m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1de8717939454b2cb7e1ec40b6885476 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7f2581e23f54e779b3e0d745a25e667 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24ebaa776e1d4204bacc41ede25dde16 : SUCCESS in 23m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f815de14efb74c91a2cc87da2c021bad : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d288fe86564442fda83b7c1998174214 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5eb8642a55e443da92c7ed1ff202a356 : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90485042e5be42669964e528f9a5ab70 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dad14f10964468985d1207aadb20586 : SUCCESS in 1h 05m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41efb32eb1df4a1791bc98ef48798988 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9e14fb9fa11d93ea00f40ebfcdbb3ec135e3ee48","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28f0cf31c49dd1457c0b1659afcd36f968da689a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:06:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbeaee17fc94ae1917aaee4cdcef69d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cb5a97a8ce2b4ababca5cb03023bf1e3 : FAILURE in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cf9e9b8d3aa1a8a48956be025c457a5b76008165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:52:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cab35692e954bcea8185d54f1962eed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e27d78179fc49ea8829095208ddf6d5 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3fcdc7ae63042f7a75ccf6009561ece : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a2195eee91543f6878412e1b13ed63d : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/338e2323b27d4f6497521209f047b81f : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24554609fe45405aab157bc4a56933c4 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d554af0da43c4962a21bd829783f3972 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7334b46ad72a41d19bbb796e4a900c0f : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a52ffa5a6fb489a9ad9a1549efb8d2f : SUCCESS in 12m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1257c88a5e1449fe95b0ab7a20d8fef6 : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fe84cb6364dd46a3863476ad499766fb : SUCCESS in 5m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/be78582e8b3e4bcfba37488d5de57676 : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/984ef813ae6645a99b3e9d60131ad634 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a4c11b49ef024fc7b3783f7d1942be20 : SUCCESS in 3m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ccde98267f7d4d77b3804e97a6ea2e81 : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d208881f56e24fbd880d4184d42ed73d : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8b1d0176677f463d8b74715d69293679 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36eeaeb12dcc4711b22a1a31346817bc : SUCCESS in 7m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9c730dfc4b57449ab2a2029b608ded6d : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/917494caca884814adcaa714619ddd23 : SUCCESS in 17m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0479fee81b7b4b11b3e348634c153a65 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bbab582814954da89b6e5e8bb4b0f1f5 : FAILURE in 16m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/13bb1b65dd124cb086a588591ebce05c : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a87310ae1ad34b848c1ab190f8498dfc : SUCCESS in 9m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/afb620bdbe0c4543921b210984d258db : SUCCESS in 27m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aa941df727544540bd3be245cd163790 : SUCCESS in 19m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b9a87c796d274d6db59f66588952e383 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b92af9757ed4326878159fa227ffba4 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b2e769e2fcb47a9b8d225304bd8b2c2 : SUCCESS in 21m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4847797b25264df7b55797eb6616e668 : SUCCESS in 5m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6d6ac985d7c145a79888cfe52514953f : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecba2dbaad60430f90828d797edcd84b : SUCCESS in 5m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/617c2dc467e54c37953219c2e21e6338 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bde3a1af497472e9f797bc3024e0b7b : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a652b02950c04f5182bbb3f9bacf020f : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"aa05a76093bc8cb854dc38e9c1dc6e86c1ae45d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5fa5eaa23a70f8ed71b1397b809f1d1ee63ee848","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:34:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f2a8e1fb24485cb08a8c8589817c95\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2fde2567c6484b87bc2a9dfa87ae72d8 : FAILURE in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7435950cca61b3d40133d5eb7808d0fe1df59115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 23:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad8775c23f214cad8abfab6e1c6950fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1818fd4c3a7e43658f3920b405f37705 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834ba82d1bc047b681b72168ea5cf53f : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6fdc2244ea848dc9636b5d950e3f949 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f6af9676d6d4d72a346a51f8e56784b : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67e1044acb5a4e649adf3357f98ca1d6 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29627eb0bb4142088014d79b50f88398 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c74fcb83902a4006821573497b3d7eb4 : SUCCESS in 8m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ff0e15682f34d889ea03306a9125562 : SUCCESS in 12m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d08b868c38424fb3bbdab3832665ee1e : SUCCESS in 7m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4d492a3b1cc645e2ad34164f395300bb : SUCCESS in 5m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c308a23227e94816a839204f57c28139 : SUCCESS in 7m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/83dd2fd1e09c4f30a71216f827d421ec : SUCCESS in 11m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fae4772cf7bb4728a7a80f927ac86ca8 : SUCCESS in 4m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/35382cd48ad14255af02ba099deba462 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b14b58a7b8ab44b89009324ad30cc668 : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5ca6c68a0be9498aa27c3aecc68618cb : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/880c906404e94680a85ae043598b42fe : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/47f4de1d27064eb49452d11137bc4847 : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7cfa4bc5354a486783a355fcead3c9c9 : SUCCESS in 10m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8feb77db47b64e6fb6f3ca7a66214ff5 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/29d58dcfb08a456f9a44ba1ae890e98e : SUCCESS in 8m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/54ff21e23f0044e8abea494667702b68 : SUCCESS in 4m 30s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c9a6cc17183f4f94a341ab5d6a209203 : SUCCESS in 12m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/31d99b7b0c394dd69ac7da17f43135b3 : SUCCESS in 23m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/db56fc7347a74c5f847e4c7e65b1fd24 : SUCCESS in 11m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/863dc252562943479a309241bdbbb7e9 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22c8e6d881824d7a9f11a5e3b4bfdb6f : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92c479a955114ff9ac2a3da9268a1d22 : SUCCESS in 25m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e61ef41c7eef45f0b982613b9df1d4c5 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/883ee8cbd4644110a6b4d3f0f05787d6 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a689e6535e904f2d8a2126721a4b38a7 : SUCCESS in 9m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9ddae0dc8b5f4d2f9538e3b03d01b52f : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58e60b31c0e340b98540e2b7c0ed5621 : SUCCESS in 54m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d43c8f146b1144029d1bb09b1e20f6ae : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":1},{"id":"814b6b802feff48469709a201ce68537af42f5d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 00:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a489297f99a54ad4a873436425494d1d5b4fd2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:57:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6355126aa8138621a9c5347e4baff354226abf35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:34:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e8f1b7ef28d452690c9a961006f1d2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f916baff67452da7660a9910105ad9 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bbead939a54d4e83a228e6fb5622d0 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d0302d17e4747b8aba846b278e95280 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/803911c5494c4768a5a748288b780866 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc381a7acb1b4c0d83d8eed1e85d56c4 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efd16a165b7f4267924a6d0186d7df6f : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb49c14dd0594e3ba643feb1a0daca97 : SUCCESS in 8m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/caa29aa04aa542b7ad06e6f27ac60135 : SUCCESS in 12m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f9d6cc7580cf4ff0bfe123e8622969f0 : SUCCESS in 11m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/22f860012d4a49bc8c72f6726f8dfae4 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5799b86ece2f4a6f9fe97b0c70bcce31 : SUCCESS in 8m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4fab4cb1b8d5477499523bc04e18c8ad : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0bb08c7861ca4caea94a78ae9c8dc16a : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29e85f48c79c431c9ce11894ed207026 : SUCCESS in 9m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c6b5535c105442089c1dd280febc02c1 : SUCCESS in 10m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/97dca2c3abe3416eb5b6986e35839a98 : SUCCESS in 13m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b80d2874b8b34f16a620eccc783d4b9c : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1eb84eac19834a1b8d665d9b2500a197 : SUCCESS in 10m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d72f197c4a3e4bd9be0d24b2b6dde2b9 : SUCCESS in 4m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5dcbc8b6391d4a64bc11a515c957c5bf : SUCCESS in 25m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/722f2180a59c4d8db52228b58b7b801f : SUCCESS in 11m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f9d297c56b234d1f9734d93604da81e7 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9205130f8a1b414c9b5f3e8fd3f41a3b : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01bb40a09fa848f1bcbf642a0b2acd67 : SUCCESS in 23m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2932dbb59d96405cad7babd7ae66fd78 : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6df3c1422c3e4858ac7c6fd53c9715ca : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/605d751fd6fb4146a8964c14c138bca7 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/38e385c39e8d47ae833e9490657ab776 : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/157c0c4be8a94653884c415bdf47b767 : SUCCESS in 52m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e30db014b18f46f78fad5b663bfbd197 : SUCCESS in 1h 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"461f36bef6c2d95aeb888694508f4e31e1cf6fc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:34:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f594d1610d6c20c09d4cd3a5ff499ecc51ee0d2","revisions":{"7f594d1610d6c20c09d4cd3a5ff499ecc51ee0d2":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:34:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/873400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/873400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/873400/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:34:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:34:58.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 5.1.0","message":"update constraint for oslo.concurrency to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I131db402bcd2ecb69af23a9f6939638c3e4e9186\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic9bb8bc1203e0158f0ef143a5883ddac78b3fabe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f594d1610d6c20c09d4cd3a5ff499ecc51ee0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f594d1610d6c20c09d4cd3a5ff499ecc51ee0d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
