)]}'
{"id":"openstack%2Frequirements~873402","triplet_id":"openstack%2Frequirements~master~I860fb203cfcbec3de0f4226f3d3ccf98d2ac917c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I860fb203cfcbec3de0f4226f3d3ccf98d2ac917c","subject":"update constraint for oslo.rootwrap to new release 7.0.0","status":"MERGED","created":"2023-02-10 16:35:27.000000000","updated":"2023-02-16 18:32:05.000000000","submitted":"2023-02-16 18:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873402-new-release","meta_rev_id":"8b8c5abb4df29a7e5145b4b04515a3a2f8cb4a9b","_number":873402,"virtual_id_number":873402,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 18:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"778cd11deee9505b28b4da7937ef9cad68758326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:35:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cb4ddac9d7bc327453d4d9a77c6e09f9ca3038d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:39:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a651d932d141416a8cf14b9c11b13701\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0eb3f59deb9452cbfde14b82ec1daa9 : FAILURE in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9c8b1259e2d3e50acd0a0a52636eec0e932e9a47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:46:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cd1e40783294fde90535dadf583c39b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e693f770f6da4c2c919b09ebe8c31d2b : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132d1ffbb48a48a1861b418a10476442 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b85ad646bd5c4a82af8cacd8c07bf3ef : FAILURE in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cddeffa4eb149e880909f1e4aac1cf7 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8642cb23ea804b8f932cec9f6d84e1c5 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5ed26e571ab4c1181c9a4b14b518e38 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b374c8909c5944f08c604267048bce07 : FAILURE in 7m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/870e5a468f9e4fa18bc01c0b592dfb26 : FAILURE in 12m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f7ef8217e9bc4c1aaeebba00bb1d5523 : SUCCESS in 10m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b3cdc149208e46b886d050fe622658e2 : SUCCESS in 4m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c3d3179c85a049f2bcfd2f15e4478b09 : SUCCESS in 6m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/159ef5cbf62a4f48bd3f4ab2b47a9531 : SUCCESS in 13m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2b2641670a60413aba86b2e2dcfb9e80 : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/235ab721d7c04e61b50bf09d9e020612 : SUCCESS in 7m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3839e02a60d64fb0845abf03153bbc9e : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4c73ea50692045b192f80e5fb9f652db : SUCCESS in 8m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc2b17c9444f4d82b47e16a0a4bb39b2 : SUCCESS in 8m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b5f369f58aab47fd9c28fb67999a2a74 : SUCCESS in 8m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1a0a6e579c8844c2942b7666de3f0f40 : SUCCESS in 13m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bdc7dd12a2934cb8bf0d1ebff66c00b2 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/56979d6b9ba34d0b9d2f77ba9664797e : SUCCESS in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75779c57804c47efb41ec2fe90790d78 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8006eb8c8af04995af56c621d4c77015 : SUCCESS in 9m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/485c92a99a3e44da9472120ea318b68b : SUCCESS in 24m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/24f62a8dded841a38a161b0ab1ba5498 : SUCCESS in 12m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/08354195c3334aa1a7d1d2b603035927 : SUCCESS in 6m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f021209db9e47d7b96152a56d9db6e6 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6db91c3eb21e4b50bf5f297291fdac2a : SUCCESS in 25m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/98ba8eb0ebef4a098dc3526b7dcaa6af : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/28f1226d138f46f7abcdd32ee4ff42ab : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d17c4da1dec401eb2faf7b369475aae : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3130a1c2c34847c59906fb653caea173 : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34c685d7ec8843aa832c1fcadc842466 : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41273d86648a4df9b9654636618ebb12 : TIMED_OUT in 2h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"34c1ff71638c3e793b7d79deb853716597561ae3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2d0f498e6cb3b1c6e5b152b0df5fceee4d7d9f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:02:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea73ebe7f7f4b22a20db976a1c9b999\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8e777d291eb84c888f7231ce15c7e19e : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7a90a9f89f0221b4b5432448124c840895c41c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:46:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44e96bde42554696824e1039c4d0e671\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/025f310c65ac43b1a7ae0c95e36166f6 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfe50a6f2064738adf1bd4653852aa1 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79550fc51104dca87705c47efd1d19f : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcee83833a914cd68a343cda809afe97 : FAILURE in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a81eb4108e8640dcbc91f1e2fc6efdf0 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61f131b3829b49278f81548b7c20e79e : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae545e3b11864c6eabbe201b0203e401 : FAILURE in 7m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0257327090d9478899f49e3d5e3be884 : FAILURE in 12m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3938e5e77ba04c2f85f0ccd3ec226cc7 : SUCCESS in 8m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8ac6ff28662f4a1499fe4907878d89da : SUCCESS in 5m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/dfedc2903ff440a4b42e1e263195745e : SUCCESS in 6m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6c26a19c5ab74eb38f3220088d2f0f42 : SUCCESS in 13m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dc6c0686edb7468990719b68a1bce53f : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2376d42b08fa43b7b4fddbb634114e54 : SUCCESS in 5m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2e312fc1115a4a5f9ba55caaa9aa7817 : SUCCESS in 7m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cbc15f7d84794bc791fc1f9f4d134806 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5dd0055da3a4064905e646e3be8f93c : SUCCESS in 24m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0c1bad588880438b84641594850c6817 : SUCCESS in 10m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/192e1983b7674d9b912c99a7ac7dba32 : SUCCESS in 10m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/43b35111f8ad47cc8084a822f0e000ff : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2c3b330ab94d44f7bf1b1a8c68b3b99e : SUCCESS in 8m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/763a3c6c316b4ee08a6bc445a4f75f6c : SUCCESS in 4m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eb8db770fffc41798a896f574e8b5d22 : SUCCESS in 14m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/813a273c86074796bc233963acc8cb6a : SUCCESS in 26m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/03dbe4aedf544bfa8ee08c69ded8002a : SUCCESS in 13m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1d940bb972eb40209836b07f667734b3 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b8e910325a14be8ae425fdb4862646d : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65a9c64fd7004a219b3a3f610461102d : SUCCESS in 24m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/39bca21c943d427aa551df54f5d5b02a : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6f7a66a5f60444a2a8a594805981dbff : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66f3c2e3c5504c638a059f07783df58f : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/60cc23c4fc104ff8a4bee9dc16bb28ce : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5fc68e2fea4db492bcb3d12afb99e6 : SUCCESS in 1h 07m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/158db674e7874ec48a13d6b6edd1e612 : TIMED_OUT in 2h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"103d08f7adb375c1fdda0f6f364e936f80ba6660","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3f9c4d3702fd4606a170a0ee264136edae7e52b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:29:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494797d4a72a4df4baa373a151ba6ceb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7d9a55bc2d7d4242abb5701ee11ed9b0 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1bf86811991ecfa8e978f13e91401895b2768718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:21:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da65bc26d0c44bea250e2adeb2c58cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53789aad75740449d724d44bf2abbd4 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db7787792374d93b2bafd0f286e1625 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/801a87ca1360473489c535dc9311d05b : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84da541f1e904e7ba96401ddcd4696fa : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19844dc1f7824a17892b9129d96b73c8 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9eaf1df2e17f4625b81981c698967569 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5dbbccd29f084a7aa22bd14357a4b7a0 : SUCCESS in 8m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75843c595e344cc3960b58b320541111 : SUCCESS in 10m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/18cc06340bcd41d9be37fbe4e7f7ba8e : SUCCESS in 7m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0e5bb7dfc1f943bcbdd620e33c8c8ccd : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e3e0ac4f12cc4e8d8b4d66e3123e2582 : SUCCESS in 6m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f792e8b3594a4e9cb40c0a35e1ed0ce8 : SUCCESS in 10m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0ab7190b7e024353808722ef97c06af1 : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0bdb05941fbb4db8a02b40b12b5d8054 : SUCCESS in 8m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e8d5735f750442b8af063944c468abb4 : SUCCESS in 9m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/136342de8d3a4347aabd5750e4895933 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2564d45d43a40b8891f25e0abe42d61 : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/75a9b9b7b1d9480a87a4d562e89dd942 : SUCCESS in 7m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4a1da39f78fa48c2a97ee2a766abf252 : SUCCESS in 10m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1cc6eeb1394e4b2e9077c8e59b579510 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6953d6fadc754b47a2d33548cb9f7498 : SUCCESS in 8m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/eab91199b49b4df081da2ed29106f176 : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b8af4c48d8394d3d80eeee05165255a4 : SUCCESS in 9m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/15b6b5ee3b54442394b16168a58f4b8a : SUCCESS in 22m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/57a4e513daa647a9a445fe72227e0b58 : SUCCESS in 16m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fda11fa5ec1b42959f90cdbe67f489fc : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b2b6191477f4d319e2aa3d23000f996 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4086d6a431674a729ade1dd0958864fe : SUCCESS in 24m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/65f52ec3568a4232ba8d3d2d9e7e5dbd : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a7787b8cc3454e768138341e37a310a0 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b8d1ee25c9db4ed1accee358e34a6b38 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2cdd6c0f7ce4218bc6d0e907f56307a : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8007e5536491466da011bac059b1d096 : SUCCESS in 57m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c25053fc62cc4b709568e13ed7e0b51b : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d711fa7ab21f9d88d27b9bd7c52f94bd905b286e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bf253bc81f8b326cd50b0765d94dc754f35e8b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:32:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ea73b86f06f8a2db62a7b81deb3b03d2e587560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80bceca54754b848747ee614cc621b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4d2f123d9544289ca411b57c769db5 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fccd31b2c22d46059f2fa020df147de9 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f92f00030a54e2b82e735252aa8cd62 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9c6ceded7e6412ab296378daf081d21 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa89a97b19f141d388f8ff08fe869525 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/411af34443fd4eb69898cccaece20bb3 : SUCCESS in 3m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3574c698de944e598d57c2b344509ef4 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/34d32de220b84522928000aa88c1ba71 : SUCCESS in 13m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e0ee991360274334b0aa2303ef7f8c94 : SUCCESS in 10m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8af8877f3f4e4c08a29d5c963a082954 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5818b0ded8274b589fa192c3939cbc2d : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/00395d35ebc84467845f985fb236082b : SUCCESS in 7m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1d214386dfd8454da400c8862ccbb9bd : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/040668a4902b4360a547ef4d7f5cbcb2 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/66040a54cbb542958b8ae4f62f9ac7aa : SUCCESS in 13m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/56c1ea278c214bfb908ab902625ef758 : SUCCESS in 12m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ecd571a1c269424089217c7b3b7a2df5 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/987978cbcfbd45569a143c6be1f02d25 : SUCCESS in 9m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/156e02cbdba74879827d841404de532a : SUCCESS in 6m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/af11668a981a4b3aabe8b57a511dc87e : SUCCESS in 29m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/73c0e57b8fcc4ea2a2e97f46cca271ad : SUCCESS in 11m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4a1798d9fdca4d38bbeb6e04e56bb882 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36e0981221cb438ab0d2701a496cbbcb : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/026dba5e7abc42d0b306fece80669918 : SUCCESS in 31m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c1971a116e734852b6a376cbfe613d72 : SUCCESS in 6m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b6f6053e259841979ddb2d23eb4e7e04 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7adc37a4461d4d0a84c0035857fb038b : SUCCESS in 9m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f855110378d44dca6bfdb1cab67cd0d : SUCCESS in 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59676e09398d40cea7b59bbf14e1cdb6 : SUCCESS in 48m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0f37f7527dc481dbcd49c7ac3732c42 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8b8c5abb4df29a7e5145b4b04515a3a2f8cb4a9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86429af50df3d90a4dc316920fe2469a2f2d0d23","revisions":{"86429af50df3d90a4dc316920fe2469a2f2d0d23":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:35:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/873402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/873402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/873402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/873402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/873402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/873402/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:35:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:35:26.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.0.0","message":"update constraint for oslo.rootwrap to new release 7.0.0\n\nmeta: version: 7.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If40c802b34c4743118ae73eb7b4968957c69692c\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I860fb203cfcbec3de0f4226f3d3ccf98d2ac917c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86429af50df3d90a4dc316920fe2469a2f2d0d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86429af50df3d90a4dc316920fe2469a2f2d0d23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
