)]}'
{"id":"openstack%2Frequirements~873403","triplet_id":"openstack%2Frequirements~master~I4c35794719afc53122c3c5aa1bd14f79dfd48742","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4c35794719afc53122c3c5aa1bd14f79dfd48742","subject":"update constraint for oslo.privsep to new release 3.1.0","status":"MERGED","created":"2023-02-10 16:35:38.000000000","updated":"2023-02-16 02:26:13.000000000","submitted":"2023-02-16 02:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873403-new-release","meta_rev_id":"75587e8270046362a21a59d935afeead9ab4ac0c","_number":873403,"virtual_id_number":873403,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 02:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 00:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 00:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:07:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-16 00:33:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6179ccde2f31c94ab3cda4a488e30c7fde234ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e177f095712463d93c7f23f5564571169d3e54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4853a3ba8e2044b99b3dff02dee91978\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c32c68bb8bf747329a782831f1c4c21c : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"877f17704d1632bdf9f7d1d81da91938ee855348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:31:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55b4123a227248ccbadcb72df6d55e78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b079ab8ad6c141529cac78f7dee79805 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63310d553e24144b45315aaa948e64c : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d23cccadbf440b2a64b241c9a50bfb2 : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7851ae61704542c895b66b7828258504 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b900ad1944874207b26c1d434ee70bcd : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be8d3ebb12384d8caafaaafb76ced6aa : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb8c27711c434df5b50bc49856327894 : FAILURE in 6m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a1520f026a34418da299c2df4858413e : FAILURE in 12m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d8c66e50d1914451abf4f7f85f9037ef : SUCCESS in 7m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6847c3d0529b4c4c924a88281b3dd061 : SUCCESS in 7m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a968667afb4849b7a6abd6b4ddc8cdb0 : SUCCESS in 5m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0d158ceb50634ed49b8994d7bfd07a88 : SUCCESS in 15m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/05d7214164594459b7a9480e7f8adee3 : SUCCESS in 6m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5bbdfd6bd55e4759aa94f26754d73307 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9e3b49b95213438489e9ce74f2a0d074 : SUCCESS in 9m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b31c6bdaec4945848a405920fd774609 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37ce4ae279014aa9b971cf5067d46fd3 : SUCCESS in 9m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6eb742e4fafe459891e5c828c0070c6c : SUCCESS in 8m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8f82e70f65e042dc81fbbaac3cf26f1b : SUCCESS in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/722a329ebadd4a299c42c71da3d3116a : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ad8ab9c9528d40df8cc527575d39568b : SUCCESS in 9m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/613b93dc1e78471fb3ab0e485e463bd1 : SUCCESS in 4m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/92a577d61c4f46f6b4fb3cd06c9295d0 : SUCCESS in 8m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b4026266b26748d09d4d2a43b6735daa : SUCCESS in 25m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7338d76e7a9d442bbc45963d0c174b47 : SUCCESS in 12m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/958048225dfa43849e46b7086d8b9acc : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8adf30b1df9c4532b5ce070c7ce3da33 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f44024f99da4fed8ea3e1e8e039db87 : SUCCESS in 22m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/052144996ce34bd381b79d85b03c1aba : SUCCESS in 4m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ca6e4f30a69c468fa159d2e0e0dbc944 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/834ddee559bd43b08754bb554cc12e5a : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6c098082ebe2425589cfa66ce04c2384 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82b0fe10a05040a29c2d5473cbc918d4 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bafc991f88948b88713490f8960cfe5 : SUCCESS in 1h 49m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5e8d4a329cd88d8c0be08ddc72bc08bded3db222","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53ac9c366eff39c0097b0eb46d8ae7a3cac136d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:27:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cfbb3b3eb2a4263b595726443a8f7b0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bc8f9f04156c4d89991d02ea7d29b23a : FAILURE in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ff348a71a9acdf5c216f2dd854439e4f845eff73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:50:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3473b6b65e8f4b79853e98a86f612437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73df77e2970e4625a8f352eaafb312e9 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e008df6487594268a8f462b2e05d43c9 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c21c6aa52c3470499b835039aa77fcf : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e995fa0565914947b3795f63b2e72308 : FAILURE in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6017a3a102f4d369a62df14ea1885a2 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6d93067596641dab1b3c951ef6b8bf1 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8612302168a4e8b9836cbf6e5b711fa : FAILURE in 7m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f628659857824467897f5aa3343488d7 : FAILURE in 13m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/70fde32116b645ab8d2f5711d51ce675 : SUCCESS in 12m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c19953d7d05b4901970f86909d491f94 : SUCCESS in 5m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9c578a02eb884bdbb6ef460a520e2b56 : SUCCESS in 6m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b2035302c90044458d3bad2abffe9107 : SUCCESS in 12m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f84b67c2294c4bd49c5937eb53c55c5e : SUCCESS in 5m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/21e4f78c0f7f42b2b31faa467b413113 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b9f3d6d256df46689f02e3dc4be1fc5b : SUCCESS in 9m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/37db07aaa11b4f1898940f95356435b1 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0014e010a4bf446ea2afee3c2e710cd0 : SUCCESS in 27m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5c4c10cdba0e4d7586ba18677c0043d3 : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7a2051109a6d4e33a30f23e63746c9c9 : SUCCESS in 10m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/024977d464e045cea276d808fbc6dbdd : SUCCESS in 3m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2100f0fe33df4f03af46a5f2a8324f14 : SUCCESS in 7m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a8691cfb0f23482e980ece863d04adeb : SUCCESS in 3m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3d00259d50fe4a5184573b9c6c428006 : SUCCESS in 9m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ebe5ff61d00f4a13aa587cbcdfceb141 : SUCCESS in 29m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3e1bb2dc04794e3abe691cd75000b3c1 : SUCCESS in 13m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/efadb4973efe40cb90fbe25914737982 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9df795014a1476ab498374c67859489 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f2e1f1ba12645e0aee639873d10a093 : SUCCESS in 27m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/abb923fd02d8444a870d04088918382c : SUCCESS in 8m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/151af3e49a6043b885e43df8c5244e77 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0544d3002b76480ca1ccb1343de06f38 : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/80f7f8b75a8e40f894f5c797cbba03d6 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27aee59adf474f279c6a557190c1a97b : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/158d0767fc824cde871600724021e919 : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d4be3b59120a1e2a2be8e2392cff4d7ccb4770b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:07:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5746d67da309cbdd0b58b9286b9465849f71f59e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:22:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c42632d2ea438fac51c2e69bfbb800\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25a82e0f3e104d1a80496046096f4223 : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"299a62c5f806669b9ca44d9ea5d5f222a2039e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97c2ea5fd3b469a87bc7154c61056bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3dec17d5024c2bb28b6b30d041f78d : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab724949365b4fc7bcf5e16869c6f083 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26ce86c839564a0a9d66da83c6d09d71 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f49a6b8921c4a598590612d562e2e43 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/244d09048882461ca65a596537da37bf : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d5b82ce28274442895b5395185df677 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c78eaf85346f49d6b9de88859697ec39 : SUCCESS in 7m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b05239b7ea62491d91614fc0e6b0b019 : SUCCESS in 12m 35s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c7324a3eeb7b4d2bab55d45c15f8114d : SUCCESS in 7m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7e4e3af194994131a269d61eaf06b49f : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a33673c459994ce5a5f91acfc4596c26 : SUCCESS in 6m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/03a93545a5204742a4611b713dfd682e : SUCCESS in 13m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aba46ace6fd64748b2dd1fc0ab066a25 : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5d71e772bd204323acd04dd3e3d98262 : SUCCESS in 6m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c393023af6774e9ca153a0aa404b656b : SUCCESS in 6m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c397dcf98b4a46af81908edf98c06173 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a85e42290e024a948e8733ba69507eae : SUCCESS in 9m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/37eaeb8d773d454093ed001a8787128b : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1766bd0907a34cf3b74c004bcdf4fc1e : SUCCESS in 17m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d51b09a4748944278809520edf654479 : SUCCESS in 4m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ab68d6a9799c44b98803604f83c7ca84 : SUCCESS in 8m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bcd7f1fb40c0414d862cce451364be33 : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d354ece6ee6248c4a6acc7d0f0118192 : SUCCESS in 10m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0d7b6c2123ba4ad3bc2bd1aa4a421864 : SUCCESS in 31m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dded0022671844ee950544095ffb6cc2 : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f358ddaa9a724925bcd22f7cbd092104 : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/132bfc95b63c4e84afdb7b9e5f34ca1d : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23a312962480419991821fd7eeb2e2dd : SUCCESS in 20m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cb571f2f0b444fc3ab3fd7b5eb9bf576 : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/89dd4e2289b149e9b6affa69ae5566eb : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88f8071f8b664b73a3bb65a16e2a7208 : SUCCESS in 6m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d31bc7cf4e4149b58dfa2d3eb632f2cc : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/596813014cc14ae88f16721c78595005 : SUCCESS in 43m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/471130a8a9fe4dc6a53469fe654df534 : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3ff6f4d5d0922f6094fa79db3b3861f4e4013d19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 00:33:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73ea48b7f374f75116f54c4724fec49b2833e2df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:34:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7534ae78a82f362f4ee5bd9704f6f191cc1c512c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:23:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6577912fe864505b5796eb6c0bf053c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb62e0ac2cc478aabbfacbce07b3cd0 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca0c2cee93f492da042d2e2e69e3a20 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8dd6434943d49cf819ece3ce8b4aebf : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/297cfc8a8c204b2d8918ddde692a04a3 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9ac6e1f92e04225936dc1965729918b : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea3828aca1ce47e49225506fda966608 : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a32595742104416cb3b24b6e4b0eb7f0 : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ba8f4b5af164a378cff09499377f098 : SUCCESS in 11m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2e07609892a04cbfa7ade309c9dba5af : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/36ed8289a41c4f22aaaf3508f429d57b : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/53cd9878b5f74b5a97e80264ee13c1de : SUCCESS in 6m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7b3c4b787fd14391b0bbb8b23bb2ae69 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f84dc31ce8a94099915ec9ae794bd33e : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/320ecd5e3d984dc09f0692d40eb49cca : SUCCESS in 10m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d7830df3388d493bbfc3877db12b90b4 : SUCCESS in 9m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a694d99cb84242d88870b92a19cd2a11 : SUCCESS in 11m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7fe1babaa42347c8bc26880f1db579c2 : SUCCESS in 4m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/58294b3e6ffa4bee966f3f641dd6b555 : SUCCESS in 10m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2fc652cfe2514cf1a1f6e8c7c4e17eb8 : SUCCESS in 5m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/685555d079ef4dcd81e0e0333ae277fa : SUCCESS in 26m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a150c602f75c4577a1d33dc525ee19fb : SUCCESS in 14m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e788cf2ba5c948408b6f3b6c11bc631f : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/211b99c7adf34326a00bf79f706338e5 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2781af404f244bdfb36684c26d82a8f8 : SUCCESS in 24m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/36bcf72390a9461490c34c04d094e000 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/623d14ae24be416b97077ddfded01828 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/170e7d2059ff429e8d54471d6ea3f5a8 : SUCCESS in 15m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/46105ab81cbf491c959935cf1a078bc2 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d95c046bdba04d23a1accea2ca21f578 : SUCCESS in 44m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a8efd2919144327b57c700836a1c621 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"93bd6963d2219224a7b110c2902029b50d542645","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"75587e8270046362a21a59d935afeead9ab4ac0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:26:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e2915ca2074eb19e73d8728119a5e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660ec4b00b86440cab903f02de3c8611 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c438ebfb9bc26cf79aae8ffa0ddbc06e3e187184","revisions":{"c438ebfb9bc26cf79aae8ffa0ddbc06e3e187184":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:35:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/873403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/873403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/873403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/873403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/873403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/873403/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:35:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:35:35.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.1.0","message":"update constraint for oslo.privsep to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I439fbba05c648fa69566aa5a0abf4b5c345237e5\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4c35794719afc53122c3c5aa1bd14f79dfd48742\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c438ebfb9bc26cf79aae8ffa0ddbc06e3e187184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c438ebfb9bc26cf79aae8ffa0ddbc06e3e187184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
