)]}'
{"id":"openstack%2Frequirements~873415","triplet_id":"openstack%2Frequirements~master~I27c051da7ad8b8b0f261f0d85af3836ff7d4b4b6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I27c051da7ad8b8b0f261f0d85af3836ff7d4b4b6","subject":"update constraint for oslo.upgradecheck to new release 2.1.0","status":"MERGED","created":"2023-02-10 18:03:59.000000000","updated":"2023-02-16 18:34:23.000000000","submitted":"2023-02-16 18:32:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873415-new-release","meta_rev_id":"de11e31365855430324d80f16d27f60bf11d9a8f","_number":873415,"virtual_id_number":873415,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 18:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b984c28e1f3f0d358068f58db5e73b105cd56038","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:03:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0406f8cc3fbf65135191b38c65871698a5ad18e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ebe29ea955d4507bb91a96f452be5a8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06d24487bb45452f93203d2f374d21a5 : FAILURE in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2e14063dfd33a10bf5ae3be2df32a638ded15041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:38:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8afaf6bb22c84916a0be425f78ce8894\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed7e2c0e84c4f6ca3de5346e5089fb8 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b555b5beda4791be8aeb2b85b32b8a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ee7b0335904a46b0001503f0fa13b1 : FAILURE in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e22d2b39b8434482b13d721db3fff170 : FAILURE in 2m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f5edd5d4e9a4252a6f321d629738e8a : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4bcf6d036ae4e958c57d6ab17c846d6 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab6c674e1c264e8da364f2c15f49c0b2 : FAILURE in 7m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f3151afce3c4637b611a3e51d3f0693 : FAILURE in 11m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9cb730d0ed9a49fb95685c6deeb86734 : SUCCESS in 9m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/04fe20f245af4c2b9069477abd30ff6b : SUCCESS in 5m 23s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/07a14d88952a41d3b1cc1453b7b9e4a8 : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/439ffb350e4844abbfe43f45512c9d69 : SUCCESS in 11m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d6ae07f3c0c245bfb6a1856a8352d708 : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7bdd05919bac4f73b6d3ed427d53c83e : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ca7bec14698844878e59a7636fb663b1 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0abf30c26fdf4607af086baabc0f9e29 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24e588b1d8c84bbd988971b40f975b98 : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4c146b86aaeb48ffa32c0f503311e5d6 : SUCCESS in 10m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2a6ebd94929a443fad0401875a76c083 : SUCCESS in 9m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5d1af0eb3cfd45d9a98f384a81832afc : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ebcb2071cd4b4629bde52473c0ce96ba : SUCCESS in 10m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8ed3e67dea4a45e199c9b5bc09a7093c : SUCCESS in 5m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/06a07e60e92a47d1a6a2135c43e3714b : SUCCESS in 8m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c447e14df70240a294ff7e697853ecb7 : SUCCESS in 21m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d29f825a0e2f45039f53cce7dca2dbc7 : SUCCESS in 14m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/654f506c44fa4ad1b19b705ff654ca1a : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98403cbb152f444db6e0a1305ae4aa50 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ce850326d144f67aa0d1ceb593c71ce : SUCCESS in 20m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/415d1743e3664753a357e14c108b92e8 : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c2f8313bebe84a03a4cf060b814d5e7b : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08718d2b4bf04c429e68d66402bb7c2b : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/572e0d2e87924ef293394164f477310a : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00cd518535f74a88ab3a9267f38d64d2 : SUCCESS in 1h 04m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9558cce728b47b9859e45516052b15f : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"77f0a07a3bc64a82075ab4bba3ce61a6c69896a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e2c3a9dd96ded3d1c8636634a980a8b293c912e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:12:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91a8c67c1e7846c7b2390254a6d83688\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8775c53b976c46cb9cffda5b7190cfaa : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5fc54fafada705607464ae958d0a28aac6016621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:49:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94ef2d64dfd9448d8eea0b11b90a16de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496f7357dbc94d12a194f4702dac40c8 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843eb5812f454b95b404af7365eceef7 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35cdf732720a4872a84284100dad47fd : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c112233d7cba48f9874eb6ffbaba2da0 : FAILURE in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bd7dd67f0e64be99c3634dbab9f7be6 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/580f189a821d439387fdd15a587234ab : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8603eb248afa4a20a9b9d0c6189f096f : FAILURE in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c3571c168b42408796afc790eac8899f : FAILURE in 16m 18s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/880940c60d554d6e8fab429b10e018a5 : SUCCESS in 8m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b90f55846d1c4dcba74db27f1f28deb8 : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3742341192944e1bbd91953b97158b31 : SUCCESS in 5m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/05dff2f4bc4a4e96934a19b773dd9b74 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ab851cce3afb49ef8159f807bacb80ee : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dfa11aa243a6432a94d56951cc98e324 : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fca6140e882d4cd7895c24b3dd8a0d85 : SUCCESS in 9m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dc1e9fbcacb34b25bb754f7c7bcd554d : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51eb155bbe754e8089e138e8b1c68bb9 : SUCCESS in 14m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0a48beccb3fe498087d90136a603bf5d : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ab0b82a877034c34be4a74437f233b0f : SUCCESS in 13m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d4bfafbbcf8046128fd49e824d9b6aa1 : SUCCESS in 5m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d683b63b01434055a8190c869e71eca8 : FAILURE in 21m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e3c439e3b74b4d62af2bab46f3def1e9 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c44a7518ed9a4d4ca88a83039b5afe3a : SUCCESS in 11m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/41c84c6e01c34ed9b8711df8d03def71 : SUCCESS in 24m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/43acb97abd074f0288fc9e13c9f7615e : SUCCESS in 18m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a726a78683bb4970b3677d9cc6944955 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/868df001c3d249f08242a8d8c812a662 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d195c39269a4969ab9204d0a1032be2 : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c8902d6aeea64880857c915123958a60 : SUCCESS in 4m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e6a2fbc67a9c4dd89968b7cc8120d207 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3cd119c6b3e4205bb8a400e78b18e6c : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/240d5d9582ce4204a24e0ad976813c9f : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e9a1924efb94848890397fe31e634e4 : SUCCESS in 1h 11m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2afa10efbeb6421fa79253bdb877536a : SUCCESS in 1h 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b25e2e2c01d29be91ae1861cb3f1966641376b70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95c03489a13716214b1ce4b740542e46b30a4db3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:10:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b88bbb4cd9e4c12a980101c2eb7d7e5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aed1cb8dead4430ab2435d2d50863abf : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3a4224c6f0fe336a70c353248f1b36e2648853d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:36:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b67ff434674e45b2468f5df8d1279d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a34d38c7c194a1b94684eb3dc636eac : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b266960751a43f788f19567a689991a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ddf22c97d30468b89c3f7f91e1dab2c : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58e2f7c7c433404d9df685dd8c608fe0 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0c117e5953049e9b1e09bec1d47f753 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ea807ebce2141c2b5b8393e00f60c42 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcb9720d7fdd499b838c5e920852807d : SUCCESS in 8m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/705d45fdf24348fa998c59abb1475dc7 : SUCCESS in 11m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1da0aa119d5b448b81fce4e5aa1aa485 : SUCCESS in 6m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/10bf5bf8b2f44666943ca897cac96a25 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d102490e83bb4d019c81b6f9af12e2c2 : SUCCESS in 7m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/63429980672844918f9b5adc6d65bd19 : SUCCESS in 8m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/736220a42a5b4f6f82c89c6a455bbd06 : SUCCESS in 3m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0a5fee6569fb438caa4cef806f2e2784 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/288b84039cd64495bd79fb360572ed9b : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/94850879a0804956b0dbb7982f8f69cd : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00a257d49d384a769b1aed884e73af9a : SUCCESS in 6m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b0eac2dfb41f4236883c6bdb95f80ecf : SUCCESS in 9m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3a88576612bf483eb033fd7af07163dc : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ade83d956b114d58a85ef3febc9010eb : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0633c74f7992419088e9a7c9130cc96e : SUCCESS in 8m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d0c139aa713546d9b117e8dd1ad9391e : SUCCESS in 5m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4327247d3d054e1a8a1a3bc244adacbd : SUCCESS in 9m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b856d9cc79e44ff1ab3e25088ccb8d50 : SUCCESS in 24m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4bcc8e2c0ee04efd9cbdf725417a9e89 : SUCCESS in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bc90873fe0414d499c4e41fb724a5fbb : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/511f03c050f048ac91721fc101631355 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a6f1511ab7b420fb9272611ddb7cd06 : SUCCESS in 23m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/11099605ec724aa28807fff17a3b7a34 : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7611b31446334acbbf451cdca1054e98 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5f34754dbe54ebbbab95a559ff836ba : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b65fbc06d908454f9da9d4e6bfb48cb6 : SUCCESS in 23m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac78917b3772436ab24d514b9b0f6566 : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28edec7e8d0e44429bb90288470c6510 : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":1},{"id":"502f05313b9c2d24f3f722668a1e7efd1065fedf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebf8eb887eb0dc3cc2b01d8b2dfd6fd26570a966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:32:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97e449eb8790e019d0c8f989e565de19bb5140d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d88b4c7e31a4e489dabbcb382249586\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc68bb181c6c4ac99e5c8acb253fcd41 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f428dbfa48544ec5a60216fff409716a : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4e9a8a90c0422da9b38b90621ccdac : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a73aa4598fb4f6dbc5fec5068115bf0 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef91ba8176e4af9804c7b8c263d77fe : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e91bb99ffe84631980edbda8312666c : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f92599f8214a48a793d020fd36954aba : SUCCESS in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/96e6d2dd452d4371aacb256054422c4c : SUCCESS in 12m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fd19be6947234188bfcd6434a51ee762 : SUCCESS in 14m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/147e55e587194fc7addeb34528603abf : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f457124662264688855b66eeff27e8c3 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dd1b66ae15b543888c2202ebb2a6c468 : SUCCESS in 8m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a46028d5455142b2ba512ec175c451e7 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a841c476635b4142a97650baf23c2b39 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3631b48b720b4a17b61cfa513ebb25dd : SUCCESS in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c2b4122a16ce48df9a0b6da6679d9007 : SUCCESS in 15m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f1feb53f350746e5892e19cd717aad0a : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c230dbeb8e3749d4ab394678c5f5633b : SUCCESS in 8m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f712424977244829aa63ac8cbfcb2791 : SUCCESS in 4m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/56f34c3c927948dfa1e4c4e3afd34e0c : SUCCESS in 22m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/84e9e06b248f49db89da46d002b8dd58 : SUCCESS in 12m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2b91dc7441e646d48c64ce4077c6ba53 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c94dfd0d1064e0eb371772bc7bb10e4 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/000201a4fbd0464da4244df6765131d7 : SUCCESS in 26m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e6b5317ebaff484581af1815e87a369f : SUCCESS in 7m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c6b11f2fbada423ca27a0c53c2293798 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b713c553008344de855e8de162b7809b : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41e53224b82d4391810a5dbd6b43d76c : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26f5671d889d4c5f97790eedf42fa3f8 : SUCCESS in 54m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21e5b1cb04b8470b9a3bd5023bb00201 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f48d15935b1bd8ba63d2d951afcd696bade86297","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:32:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de11e31365855430324d80f16d27f60bf11d9a8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc529a3af9a641b3a3bf987d37f7793e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985d225ec2e6423fb78b7b617fef6fd4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3291f9c45e389ad4b1f8307aac53ca184aa80080","revisions":{"3291f9c45e389ad4b1f8307aac53ca184aa80080":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:03:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/873415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/873415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/873415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/873415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/873415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/873415/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:03:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:03:55.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.1.0","message":"update constraint for oslo.upgradecheck to new release 2.1.0\n\nmeta: version: 2.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaed592bf5dfe0a0bfa1ac2012e65ad16d6e0f2a6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I27c051da7ad8b8b0f261f0d85af3836ff7d4b4b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3291f9c45e389ad4b1f8307aac53ca184aa80080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3291f9c45e389ad4b1f8307aac53ca184aa80080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
