)]}'
{"id":"openstack%2Frequirements~873416","triplet_id":"openstack%2Frequirements~master~Idfe4c6a4a593d11982aa22a933d073364c5aaf30","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idfe4c6a4a593d11982aa22a933d073364c5aaf30","subject":"update constraint for os-traits to new release 2.10.0","status":"MERGED","created":"2023-02-10 18:04:23.000000000","updated":"2023-02-16 03:00:33.000000000","submitted":"2023-02-16 02:57:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873416-new-release","meta_rev_id":"5dfe752609e95877a3283b6e3b3547dd1e4195aa","_number":873416,"virtual_id_number":873416,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 02:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 00:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 00:33:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 08:59:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2023-02-14 15:54:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:23:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-02-16 00:33:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9a7a30924b4c6a778cbf64d011397518bb647ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a314a2c3b7c91289b164e94f8ecbea06187773fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6772d8d8d514c578283ea595b1384b6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6a32c63531f84123b40e47c51cdb01c9 : FAILURE in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"955d7d9294d014b59ec662be378230ef85a8e26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb3870cc48014152ae33b4aef699b5c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd68757786441bbb27e9d79fbb5ab05 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e1f6be67b54ffba5e0a929497e060b : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/848f187bb1134421b6e03c8ee75e3eb0 : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01d16ce196c14acf85eb4119ed10cd49 : FAILURE in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71562a0e7be24d6cb8350029b5f42236 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/afbb31d601c44581b6946f76ca33e6ea : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa355f7fb20a49de9e6b52433241f089 : FAILURE in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/393780f3369540f6ae766ad61fb26ea7 : FAILURE in 14m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1d24c6eb9bca42e7835b912f9aaa9391 : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8ea78480225c4cbe988f2ad06c1777de : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e2f1aeac18ea4be58e026875b195c198 : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6273ed33cb4a4f069429b0c441a3efd0 : SUCCESS in 11m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2e5b11ab459a4421bd7d3b8826a99c49 : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/60590aa397d94685830b6048a2bee3a2 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6f79f2c5c988405e9d67b8579083e196 : SUCCESS in 9m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/264e8d856b874850a1191852dbbcf5b7 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d819d69000d4ffc9441402cfd84ce01 : SUCCESS in 8m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9e998ed56d754133bfa7a14e3d54fc34 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9112608bb1ff4fd78826dd354c445b0c : SUCCESS in 13m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b435e4ce66a6425381a1cdd114b603d5 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/55a9760d949a470c8f3845e6c6978336 : SUCCESS in 8m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d1ec2cef2ded4580af51cd1a3c29910d : SUCCESS in 4m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/dcd070422822445bad29a6f53ce0329e : SUCCESS in 10m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cce158b211264de9b10e0e493b6b43f3 : SUCCESS in 35m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ee33fc14cde345d09c62b6a173b2ffe0 : SUCCESS in 17m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f2341c42e64d447ebab798d2040d46a1 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df32793471d045548d689fbe33fb8ac9 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fea2cc67d54841699890c7ce01af0da2 : SUCCESS in 25m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/09c82ab76fef42fd85141e0dc654a068 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/32f7086b1ff84f9084c6f36c68ee6a52 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7943e5ed7db4b2ca2a154ed870acfac : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa80d807ea8e4864b826799421ae508e : SUCCESS in 18m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d0f8b29fa94323b5f88895a393530c : SUCCESS in 47m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b82615f9ee1c4689b224d3ad38976223 : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1176eee5bfc228c09052a0b305c02dde349096ee","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-02-14 08:59:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f44e89d8a9e20b313eed7db9ae350aa677a343c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 10:37:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96048eb26a1b4bc1913c6081088ae2e3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8e262363d31c435495f04f4e259ea667 : FAILURE in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"df98bf629a6c29f23fabc8cd0ee831be8d7d184e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 12:24:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d52bfc0e1c74dd7b4df734cac40aff2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd1a5cdab804526947615241706ba6a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d436ee9ba0645708b9d09e14a28a30e : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7440583829e343089d566af6ece786b8 : FAILURE in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a9c56a1797b43bbad8f044abdb373b0 : FAILURE in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ec0149732ab4287a22bd80d7de0cc26 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2998b2bc9dfe44679d396659739a27b8 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a90eb9b10698451e9a13ff9fef22f055 : FAILURE in 12m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9047a5beeba74f9a9d371f021d13206d : FAILURE in 12m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4fdcd01800f3470a81fce3621693cb88 : SUCCESS in 6m 57s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/afa5e788039f422b84f750d043f3d997 : SUCCESS in 4m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d5e0577826704861939e0c1f97f8378a : SUCCESS in 5m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6100cea5d7da44b684776dd75dc0d04e : SUCCESS in 12m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2d115023896a4f2e9a56305e4d6bd837 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/20d8e97ee737452aa177c968c84e64fa : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/efee60622df64380823464b44fdfbdff : SUCCESS in 7m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b36041d7d2ec4728a124c1b18c6ae407 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f765dd9479fc4db7b2d6dc1b027cb180 : SUCCESS in 12m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8c4666e562804754922aec2f77a5316e : SUCCESS in 9m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6f1b17e350544811b29b63dfd4c12288 : SUCCESS in 17m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7a66d3e913794e6da96b73ecf2175217 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/36c1e005e5fe4b67b6bf8ea581990017 : FAILURE in 18m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5ca9ae5c0c6c4a47919ed40772c9d6b2 : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/19ce404544874f3b946857bd0bb7d068 : SUCCESS in 8m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b3a8db4b71dd470bbfd137fe8d6e38a9 : SUCCESS in 30m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/831a5742217e484880f00e712152167f : SUCCESS in 12m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d926aa70ab374864961900153df533a4 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1d475a2c41a48088b3a19bed2997e77 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7166dce82f374ddda62fdd37ca863988 : SUCCESS in 23m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2a6394d9e0d44469bcb69f31032d9b9e : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d0f871d367e2496f94f0dc13adcfe0ff : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fdd2c8f780549e3be3438d751b2677f : SUCCESS in 5m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee4d6e323e5a4a7a81ae9272b7381510 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20339c65484d4914ab825f5d3696980a : SUCCESS in 1h 09m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86b2261813d44a989b8c38da257d2efe : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"608a00df9de357ab2f5c5be6a7715d13c2cf4ddc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d1f6d835f93e59e11115ad9c90baf1cef00c48f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:11:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e357896bfd2847449c1c73cd3ba34627\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/02fb92bb80744cf7bd0c341209d7dfff : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e51ce5f44d853f89c80096660866ec57e9d51889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:18:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a16bfc2dcaa94b8a8b0e742aaa9a19dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa9c16fc9ec4305ad0fadc7aa7328b9 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33cce579352446b0a82d177a989ea13d : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5765592307cb41e5af0354f3c5d3d9bb : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51dcc6b2d63f4117ba309bcb329a7b10 : FAILURE in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/850d367b1cb742eda39d7e0cfe185452 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44a5a249e0f8470db9a7002bdc466d1f : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c387b306a9da4a2b8ead66893d64b679 : FAILURE in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05949dcaad0c4ab680b7f73de05d7b41 : FAILURE in 12m 55s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a05df211961e41c5a7c941475a74557e : SUCCESS in 8m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c9de8024c39844a4a69782e5d2b8c0d4 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6852a6cb7dd744f2923c44864ae99d90 : SUCCESS in 6m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/458ec13438124a21b84b57c788238d55 : SUCCESS in 11m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9f533bf190fb4282a2d93577d19b3dd7 : SUCCESS in 5m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f4bee184e462463ab639c4d5af7ef7b2 : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b2c17700398e41d49a7875ec6ffb7a3e : SUCCESS in 7m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/333970fcd9b44c4c8b6823ba102b5877 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08122529ed194d7088bffafc7adea442 : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c6784c8730574c24994fad9a409c845e : SUCCESS in 7m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2ec2be36af284b51a5a83a585661830b : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/567e8c781a4e4ce1b7c0dbd3e7ab39ef : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/876fe2f3aadb4b0089c9efd209b707fb : SUCCESS in 9m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fef60e33b9d84ab0bc3a7c6c3099deb7 : SUCCESS in 4m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c91c8d71ebbd40989f23fc2bf0f5fbed : SUCCESS in 10m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/58a7e3f96cdf4a48b060bba24329b23d : SUCCESS in 30m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/32f10fd7b72e44b6b730fa8afdb25ec2 : SUCCESS in 12m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0df417f31bf742cd98ae88d2a6a669d9 : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8d958a1cbd5476fbf1e263aad479cec : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d07188295be4ee1ab18137f313d5a5f : SUCCESS in 25m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/92a6812fdad44af39d1f1f36bbc9cc63 : SUCCESS in 6m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6dc35380074c4f248d94939791faf52b : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce6f22b9e6b4482cbbf8b2ac4254303f : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5ca5d589b9b64602a40254657e140d08 : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2b15a3ef07b46efb4f22d4898fe9676 : SUCCESS in 56m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/224ff09c596e471581274d3fd7f8ea5b : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"36fa18f3588a7e537f244f618389fa1554be321d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-15 18:23:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d7e767808e383e49f8267d55bad8e4a01e61a24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:45:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c471cf62a5204e12980f0580f1a4f711\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2d194cb51ab64fe582bc47a6524e727b : FAILURE in 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1e837189cadb098ffe101d7c283113ba2e8b1927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:18:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6366931699014d76898f4c0d0f4d608e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67965fab977f4b53b386e7a564dbec5f : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e6ff15419441b4bda44ee748c5111f : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0c9b159e01542338599e0af8269b410 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05f436b5aa5c4b778e5e1aa4b0303738 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fb3bead673c4dda94c0cd3e44a93834 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7bc32481d574ad2a9590690c8e16a55 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8db5838b40b4c65ad512df63e4ba0fc : SUCCESS in 8m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1cb0331be5544009016c6de785a491e : SUCCESS in 11m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2702a6d5cc064474ad8f50d444089387 : SUCCESS in 7m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/66b11a6805d441d4b611fe82648a58cf : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a5b38a6dd4544a02b5e238f0b704aec7 : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/21c02afc04414d3fb3a54e7878ff6460 : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/329af87bd94444bda5840f3bce2230cb : SUCCESS in 6m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/039fa8d6d2c24a13a6e14d583c810765 : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/208c23fdaeac4cc698f19c309d6a003d : SUCCESS in 7m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/15a9d6534a994cb3af8b34712a55a550 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/457998b068a54390a7c73d2c81d3a60f : SUCCESS in 8m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c96139762eea422ab016165f08f5455d : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e5d32756415c427bba60b7841bafa850 : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/61be0c6c4aea4f83aba933af783e1929 : SUCCESS in 5m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/14a7cfd491074017a5d98cad3d4e686c : SUCCESS in 9m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cda19667d92a4ea7850e85c86c4978df : SUCCESS in 5m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2adaa48593ae4f80bb8d5cb10d25adbe : SUCCESS in 10m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4829167194984e95912c6677445a4c65 : SUCCESS in 31m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/babbd913885745deb33ecd08965bba44 : SUCCESS in 13m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4518f4bb13c74be58ca3141c5e43056b : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eba00c86b7e24ae38af492c2cb62dab1 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56071cfecc9843d48f37ba43fc1c0895 : SUCCESS in 31m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/475bd0ea85cd4c0d972d792a98372692 : SUCCESS in 5m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/87e5ac50b5b14660acc334cb4bacc2fc : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79f47c4772234598aa6e368caa7c7ac5 : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f9aa75a92dd743bdbd3e9175b43467b5 : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0227af5357824ee4b6e1378ae5b96f6d : SUCCESS in 1h 05m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/736ce59ad22a490e8a9635007bb24674 : SUCCESS in 1h 23m 25s","accounts_in_message":[],"_revision_number":1},{"id":"440726e8741efa3de51bbbeb7f16add8771187b7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 00:33:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a37cbcb1572391d2c19fa61ce39b84c23d039fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:34:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ba1740cbe6f7062946858c19aced16bc5706dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:57:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53d551e2a544fdcb64a7ad721cd343e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac04826a34dc48568001bb1299f40fcc : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76ace3e96a44e9086dde483825b894a : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94f5d36b75a04bc29722675cafc6ce05 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a383d81bf0b412da4f1343272efc6f6 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b78207bbc554353b6ef46617613d7c1 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1706e3eeddae4241873d569a57d9d739 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7016b47b25f41e49afd30a651e23e70 : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56e171f514c84985931edc2fb92987d9 : SUCCESS in 15m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/82eab7a0953e4931b4726550914efa9c : SUCCESS in 10m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/971fcce4fe0b462ead38ba22864fadd2 : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d80cfb84d430437aa8fc652d65d77bba : SUCCESS in 6m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/870e2eafd1374df18e1287b181d11c5b : SUCCESS in 8m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cc7af4c77a8c46ca99013ea85b7b2432 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53cfd30b8d844c9396354c30ad00a148 : SUCCESS in 7m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8a40624ec07a496b8ad6cf2d790b515e : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3dec4f7ebc7e42f1950daa79f74b9bad : SUCCESS in 15m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0ae8cee1965c49e98011c1a00b6ca2ec : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bd68fc44e7184aa5ab9874efc312068d : SUCCESS in 11m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a31f8aca30704359afad9116e4465b58 : SUCCESS in 4m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1d386903b92540738c3c8f943b6ad048 : SUCCESS in 31m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14b9407f1c1d4139b210ecc3052db025 : SUCCESS in 12m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/215abb8288b24b30ad2395da66a54ae1 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1b6b6be8452463bb821c9bd46fb3297 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54bed3aa9d6644749b9f06c021023b11 : SUCCESS in 23m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/10af218533e843ec837e562e80d83189 : SUCCESS in 6m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ad8c884379f84387b985e8dec1a8ed33 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aaeb1a9c757948138542f0987148961d : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6ef8dbf5498e4167a13c0b3939f3ab1f : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b35f4c13187b4c24a3024a00c8d3484e : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d4029cd55cd4e42988e401525f1e50f : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"56c7ecd3a233a786caaa7241cdbb7db94fd8ff06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:57:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5dfe752609e95877a3283b6e3b3547dd1e4195aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43495fe967cb44ca886393ddcea338c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd37cb51b3f249da97d037405a55af9a : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9455449052214cd5b40fb0bfae3d84f8f7203273","revisions":{"9455449052214cd5b40fb0bfae3d84f8f7203273":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:04:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/873416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/873416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/873416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/873416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/873416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/873416/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:04:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:04:22.000000000","tz":0},"subject":"update constraint for os-traits to new release 2.10.0","message":"update constraint for os-traits to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Change-Id: I7aef295bfa9b5af34d374b73c7ea0fa165609073\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idfe4c6a4a593d11982aa22a933d073364c5aaf30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9455449052214cd5b40fb0bfae3d84f8f7203273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9455449052214cd5b40fb0bfae3d84f8f7203273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
