)]}'
{"id":"openstack%2Frequirements~873417","triplet_id":"openstack%2Frequirements~master~Id73719d16ae18281be674d2be2c5493896b7cc38","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id73719d16ae18281be674d2be2c5493896b7cc38","subject":"update constraint for octavia-lib to new release 3.2.0","status":"MERGED","created":"2023-02-10 18:04:26.000000000","updated":"2023-02-17 06:45:00.000000000","submitted":"2023-02-17 06:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873417-new-release","meta_rev_id":"70503d3c0effc140d909f63ac3e1133d3d82a7ca","_number":873417,"virtual_id_number":873417,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 06:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 04:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 04:57:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8881383d70661d6ef7d43da0d274d7d08041f6a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4318a886f40d15f8c7099a2cc9e7de53f031139","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:23:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a08529abb74b868f4c7651a8a5f1aa\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74da77ab392a46cdb6edccb091de0e34 : FAILURE in 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bd9aa1af995c4f7eba8b4c317cf02c47ffb66dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:53:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0e14acb8f5b4bb9b2712a1e4fa9aa7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2bf50d16bb433ebf97c05688a1c3be : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9836bf775c045da96a077da0c060bff : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0614511fc18d46df9e80fbb24b0d8427 : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e6aa7e1ae44bf09b26cca958974321 : FAILURE in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c366b2587d74f5186bc4a35577af825 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d90d84603db34ded8b9d8f1ffaea8cb9 : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/835c75f9c7ab40459b68a7d64ef5c633 : FAILURE in 8m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/31289c38f05441a0a55536ba7102d97b : FAILURE in 12m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/621fe8128dd147fea4b0ad5530b840a8 : SUCCESS in 15m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/81fe677a9c534ad189d115f56947bd93 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/12d71cf5c6954950bc2c67f8b3140e00 : SUCCESS in 6m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d0df246a28c54d959e4baa28a526e955 : SUCCESS in 12m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eb60780ab36b401c869b075b6826c4f4 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3fce7ad58ada41bab9dea9436df9c0e7 : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7715e34c874a4bffa5ac63392d0baacf : SUCCESS in 7m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d496e7c4c8924b079f940717419bc61a : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/493df82e386c4d4299e95ea29e33018d : SUCCESS in 8m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3318be7d1cfa4f48b3602067ce70fd96 : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/48b704817ef14ffba1510c95faecc921 : SUCCESS in 10m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8fb8ce9663b44e5d8c8870c8491947c9 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c248529050764770abd54867a231c902 : SUCCESS in 13m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/48a95067bae7422592d27d9fcebc674c : SUCCESS in 5m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/95821fbaca7b4cf7a03371c219a87cc9 : SUCCESS in 10m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a02700f671474cbb9c7b965f71f24703 : SUCCESS in 31m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1f7cd3a30b654f53a7bf3f37bab09974 : SUCCESS in 14m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fe3b616a2113456ab723c4230e7ca6f6 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1426acd8e2ff4793a0efa176be11ea8c : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9963d0b1905040fd9a55e787582529c1 : SUCCESS in 22m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fbc186ee476f4dc2a55cdd19d21cfa7b : SUCCESS in 6m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/238efe63c6f046b48472b92489f32881 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0f3a16652d8471f84ec07863068696c : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2523396f51641a3b52eb06b8b453abb : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e037fcdce56a4f999a6cc00abacaccf2 : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c899c166ef7748e496bcd673fa698600 : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fe918ce434e31450ddee97f5a82d7c9771408c71","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"426c49f9bfe3c45962071ff7f01619b913f6b65c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:11:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddeb5c773df4bd1b1de6dd61f544af3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7529be3f084341c3ac933275b5f09644 : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b973c4c0e32443f5f51b6d8e6064ed51da5f7f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:07:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2945e44e9f8741daa7a5c49f63fd14b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273f98babc934d72b439bf6b3d867f27 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44fb64a66304491e9437fea1af12cce5 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3564f0bd81284f42a1335e8d823959a6 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4cc6b4e9452434b93ba5d7c56916db8 : FAILURE in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29d57637705b4f30be48d4b4793e3b1d : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d38e07791cb2461b92c7be2aa22a67e9 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bda6554f60bc4addb1fa54f7fc9c1756 : FAILURE in 7m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/149c713920584c14be800a1d60769dfd : FAILURE in 12m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/730778f9192d4153960e32337ab39746 : SUCCESS in 7m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/41254a607f144cd48e53ad689819c9d4 : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8303649e45574101a1c1361937a49c69 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c82814bd8546431b9e631eef82efa60c : SUCCESS in 10m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5d35c705e610442f9b0d8710faf49683 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/217d0a53ce004f9d9ac5b85dd856b8a2 : SUCCESS in 5m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ace6aad16def406fb9cf81ab59feb2f1 : SUCCESS in 7m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ef32fd5a3217431cbb32007803a570ee : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/598548c8a75b4683b0c69550ba25785f : SUCCESS in 16m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/30bf389c93344d0e84d5df8d27dc3a5e : SUCCESS in 10m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2e0d068ef38f40ee8123ed17bedc3900 : SUCCESS in 16m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/080dd77243fd44d3affa8e5a3124dbf1 : SUCCESS in 3m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a5ed5c6112934a7cb6984ef06245f323 : SUCCESS in 9m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c82e642cc9d348f6bff2ae11bc78d778 : SUCCESS in 3m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c6eaa424e81145838d060a4ba432c263 : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1e1f02b4dd354827af8f1e2fae0232a1 : SUCCESS in 25m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/836c1ee6a0324c45bfffebfc161bcc83 : SUCCESS in 11m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4c3e0a68a3d64d58a29e958304ce883f : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49616ffa7f7b4eb69bb3bcb82625393a : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76f42153b7644c34a9f8990b0145f357 : SUCCESS in 21m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ac29e0876dd6472bbb05f532f73d2b6b : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/143657d15a034e56a9cb92e8e43f6c96 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61853308f2954c9da69de3a8d1544c05 : SUCCESS in 8m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8868168f613c49c681e41035b857f4fa : SUCCESS in 23m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54d25cd0af3543e9bad605c43a4e642b : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef500dbea39342dfa2797808bcd9903c : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"413bb3063e0b3182d1b7a9bad134f20baccc5643","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1576d27f3630bd5c7243fdb351bf2147bdbcbc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:10:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f4780a65774b49b40093922b2e4840\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4302afa4c1af404b8d6b5b51e3959327 : FAILURE in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"09607a170643b1e26f01aa068d8114bf4a587888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:30:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a99d5ba2004c0c8bb212b444ef3b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30f7348a5b54dd4a93862d3fea695d7 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3687a673fa664c18a91d0b6dc3957415 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f96913ce0153403a8fe33cfe7f541073 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71233fb252194a56b0b34be7bb14a4ae : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00ecf8767d1a4361981f62a42d8f2145 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f5686f177d84feba8bfaac253391b85 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a664c79fc5b4bcba2f28baa35eda829 : SUCCESS in 7m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/65e8569fadb3482c98e0279476d84530 : SUCCESS in 14m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8b90b904ad8048b293df31a84daab809 : SUCCESS in 8m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/88b0e706953946de9a20fc4c9663abcb : SUCCESS in 6m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7b3cfa7fcaec48f3862083f3b5ff1757 : SUCCESS in 5m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/00a215b1866543c7b8261c69589abed0 : SUCCESS in 11m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/59d403942ab04514b14d90ba54a6554b : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c4ecca72629e48f4ba884a01c00b728f : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f34bbb8ec1f6480bac03f9c2a2739a82 : SUCCESS in 7m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ee84cb09b94f4a1399c72adb693097fa : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0188579680c7414f86a726a090e31bc3 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/090b9cc6073c4e648fd0a4c5c6665bed : SUCCESS in 8m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c7f3c3da8da54585896587e95ee9c284 : SUCCESS in 10m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cdbe270b2bc24e56bd3053b889dd6577 : SUCCESS in 5m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/087102573beb421fb67160fc19623894 : SUCCESS in 9m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c6227b235eaf4062b98dc4031b910c49 : SUCCESS in 4m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6848a69cd7e640fdadb5292bb289c1d2 : SUCCESS in 9m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9955af725a364e7488ed292fb518da0b : SUCCESS in 22m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/caf9560231764b55a9e537d398fdc8bf : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/336c8ef782e444359d9dfb079e9346f1 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57c1a641387f4474989dbd63c525d54d : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb116fdbd67340fe8f0ec1cf13899e14 : SUCCESS in 23m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/12e7b8f4f4e747d2b0eb195e912df319 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c2f49cb56b56447fab80893448517108 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f0f501e175c4b6ca3731de68d06c7df : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41a07d302d4c43fa941b1841dd1cf210 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/debd3284822f429598462e24939c7e1a : SUCCESS in 1h 01m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99b4e0baa98b4f5f934cff4695385515 : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"79b2111bc56893fee8c32a312bd187598f975b5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c1ed3e018ff527b5008ddea3bf0039dda6da8b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:32:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48782059c2229620ce109695015b8b0f728cf099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:48:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40067f9541c946c9998f5af3710dbc2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc21cd2c22e4d818c083a7244f9e122 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a891e9d70b444a4f8ad9a7b068d87f59 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6232ac6f22bf4212a2aa5535483177bd : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe7c7b66e6984708908481759e04b583 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ea442927482483890d2de91a4e5e39c : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/984ec2e85cbb4515b5c8353658ff1cda : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/18ac1a33b90f4ab5956b35ef79361ee1 : SUCCESS in 7m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c98a113c8ba420285c663afbb3dd406 : SUCCESS in 13m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/018d1f7ed5d346c59f5c669b25412852 : SUCCESS in 10m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/111739396e2040a1ba977aedf9761c3c : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c7b6725633c343faa30279a1a1e194b9 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8ae7df59869f4b33a07739bc85913d27 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/81a3470f65474c878aeb3ed21e923e77 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/31dc95ab15e0447cb81725352edbed94 : SUCCESS in 6m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6c94aeb3056a4e9b9ea9fae43a92ba92 : SUCCESS in 9m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f0358eabee25499c9820ab98eb1b32d0 : SUCCESS in 11m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9d3ba8aa33fa4c3bb451a742411d2ca5 : SUCCESS in 4m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9947439b3e3f4861a8c4ae7f056e85ab : SUCCESS in 8m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aa1a95a1cb6843fe976a5db82c2efc5c : SUCCESS in 4m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e0ced18fb4f949bca98ebc9022444fbd : SUCCESS in 23m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/85e5fd7f36b94baab624f697360d1ca5 : SUCCESS in 12m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/724d021886bf461e871433e07c147822 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de8366c72e3546efb692b14a7aeb18e2 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04f2698fbdcf47dbad55f5490c0780c8 : SUCCESS in 23m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95f008ef6df949d68fad0718a327508b : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4545de56f0174892944e016f805fa7f0 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fa4675405144c2697c16aa4ee97dbdd : SUCCESS in 5m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9ca5bd6d78a4462e938f178caf0ea91d : SUCCESS in 23m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5ca854ae3c14755a6e5ab21c7660bda : FAILURE in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a67f52ea5327416993f08aa35a244337 : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":1},{"id":"aaa51a54d14fb5039c8df748388e7a51a2699eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:43:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21ba05bde708dc8de5c61bc01734a702e598760d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:58:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc11a0c7dd7497fa0da89ef616b2057\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e64db63506b04b9180a8f01db1e7fcb9 : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"165642e80148dc796873f421d6561d632d53a782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:56:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1ca10fabc746b1994d98e2f278a5dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b8dd64ffcc45f8954cec5ae1aa0c5f : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2cbad37e61466bbe8bf7bf9f54dd9d : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c04d3b072a3f47bfb8210d780b3d462a : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0fe43ec095944558fa271a12bd9c958 : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a21a255f2c564838b43389919abd8328 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/676568847029464bb5b20c549cebeac4 : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fec5004aceaa4f18a0b81f9894a6d4a2 : SUCCESS in 9m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8440b9679d5c44b482e98246605ed085 : SUCCESS in 12m 13s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f0b67f53608e49b9ab298b7dd125b738 : SUCCESS in 7m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d87ba86260bd483fa5b051c25a94e4ee : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/710e97730e59462e8d9585bb256b061d : SUCCESS in 6m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f999e1af07824c7884a12387df8b1a14 : SUCCESS in 10m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fdd8ad5af7db48e389aed1d740a9dc3c : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3e81006dac7f4ca5b9b122342c026862 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dc3e04fd707844858daa943b0a0e3717 : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/086a8909b0c143779a2911e0394fde38 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b17ce03254d14f3cb420f0a2b892ea9d : SUCCESS in 8m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6f0787367afc4f95b6a6575a5aa00398 : SUCCESS in 10m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b3f40eedcabb4cb2b8e4d67fefecd8de : SUCCESS in 11m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e8ab6f080517420e940e22cb06463afe : SUCCESS in 3m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d675fc95092d4e9a87cc0c9447bbb03c : SUCCESS in 7m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ada0d7c2b6af42d2b59bc6c4b20e4192 : SUCCESS in 5m 50s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8319acd8a8054046804cc25b948cad39 : SUCCESS in 10m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cac4c93c9a2a489dad4b77ed4736155d : SUCCESS in 28m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fe476a7be5b34df484c08d95273288a3 : SUCCESS in 13m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/45329de9db254a65af161799e2f8bab4 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dfb8c2aab748410591690a75832647af : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d818847940434d6a8a0c9b94ba0b8326 : SUCCESS in 20m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/43df20cfed714797ac85c587296ac411 : SUCCESS in 4m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b7b0024c2d4b442d80a8d05509323cee : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9120c2361ffd42aea31b99d2c1354dff : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b85d232c2b184ddd806e1df9b98b71eb : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64bdab6cce4d490aaec9214001bea01b : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50d34d0dbb4a4b2daa899447f9e7ffb9 : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":2},{"id":"19b5932379049c16c8fc9ef0647081b8cd5fb0f1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 04:57:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9feef40e06f75d617853e6522c66e5c87aba3cdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:58:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f645ad129b19896526edec280316af6d297e866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:43:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b86e89126ed84eee862201f32c403e64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754c74f49735445abce65108b9af819e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaea1dfa37b420588af95e06bc44efd : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6219a78e2f44e78c430b179d0829d1 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4518e8784d0b4e5a8a05bcb132bf19c0 : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad456d8363a04a0bb26a4ce47a1e7175 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b838e81f2c0489d957f4c942a723e77 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/149ce07b6b0c4b0e8f7a1ff3b8765bf4 : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2768056f1124ab2ab79368dd4ad4a0e : SUCCESS in 10m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/44c881b85ac5429b89649f75a72d77ca : SUCCESS in 12m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4f4488686cdd43aaa427b448f574d1a3 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f789d4f52c9049c39eae1b6985a54c8f : SUCCESS in 7m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aed24d922585454d88b0c499cfdd158f : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7be9a262277846da8e904caeba4ae325 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d2f4b260965434daad33b4d76d241dc : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/926d556db93f4417a98d07c6d0379fbb : SUCCESS in 9m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4f509ba66384453da1dba6582b982497 : SUCCESS in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ccb0c68904284e559fb6078308cb2e44 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3fb0dd02b5d741aa920010d3a843b19b : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ee14e15ea89a4c11825a7ee6e7239a5c : SUCCESS in 4m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1e878657f8964b01973d504861e81a12 : SUCCESS in 22m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ff8c1242d3e043f4aa61d12fb6241683 : SUCCESS in 11m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bc252bd4698f46979929535e026ea03f : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5eb697bd533420497e1f5aee3ea21fc : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22e1d7f142804eecaf3eb0e3679907dc : SUCCESS in 21m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a8349c092ea74bad96eb753c7ede0720 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e2c40fca31e04a77874c65159b13bcac : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b67141b96a0a4ae6bf656947f2df6daf : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be0f51b4547846caad00ca0181eca391 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5242180923e64ca4839c5ae2f2c47655 : SUCCESS in 48m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed8517b47f674356bbc7a83cc1b0ed38 : SUCCESS in 1h 15m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ab298fd69c813382d3001bf78549dfd800d18320","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:43:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"70503d3c0effc140d909f63ac3e1133d3d82a7ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:45:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8dd7c45a3a7433b877046f74cd53fd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f7c903ca354f2281cd80bf6c953749 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fe52010a807fa0c261002bf30b8c0c4fb6b761d","revisions":{"aeed067ac0e8f6d205a8f179949405f2e76a72ff":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:04:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/873417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/873417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/873417/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:04:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:04:25.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 3.2.0","message":"update constraint for octavia-lib to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I90e47aff1c44e9ed66931321cec7529aef18cb0e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id73719d16ae18281be674d2be2c5493896b7cc38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeed067ac0e8f6d205a8f179949405f2e76a72ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeed067ac0e8f6d205a8f179949405f2e76a72ff"}]},"branch":"refs/heads/master"},"6fe52010a807fa0c261002bf30b8c0c4fb6b761d":{"kind":"REWORK","_number":2,"created":"2023-02-17 02:43:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/17/873417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/873417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/873417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/873417/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:04:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:43:38.000000000","tz":-360},"subject":"update constraint for octavia-lib to new release 3.2.0","message":"update constraint for octavia-lib to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I90e47aff1c44e9ed66931321cec7529aef18cb0e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id73719d16ae18281be674d2be2c5493896b7cc38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe52010a807fa0c261002bf30b8c0c4fb6b761d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fe52010a807fa0c261002bf30b8c0c4fb6b761d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
