)]}'
{"id":"openstack%2Frequirements~873422","triplet_id":"openstack%2Frequirements~master~I0db380ab57606f2330437d38fff822c4bff3b7df","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-02-17 14:14:21.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I0db380ab57606f2330437d38fff822c4bff3b7df","subject":"update constraint for oslo.versionedobjects to new release 3.1.0","status":"MERGED","created":"2023-02-10 18:22:11.000000000","updated":"2023-02-17 16:31:00.000000000","submitted":"2023-02-17 16:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873422-new-release","meta_rev_id":"621b6e3645706f960d557d85699fa90a458065ed","_number":873422,"virtual_id_number":873422,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 16:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 14:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-17 09:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 14:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-02-17 08:24:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-17 09:44:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"472c56bb59aae6a59736b26724be47b220517325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68b77d108f2b8fd7df8b449b47f4dbbf7211957d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:27:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a008e90bcfa4bc5b905d37997c7398f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f1c70dd6fc444175819d31b21379bf8c : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5183213c4a79805cf97675abfdbef404c3a65082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:16:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85511f61b9ef4ff5981b56ae055b95d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c3a21b9ca24369a5d6140c96611088 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e352d9b02b24292beae8584b95a634b : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78d3b21c8a4b4a3995d5f181d53abc59 : FAILURE in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8eb4406cd78f432b9c7ec9e427504c62 : FAILURE in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed41b962595c42d09c730e9626e2c9c2 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/913b530fe664408eb1ddecf6da4da626 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3abedd272a9428cb73d8e60e3cd9e5c : FAILURE in 12m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/25bf5d4216224710b136cc23a64ba8b5 : FAILURE in 12m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3c3eb78e169b490a9e274cbbd02f2570 : FAILURE in 5m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8f093e85fa314cabb2bd91493930e536 : SUCCESS in 5m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b82e76d2ad714154b626b9bf96b25935 : SUCCESS in 6m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/667545997fa4442aac684d92dfab2999 : SUCCESS in 8m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f359f3f0142b4e2987a85d837bd9b59e : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bc8dfb1a38cf442fa3c1da36fe8210b4 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/adb6fc9fefb04595ad26aa2899ad7a5d : SUCCESS in 8m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2032ee0f35f0425b9875a4e4f4ea9d3e : SUCCESS in 7m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e533e339ef542f79b3feb8fbbcfb1c9 : SUCCESS in 6m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7394b235df8940aaadbef709fe8d468b : SUCCESS in 8m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7be9647219e940ecba8a63a60e1ccfab : SUCCESS in 12m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ed35ce79e82b4f668f29925009bb2842 : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9b2a1bdfbe2e45ff8099644041431954 : SUCCESS in 10m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/daca6bbc97e443d6b17a67cc3113c84d : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/92fe51c096a042218c18c35654956069 : SUCCESS in 9m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/878d251f5a254d61b9e630de705c0bc0 : SUCCESS in 29m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/37c558a42dce4e9092f539ff08bf162f : SUCCESS in 13m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7aa9f2404a8640bc8f20813a1ea504b9 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/911e0c6877ad4ecf989ba9adea98628b : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/037e7b6acb66471e96e922ce486724aa : SUCCESS in 19m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ddfc62a811dd4d8391eedf9fcbb9ac86 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/98b2d084d9ab47159bb60cc5385b2508 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62f3d5040af2425889e702529d52c410 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41d4f802b25e45c1a607623b3259a3c1 : SUCCESS in 25m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c61a6252de148468f0ae1074ff2283f : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57095514abe74cc198aa3e2f016ad8bb : FAILURE in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"061ab315c240126ec64b5222a33709be54a8636e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6d9a9bc879b039d94ebeea546e46974947d72ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:13:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfc900038994a14ac58af308d398b0a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ebe5084f696429a88723321121f3f3b : FAILURE in 14m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b25e56414ab342f94551fab5abaa136b743926b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:37:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9047ac0aad8a419fbae7877a9dd89e48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbb27b63e6e461cab2f645c90b86888 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d225b593594a2694a7b84a75c24434 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ddeb139ff2478ab9770997281463e4 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9ca01205220455eb4db625ad8323285 : FAILURE in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46ef15bb24614c28a1443de9b6b648e1 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e241719c3104cfb99d1bf1500e2f537 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40f7614123d043aeb8bf3e778f77b294 : FAILURE in 7m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bbf849af94834b00baa963ab9f62c69c : FAILURE in 12m 26s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b4619c670f1541249e2a93b823f83dd9 : SUCCESS in 6m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/31bb9eb91e704364b280d55279b3013b : SUCCESS in 7m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ded5a14ad43f45c7bc46262e9bc30561 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b55975f61b7042cfacbbb349a997eeea : SUCCESS in 11m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8fc3ac59fec04592bfb5fcc6cf5f23bc : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/290a184a03dd402db1e03aabf41bbbef : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c07149f5b85f4b7e9796a9b0c02d9a9e : SUCCESS in 9m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/115c9d0841e3478ba31090e19681681b : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57f7276818b34d3f957eae616e56a13b : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fc7cf7e71d904968ab29c7ad446bdcf6 : SUCCESS in 10m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0f6c9880ae004f50a3ca0bb87d347ccc : SUCCESS in 12m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/265c8d9d16f44f7287a6dbdd7ec6c94b : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5498682e94354eb698e5c72d4fa966fd : SUCCESS in 14m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2c56b567410f419f84d15d28210a756e : SUCCESS in 5m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d5ced480c35e410b90523bae1adfd44d : SUCCESS in 9m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/580cbd97ccf14e529c5c69af2281d535 : SUCCESS in 26m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/39d89772513f42a7923ad817a1f34e6f : SUCCESS in 13m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/92e2aeee48be4467b584b056d9fd4671 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d427cc322c6d44f58a4b5543a724060b : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d86858c0ca9945999c02a7eabbf8fffc : SUCCESS in 23m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0e3862dff48f4f2bb22fda60bcdcf3fa : SUCCESS in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5a77c6e4698048e595966eab772e0dca : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8eb0d4ca4194b23bb6da08d7c66da2e : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/34ac4b14d8da4883a1846cc437f0189d : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0148aff8379f4516bd00d0cf5086b80c : SUCCESS in 1h 18m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04edb5f171124e7a908083289db24aa6 : SUCCESS in 1h 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"45421a5cef1c2b2612555f5b72a28b91755e8641","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa570fd4acc58c7c57c631137cabd70e55e32ef1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:23:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bac647c05e248e0945ff89d72340632\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d6686de0420c41f4a57412978a7d280b : FAILURE in 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4f128e8c6b0d51206b775e910b17ac7d6e77f9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f37ef682cb74e8486b16db8233616d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e9832c2e1f477da706d6034b7b8958 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17242ea90a964af2bac16ba83038d5ee : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/769e89d4147840029728e1da01d872c1 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4beabb840d9843869c1b447fd4365ed5 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40b05ee6b59647a6bcf780faaa7e1d79 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67c5e69b7a2c4a3a82d5a3ea3301d044 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7448e8065e604b149c59ff46905f7c30 : SUCCESS in 8m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ab04051793b6448daf6f8d637c1fe431 : SUCCESS in 12m 58s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d888a496cfd54dfdb7dbbb73f36ce380 : SUCCESS in 7m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/908e3594d7f5485d8613ab1f21e9093e : SUCCESS in 5m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/517e92a7f26e43f8ac3d63fd0bf95175 : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/228eccbf7e6a49b28024fc4a0238bb0d : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5efed8c722de4bc5a333284be68dd1c1 : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a9a7b25af7cb467e91a72b9f2b7ccdb1 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/639505ef234b42b5b3666ab13066f33f : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3f02d4ec5f2742e9b1d37fb3fedf93c8 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63170f94788f49158d03c5bd6f4ad8b0 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c4c1118065d94961ada2a5975fee41be : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/966f0e114462441883b500180f0b736b : SUCCESS in 11m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/54ab9c7eff574f91a87c5d876393fb4c : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d27ec79dd4bd42e3add3666a5697a4f7 : SUCCESS in 8m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fc941385b23e40a6b895de1e63f4449e : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6048bcceb8914ab5aca6368af23b62f8 : SUCCESS in 9m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/58898c0afaa34e7caf735ac7f8d54417 : SUCCESS in 25m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4c927019c1594b11841e3b31a3a2be3a : SUCCESS in 16m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1508904209f7475eba857e92015a5896 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd4e42bdeba245f98c04e2a6ba716087 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4150075c6bf84d43ab079648fc31aa2d : SUCCESS in 19m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/78887f2b1cb24e18a0656a9fa84fad4a : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f4c2765994724ca58dc947071e9ff043 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/201874d2000540378766cc4409156775 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/71be5c35ded149418a1986430afb482d : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32f0caaa41454d4d892810ee4f1adae0 : SUCCESS in 1h 03m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/315a42fcf4f941a2a05b12d6da95e3fb : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"61dc8e9938c0c3c6ef8a688057badace064f6f64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3c018b7dc0ccb43cc149fb9d94d73ebea5a51b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 15:13:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41deb33da4c193174fe80c6ad8dc00c26266ff91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:03:51.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 873422,1","accounts_in_message":[],"_revision_number":1},{"id":"ca5d851e68fb2b400913a0bd60b7cda61c7fe9e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:44:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d545cd95ec46c32a9ff1f8ec28568bf9ecb3d4a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:53:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba9438f30724a90a069a97f279d6af9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c6b4f5de1b0641cd939edee9f61c85c3 : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"80e4dc11a72a830a8c5edaa36953059449c974ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:19:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0213cf685b5f42ffbb74a7b3e6e02101\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9ff10be6d14853ac4c3e545924bf4e : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7814b4315f44d09bb64de11c62bb9b : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/970ba60f25fe41f79759806e280f65df : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41be3bac732a403aa411c78d0f66cce7 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8271ae18027492b93bc2c4cffa30cca : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e04c9d1b0dbb4000acb93ae847e8acf7 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9167a5196b994defa5dc64559b84c2f7 : SUCCESS in 8m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f7cd6af2f23a4bba83508baf1d8bfba4 : SUCCESS in 11m 53s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1cb4c9a799a04e9ca5c18a1ca8f2bf01 : SUCCESS in 6m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e42484d8b9f74e76a6ea3465c8f96ea3 : SUCCESS in 5m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/caf20efe3ffd4750801d31f5ed9e213c : SUCCESS in 6m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/63843017871346dfa4fecf6dde0f9e9e : SUCCESS in 12m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7dc31114fef248bf9189ef6b7f4cb918 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/49571f26d5e5403e9b30d75bbdf62441 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f079624830cf43f89d5962fc0056705c : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c4702a2c203e4e0db29b7a811b09f8b4 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f368d8fec074b9b8b18bc4fd03a0cf7 : SUCCESS in 10m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/918be5d1f8674363bd3141e77560f16b : SUCCESS in 9m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/29f88b2305c9430a9077ad0c905d2de7 : SUCCESS in 20m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d6e32cc14c9d4917bd4a26b0a824a6f5 : SUCCESS in 3m 54s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/813cdd5b70954aba8bb4e50744c27181 : FAILURE in 18m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3ce8ee1efa78439b9592de94efc62656 : SUCCESS in 5m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fff2d02a38814c47b0f71aa94d5061a5 : SUCCESS in 9m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2dcf356263e64148bffddc07b6051742 : SUCCESS in 23m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d75f15996e834dc8a51960b36c28ad5d : SUCCESS in 14m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6adfc5830eb1447b9cd697548b71d8c7 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/441c3fc240ae4382bb7f7957cf6084ba : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5e6bfb5977f42f39d4344e193ac7a06 : SUCCESS in 25m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/492d915101b04112b49432bdc4d2eaaa : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a12fd762252e4465adc86e083221f7fe : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bb946d4e4674638ab7b2038031b136f : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b00eeccb17f44a0dbc5833622e5494b0 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e108dd5cb9664c1880c79d2094db4f52 : SUCCESS in 55m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/868200fded10409783171a92319d59da : SUCCESS in 1h 34m 09s","accounts_in_message":[],"_revision_number":2},{"id":"77ad8323689103246f3c938fb7925591da34394a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 08:24:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b076f821c675be82fea92d0c4e8b20cbb9bf9fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:29:59.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6d68f3b6c64335808b4605b18cbd5a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/61da94b2735044dcbdf77c6987a7f89b : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"14add4ba7377d44eeea890093c8e26ab84492b70","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:44:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5b195b743ab7fbb6632703d8ee10f3be4a295c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:09:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cef698d5c3f3469f8821f7bbb11b160e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971ddad5da0648d7a00d723a1765478f : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c2c2f91c134bbf85236c108805a48c : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde4a45112a74d3ea8028eb0e73e85ed : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/662ccded2fb7495b8ff9d9e702448a4e : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e7eb9035dea4f069e2c41503ed774b7 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1092cb79004a42ef83d2696bfb294625 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb0b42d334b24423aa4685dca55c5552 : SUCCESS in 8m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9bbb2d198bb04069afcbaa14d6e875a5 : SUCCESS in 13m 01s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6099e708eb9643ae90d321ad816cdb78 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d252a12a066b439bb2e8f26102150ce1 : SUCCESS in 5m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cdaf578daf8f4ed29fafd99e84ecce9e : SUCCESS in 6m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b5d73b37a5e94ccbb827d7fa114b69ca : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1afc5fe50ea54b8584f52aeecd9da858 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ec5ed457416c446c97a06015da8a4886 : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b02a1ca93aa24106bb92b0b254dfcab3 : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c734431888824b51a0a5322f910df221 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a144612dd294e4c8f8d329bf20c5f00 : SUCCESS in 6m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/60dc306444b94fd7979f1091a40aa865 : SUCCESS in 8m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ae2db20f2a4747b6b8a41a4c132c8dd9 : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e9d81705e9ce4bf8ae7e178d3863c7fd : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4e06ef186b4d46d2bb7a67972925418c : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/89ec3cc3011a4980a1f445cdc94cb12f : SUCCESS in 3m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5cdbf86fd4764c40886af2e165f036bd : SUCCESS in 11m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d9b210e5e363487bb96a1b25f1ba4ea8 : SUCCESS in 24m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a43d41962ef04a73804a342f27195442 : SUCCESS in 11m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/961e4ab169304050945382f5257bd51d : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88f8ce665b5141038ba934bb152da45b : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf9d07fa93cb446097e9095d41ae9da5 : SUCCESS in 24m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bdd9ec093b90494982f75109c26c21ac : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0413447c2ab548a0ba7da8de4c4dd440 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdf1b07b6a8348f79dd29deffcdf27f8 : SUCCESS in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4abbc8b625f54133bb68be97cbf4e250 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/652e9ce729a049c5ad133569fc40a6d2 : SUCCESS in 56m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94e4fdcf4b984491b72ac34f9b1af31c : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7f43a1cede38b862fd8af5b6c18acc85e503c321","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 14:14:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e969323a6361f2c5384938e0b88d5c564062a34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:15:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"061fe9587852f23403b48c0e67002b3ce3e322ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:29:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e28ad5595b4174880369f1ce49233e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b574649fbcb4c2bbbb6ccb482bce744 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb71efed3a3d48fb966dbf16c2ce02ba : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452e0881ef2d441b9be584c64ce52a68 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf40a7512e094e67b751ae98d814cff5 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b67302e91ed4e138fe1b1cce36df3ce : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/debcf51f882e45268701d19d098158cc : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8762ab6dbbdf4153b1f50a4590c63959 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6488ecc66a75489cb4dcb2cfc64f2820 : SUCCESS in 11m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a843d8a446574556952e05c1a4227737 : SUCCESS in 11m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/480d353c141041dcb766264d00029327 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/eac86fb5a3024d609368d3d14fb4fb61 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eb22767ffd2e49c0bdc82523e5d6fa5c : SUCCESS in 8m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2da20176c014411096f0e9e53687d335 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d8068f3d1f44cbb8bb238347aa6317c : SUCCESS in 9m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/56bd7cc5c29f45bf936d038525a9d7af : SUCCESS in 9m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b680e08349864669b18b4d2e453aa0d1 : SUCCESS in 11m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ed266260f4164d5b8a2964cb6b9fa50b : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ff77a80c4d88488c8d0f44f49494eae1 : SUCCESS in 11m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ff394e4439bf4101876b31ac816e2a10 : SUCCESS in 4m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c45e88049bda43af92508c24bdf18569 : SUCCESS in 25m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/69d7012ef8264f2297e0fbd69a1e5376 : SUCCESS in 14m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/abdef672e9394d0a96c55256e861c3ca : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38e7f7448baa4bd790404dbdb4dd6c31 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/458863b9c25c4fd497e64ad190549c21 : SUCCESS in 22m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fae764fac7af438ba03d58a756246d6b : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a4b006b1f19e4ba480133c2ed3151623 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9388554aea46430383e8a9d362657870 : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee4b79e7ad534d9a8e88c5cca63528ef : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f237fa8c6f754cba95c5da6fdf406e55 : SUCCESS in 1h 02m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3083ef5a5064e0faffc1c208c5bc60f : SUCCESS in 2h 00m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8d403a4785f933c3e7bfc0c9b21a56b337b9b74e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"621b6e3645706f960d557d85699fa90a458065ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f053183652ea46e2acec300ca9c7051d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a772491ee34f2dbf45b4b70e8dc2b4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7b0b3ab25004e3d65c6b8a60809c95d93ed91d0","revisions":{"c49ac4cd6dd79236453cac41cb605c5446b70bd8":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:22:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/873422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/873422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/873422/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:22:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:22:09.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.1.0","message":"update constraint for oslo.versionedobjects to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I58bbf4337bc4a5cdee7cd525691ae7084efd1764\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I0db380ab57606f2330437d38fff822c4bff3b7df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c49ac4cd6dd79236453cac41cb605c5446b70bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c49ac4cd6dd79236453cac41cb605c5446b70bd8"}]},"branch":"refs/heads/master"},"a7b0b3ab25004e3d65c6b8a60809c95d93ed91d0":{"kind":"REWORK","_number":2,"created":"2023-02-17 02:44:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/873422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/873422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/873422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/873422/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:22:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:44:05.000000000","tz":-360},"subject":"update constraint for oslo.versionedobjects to new release 3.1.0","message":"update constraint for oslo.versionedobjects to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I58bbf4337bc4a5cdee7cd525691ae7084efd1764\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I0db380ab57606f2330437d38fff822c4bff3b7df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b0b3ab25004e3d65c6b8a60809c95d93ed91d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b0b3ab25004e3d65c6b8a60809c95d93ed91d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
