)]}'
{"id":"openstack%2Frequirements~873423","triplet_id":"openstack%2Frequirements~master~I9330973e7151d301c36f9f9b17285d944ee863aa","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9330973e7151d301c36f9f9b17285d944ee863aa","subject":"update constraint for oslo.reports to new release 3.0.0","status":"MERGED","created":"2023-02-10 18:24:00.000000000","updated":"2023-02-17 02:35:04.000000000","submitted":"2023-02-17 02:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873423-new-release","meta_rev_id":"d93b776f98db34419cb87899e5560e2f5de001f5","_number":873423,"virtual_id_number":873423,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 02:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 00:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 00:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 21:12:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-17 00:56:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3fcd185aa108bef2582b676992f3a0a8943eb2e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a6195ee539675f7a2384ae9cfd39d3790f8f1ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:40:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0fb5cd52ea74885a1feeafd2c827db1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/19192fd2ff944313803cccd5509cb5a7 : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ff86636c6662e417c4544c5c2976de910e154087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 21:12:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f9aff273b0a447b97161dcc59cd6e3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9315ce06234860b80211440126caee : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90faedbedc154c348b168f291cfb029f : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfd99790dd1c412d964eaa73200b52b7 : FAILURE in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ca9de2a06e74514b2640601016508d1 : FAILURE in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d18a51b9059b48a283e314c3da7a3486 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ce87322b7a04beb9e833e2c483f37ea : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ccdf10654d3f4dddbceddc2a3e0c1ae0 : FAILURE in 7m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d49cd8f7f0cb4f5a9197077cad2e90d7 : FAILURE in 12m 43s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/16b2637c2bdd4f3996b0d3c24148228f : SUCCESS in 14m 32s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/65ce15771d314b19af3604dbd5a7ee46 : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e6384f4d94654d79aaf65d9ebc3df623 : SUCCESS in 6m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/666aaa0b3ffd454bb9fc94c5bcf33c9b : SUCCESS in 10m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4afe15e6209f41a1a8785b7f5ee8a84f : FAILURE in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/090372674d1d403abc2abc2190a4c247 : SUCCESS in 6m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a688804ee2494de68423f816ce27bc8a : SUCCESS in 8m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2ead7a71cdda44a292fbfddb6b5f2b23 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9deff865338d44f89dc8fca5ea117f0f : SUCCESS in 10m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0c93d8ec5e82445cb04b789b8771358c : SUCCESS in 8m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/61419a73af544670a5d679a5dbb5be7b : SUCCESS in 9m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/05bbc8ea573f41c3823e374d7bfa9571 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/871168585eb34237bb3ced61c1a667b3 : SUCCESS in 18m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b2fb8f37edd949a48839c271e16f0df7 : SUCCESS in 5m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6dbfb995d8c54c1ca7c6f39f93d63b20 : SUCCESS in 10m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/12b1ed229ea744fea4bfc1b8f3cc009f : SUCCESS in 24m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/162686ed4c0f44bd9219f74e0cecc2e3 : SUCCESS in 11m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e4c16734e8154020859b6183a80959ca : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30fc6d3fa3784e568cc371e3f578d979 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6798925f60349cf831279394cbd94cf : SUCCESS in 21m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dd4529e7aa9b4a36b5a5ab7fbc0d18d1 : SUCCESS in 4m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b67991b326bd43eeb263891abcedf1e6 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e0c44e134bd4133869025c44b5f47ac : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/404984b74eea4850824445f2f3fc9ea3 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e892bf251114cc69e9d95c8c0e8ab64 : SUCCESS in 44m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2130da99d37c4f6f9bbe337eda051b7f : SUCCESS in 1h 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1d8b5ef02d81f68ff3da24ee8e07cdda906a58e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9bdc94b8515722d675f9cef9b7f112372214ed88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:14:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e4067975158479186fe0210700f97d3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ecd619ed40004411a02e29c20ca433a2 : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f0023cf10e436bc68fe21f245e1e19e1f9018216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:57:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2610eeb17df44f45880214154e21ed77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93d5f320d2d4043960b54475df69ffb : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3a975a69914022b9603b1973f9e33b : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bcb4082eeac48c99f89111e5fa6ecf1 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a6c6aa812e4e1f99e25a730a77ae59 : FAILURE in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dea68cfe60674bd68b942f98354b4dcf : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75e7cd823b654c90b0b9c48c91aba74c : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0433689d9c8f4dfb9df54dbd43989e45 : FAILURE in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/54b41b945c90443386e58662aa550380 : FAILURE in 13m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7c992c79f03e4b6383a9bdaa84ed2991 : SUCCESS in 7m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bef44b707bd9460ba948152d271dcbc9 : SUCCESS in 6m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/74e64e504b934ce0845008f13c871861 : SUCCESS in 6m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/25a7ae9658af43bba92af0726587d4d4 : SUCCESS in 9m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aef975229e344a078ed6531e1fccc353 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/42c0f208b47e47cdad343d6c7ca0aee2 : SUCCESS in 4m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/06f6342f8a174382bf73a4f1947755a7 : SUCCESS in 12m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4ee50567ced247d7b74d4f7abcc34829 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50ded76344434c82becc06fb54c5d87b : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fb3056b2e7a249109db640180c4d8dbe : SUCCESS in 9m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cf885eebe4fe4595808cc2a057a37860 : SUCCESS in 11m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/966d4a896b354fe18ea0d3ece7ca71c2 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/63aab4c7d2e34d58b2d7c9e107150487 : SUCCESS in 9m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aec536bd912348ceb9c015d08c572c45 : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3966546195754b33af9eda1f8af2dc62 : SUCCESS in 12m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8e0c7c39ede94c78b0839a061b29b131 : SUCCESS in 28m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c47f2d32dbcb43eeb4ab3e2cb015cc82 : SUCCESS in 14m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f2c62857384d46ce91619a7eac2afea3 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cdac9d9a421a4843affbd209efbc0fb6 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66ae1359fd8649eea94cebbad74e70fa : SUCCESS in 25m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/10e6356619674b58a22730845b7a2192 : SUCCESS in 4m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/feba85d71dcf4a8b8a74c4c4d05c4b46 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/caba8933e53b43d9aa7df3ae8b49ce1a : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/464a74933ffa4c258c91eaaaaec8803a : SUCCESS in 27m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bd51958943b47c8816a21b79e4e53b1 : SUCCESS in 49m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa85571a54aa490fa11615ff19d2c939 : SUCCESS in 1h 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"597746f748fb1eb2162e95c2439a42e7fd1c7775","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97dd686aff3f00b3ebffba29069bae3049915efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:22:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908ab3caad3a4c0fb6f6c3bc4037b100\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/27f9d542832148b48b5b2baee8fabdf6 : FAILURE in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9dfbc26bab7416442e0545e4a1379e7166dfde49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:41:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/955791790aad46cc805558e4c4ca1a36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a417184008349f399f35289c24845b8 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd10a9f82a014a79ae3b4a2db9a07fa2 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a578d4eb050e44c784d4732c51122fbb : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/171bbe88bef74d8f90d0c5e3a0856dd5 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5324d6d68d4445190ac6519823b7cb8 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68fab3d8565c44789c1af8754a2bc7cc : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6033d6fd682d4d45bbdad67c9145b8e1 : SUCCESS in 8m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2e25699fbfe649b789c0086566b5ef26 : SUCCESS in 12m 36s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c805a2acf5204898b3341377783c2c23 : SUCCESS in 7m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0e0f4464878c4e9cb51ea015884bccf8 : SUCCESS in 5m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c23b47adaed74a429a0caf10670653c1 : SUCCESS in 5m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aa4e1437f92f409bbeeac8c15d8aecfa : SUCCESS in 11m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/369de3f11ed14cee892ffe6c4a4774a3 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/31b676e2546d44058477320caeac0732 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bd9fdc417a7e416794989087364f6e77 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1a74cf86535c44edbcbe9e08b3d151d5 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c431b2866d014314a89ca765646e2b87 : SUCCESS in 8m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4cc258932b0d476eb7254c3a817c067b : SUCCESS in 9m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/74c28b0af751406b9a1aa57ec8e6e97f : SUCCESS in 32m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d5124d5e89fb4583a4af594cb2921d4d : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/78a0548ed22344a8bdd9a5de51d10f24 : FAILURE in 17m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/398bb3776ca540998a686f120e1493ef : SUCCESS in 5m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/99010c8f26d14c09900a32a665434451 : SUCCESS in 9m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e41945d9a6f40dc959f3741837c9e96 : SUCCESS in 23m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d21c8c0ae8514e8e84e6f97adcd0ce10 : SUCCESS in 12m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3f9c29af795146dbb33c2214fba67a21 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71eb6f8142df42ff9b5554e7195bc52f : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/806b6e3ff1bf4dd0bca9f51bf718e4b0 : SUCCESS in 22m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f853722180854b898c552ada63a96999 : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fa73cf646567463ebf7c762ef18f861c : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b61363a8c27241b89a6df0fe855110ce : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a3464f811c294c2780d5617f6fc2fc98 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b28e04533f174bc2a21772e3f9f7079e : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29b0fa85c665444fa69bc00c92fb3ffc : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c4a6e000cbcc3b1d110c0e19acee887960d979c9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c26aa19d056e9aef008f91eb2adc869a36ef2d57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:41:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6bc5832dfb446e8b9ce41168b9e8c52\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2c7bb9b797d04c4788a720bfc422763c : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6efc1826a96974b9920c07250fa3aba4be3633d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 23:17:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d4f991d067408f8a3ae695f3620317\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ca71104cb64bc796718ba5620b60c6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b5b8afcc714aecae54e5b9bb158b30 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfb9ccb8fad44f2a98a69607c2f7143 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/017e9d6d2e6e49d7ae4bd34580d139f9 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e245ec8da7394078acb87e4e99523aed : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63ef5c5aa04e4d93b6709ab6bb1dfb3e : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/acdbfc01eb984aa48036b992af6ac67a : SUCCESS in 7m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b154d07dd8044ef68f813646f76e3cd8 : SUCCESS in 11m 54s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4fc5021760944ea08863c15214d45e5b : SUCCESS in 7m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d2007193934e45d1b9b0de906477f2c7 : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7fbfcadb0ccf408fa5c683d3a8ed9158 : SUCCESS in 6m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f250d3b92f864732b6057998625f9b41 : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/632750a382dd4bc98448c752169a43c0 : SUCCESS in 4m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ce6e1afab71344f49d081611a1509dc4 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a675a53e46424fd4add870e24dff75ce : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/05527cb13ef04feaafd91c8b1f520867 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65f0f88f60674711af5022cfd575ba33 : SUCCESS in 8m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/339bbb5d42f14bb59cd30ed9783e93d6 : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/810b3fe2c2af40bdbb33bf75f88adbb0 : SUCCESS in 10m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/331696d08c2b404aac77217935d53b85 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9acbf9d01b924f94b1f9755409378887 : SUCCESS in 10m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f5ab8c64c5cb4105a199dced58b1acef : SUCCESS in 4m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9401d4229e8d4f3f93f24acbde0feed2 : SUCCESS in 10m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a90bac5717434baabd6550270caa2069 : SUCCESS in 27m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c849f7994d1642b493458bce9b34fa0c : SUCCESS in 11m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/14326a6af8274b4e92fdaffb3cabb704 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68792da3d1924424aa29095bd5468c14 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a75065fb72942c29bcd16e9b5377b79 : SUCCESS in 21m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/74de44e539084938a101c0cfa2c1f784 : SUCCESS in 6m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dc4aeb82141e4047b3eea4025da82004 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab47fe4d91d740469c666688db1bbbdd : SUCCESS in 6m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e96bd2c2a96f489eb0591febc7d33cc7 : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92ae0ec5a0ae4f92b907b2c89ea1a139 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf4dd1a98c91473db4118e8b4a23e0ce : SUCCESS in 1h 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"96d5d97fb0535f13812a354b14b798b61aeb60e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 00:56:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5caddb907acb98ed88aac76372a34741c30a9f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:57:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e23495c105809233b7e1b18c745374f8227afda4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:35:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf71de624574347a39e2d8c585f71d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310c1e69a45d4545b7508b4f1ffd0edd : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ddc2fa7156a4524b2599c8e9a70cb5d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d212994fb13e45f0ad8aaa7eace82bd8 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6017725d9314f6ab22d1373516ad4cb : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac5634fab8494d529c979f12b119caf8 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e04cd90f85a04a4abba3f19886ca3f51 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/43e7cff8546c49f4be9c04524abd2dc7 : SUCCESS in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6442c3221b2c4c3097f538c04a22d09c : SUCCESS in 12m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f5d892277e7b4e348d7755dcd26dcdd9 : SUCCESS in 9m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/318ab3ec26354613bf1b59c20e58fcc2 : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/558c2cb1443f481a990dc411d91f370b : SUCCESS in 6m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4bee7bf8a28c4dd9a9ca55856d24bd43 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/850acc6093b34659a8deea5e5e9aad55 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7958de9db1d640028387b74aaa0f960c : SUCCESS in 9m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d8e5209a3dc946789b92389b91e20165 : SUCCESS in 9m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ee8460e95de34df0a328c6b8bc5fa3b8 : SUCCESS in 11m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2f48253b6e044ce8bff2572c42b3ea69 : SUCCESS in 5m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dad40849f88349a480ba60016241ebbd : SUCCESS in 10m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4c4062af4db6484b823353de4600d35e : SUCCESS in 4m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c65f82f25ff74e80b0ec25c7b3e5d2e4 : SUCCESS in 26m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c3bfeebdb14541f99bbc6d5a3fbe2274 : SUCCESS in 13m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/226b94c06a154c24a16e7b64707bcb01 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4485a539708345d58c7fe5abf867dc42 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/941c5a509a544f099cf684965f1aaa0a : SUCCESS in 23m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/74ed5cb3983e456dbfbd1b1fe00f47b8 : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5f52f2aa96b843a68e17bc50ff4be4dc : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79adc6b782c9465b89bbaa5e8c80a037 : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b3cc3ef7462d4e96bff208ad2d4763b7 : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/844beb47ea8e4770ac1c2c6a717090d5 : SUCCESS in 1h 06m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca681ec60a0349508ca90db4d083ce82 : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d93b776f98db34419cb87899e5560e2f5de001f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:35:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e2c8869703236c6908815f9315b769d8899b78","revisions":{"34e2c8869703236c6908815f9315b769d8899b78":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:24:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/873423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/873423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/873423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/873423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/873423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/873423/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:24:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:24:00.000000000","tz":0},"subject":"update constraint for oslo.reports to new release 3.0.0","message":"update constraint for oslo.reports to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I88d8a8edb246cd0813201ff0491f87f39ed4be3a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9330973e7151d301c36f9f9b17285d944ee863aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34e2c8869703236c6908815f9315b769d8899b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34e2c8869703236c6908815f9315b769d8899b78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
