)]}'
{"id":"openstack%2Frequirements~873424","triplet_id":"openstack%2Frequirements~master~I8cd989f416a973762440fe94fc7565d24e5a2b94","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8cd989f416a973762440fe94fc7565d24e5a2b94","subject":"update constraint for oslo.middleware to new release 5.1.0","status":"MERGED","created":"2023-02-10 18:25:09.000000000","updated":"2023-02-16 20:03:53.000000000","submitted":"2023-02-16 20:03:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873424-new-release","meta_rev_id":"0afd5ea52e63cc04cc02b4b7a9012c94c20d139a","_number":873424,"virtual_id_number":873424,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 20:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"24ac6310a51597751d4d1729bc4f1babbe26e386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 18:25:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ee811066608da1269af4d215816f9fe8c43f223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:31:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1d1b3ef08242d9a37473956d0d620f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/027a1e9ee10c4e8585f952176edf0b7e : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1cf4df0d90d107eb465d96cdd5a29b38ea521ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:42:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8e9a9bd70a44db28bb82266b8fb502e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76e8a8ddce349f697c35f1567f5f0b4 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9205c3a99094e14b02733560664fcb3 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3ce1dc1ff494ff4a78c8bd3c9da0ad4 : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4408d99da5ac404090528170c92f079b : FAILURE in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/169110b4730747fc9ae16e94e1a9d37a : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c999970dbab4db7b5b0183dcc2cd29b : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8aaf2b8edf0e4d67849136dc2efc6b24 : FAILURE in 8m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56c06e58a2974c6999f272e06fa0da1a : FAILURE in 11m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9f86b526b1ac495585eefdf451c0d670 : SUCCESS in 8m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a350e72c9a9946c791d76aa0460dd55d : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bd45a8f5c0db467bb6f245f502f406dc : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ac2235af1e2042bf92244a82ba66c31c : SUCCESS in 12m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/62239f566adb4088aabc6680e185ff5a : SUCCESS in 4m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6e09a423239e42569f477337bccc0fa7 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/054596d4e54748eea12824ec93901af2 : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/60f497d9c9554692b25cd1e5c3b89268 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9df8e2d683041d8a7fc09d77cce5076 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ed0b4bb5af7a499c963031f27d610b28 : SUCCESS in 10m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/49783d1fc72541d0bc14075570653cd1 : SUCCESS in 11m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1bcc38607353454381292f382865c160 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b45fbc022db742f7a361a680b4090ba1 : SUCCESS in 9m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/253ed504306f4ad29d0201ac4078bd13 : SUCCESS in 5m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9e855b4644534796834dda1a5ad46215 : SUCCESS in 9m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/792efdd417204b36b0870fe2fe866151 : SUCCESS in 29m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/13c02095b26b42fa8d8d1d0fa68bf022 : SUCCESS in 12m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/64c5734cab85421480339d1000a4c98d : SUCCESS in 9m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9718a15d08ff472cad4b7636f58cc9a7 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b40dc18db89d44a08981798cc243a81d : SUCCESS in 21m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/06eebd316fb84a76bc8605b4197835d5 : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5de5fa06548f464b957fb18e7b603801 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fead7c4c1fe843f6a3493a6411fec563 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2daf1202e7940e2b07e64429c564238 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea7486d1cb594911af9cf379dc274174 : SUCCESS in 56m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f198add421473e8214e83c20a1591d : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"02219826079e818d27cb8cf91e5bb68968f1cc33","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b47e65bc388149e3560135cab20b025206728d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:25:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3579c651a84eba976ecac539b2d381\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/09770575f9f14ab4acf3e7ad3caf7944 : FAILURE in 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ced55aec5c0d0771fc3586de75637c26571b0617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:23:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0396ce2e57a4627b031871e5eb2d411\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5662c68147f24c99b8a96699556477c3 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189445f7de31499d8dcfe6f72702c1e9 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db37b314a881466a8cc046814bdd1c92 : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08c6633b4999499eb996ca85241ecfcc : FAILURE in 2m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/862114a3d1d94763816aa23df307429b : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9c1966871a84f21b96be2b7176e4a8e : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0c92115d57b45bb963afef0d9e8fb28 : FAILURE in 7m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c328c97d5954f938e1f29c4743e54a1 : FAILURE in 14m 10s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/52415491dbc94120a82a70eaa051c2f3 : SUCCESS in 7m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/96fccaba5ea742de9c6400324f0813e0 : SUCCESS in 4m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/050a8622666c4e4883a3273a5d6bd41c : SUCCESS in 6m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/87206bf33ace47f187c45aaa41e62d71 : SUCCESS in 10m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/01977e65872a4324a79b8ff910164c1d : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ba537dc36e8247b08957e3fc825efa3f : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/337cb62c77984d929f8c34ad40bc24f8 : SUCCESS in 9m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8601fd48bc0f40b2bda6d656cd6cf1b2 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0a9311edebe4d4fa88bc8412417356e : SUCCESS in 8m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/25f39329276f4b8f9d46c05983b9eb7f : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3a286085607f4d57a5a785ddf2e732de : SUCCESS in 10m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9e69f22fb81b4e149b8e1e90ca153b3a : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9dbec355932242a789891f2af25124e8 : FAILURE in 18m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad3ebf2e167d4db389b08748dbea8d7e : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ef8b4f258b7541cca7bbc1f47c7cbdef : SUCCESS in 10m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/83b74de581ca414091af027797bb57bd : SUCCESS in 28m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aab9f2e4369d4fd3b18bbc0cc61889c2 : SUCCESS in 11m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/27afdcaff7fd4022bbec9b34c3793b3a : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3454897d87c4684bd688a7059145874 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d77f7ef915ff43beb9a9a5b60bf836bc : SUCCESS in 21m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/edad5f9e00214373971085fce3b1c00e : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/450daa5879dc452d87e3b405f633a95c : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f79fdb0cd8e94b0c9cc5277ec0df0d65 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee3f3681cd7c437db92c1ff02cb897f2 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b25f6d8a16e4f8c8ea78948d07dc5b7 : SUCCESS in 52m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca50fa45d3c84173999f54d803e480d1 : FAILURE in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ce696a138ddbeaacda741eeb176a0092a142b883","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdcd9d7576f24ca48a213bd6166615dac492c158","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:24:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92208da1e4d44cdbb30588c84f52b2e7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cdaad8a616c4495aa8a401716921720a : FAILURE in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a5ef3dc40757e64675a1ea78d0f8d39bf4b11ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:46:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7642a5d23ecd4653b37ae0edae693926\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5a6705840141309ca0d02f0773ca19 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa011cdc20245dd85cce25fb7dfbb9a : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4fc7c4bc2e4ebb8caa19d5b5cb9ee0 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf4e8f4161b444d287f1f6a1572780b3 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6e886d3d690428885f308f4221d3985 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f4ac1c5422d436a8f02e657b8c32935 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/adad1c830ecc4fd0aca5666c4a1a911b : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b04f342b0f9e4ed88048d1c14c139f94 : SUCCESS in 13m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8e73a5012b5647eb8e0b63b0fb8405e9 : SUCCESS in 7m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/99df04186546457cb162a93335854c06 : SUCCESS in 4m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6c58054646e04785967159bf5e70e467 : SUCCESS in 7m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f42b02b8b761444eafacb1ebee73eab5 : SUCCESS in 10m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4f330f923f1c4a80913a8ccdbb2b8dcf : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1499ddc5408c488eb117ce0512353f91 : SUCCESS in 6m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7106fa47bef54bb797c46ae1c88f8868 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3bbdd7b57883464598c17afa59d23174 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/602dc40b4a6a42f596820a8304ba2f85 : SUCCESS in 7m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a566ddce60fb4bd8b52f8fb3e1ae2c0f : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ea8fd6be66cf4e6a9736ad3b8341ea03 : SUCCESS in 10m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f546df08e954429498556c625df9a78d : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d441baabf7044365bb61616809c4967d : SUCCESS in 9m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9a01ec2741fe46a6a3030fef0cb23a46 : SUCCESS in 4m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/afff3d2601d44bcead2b356d51ba6fa7 : SUCCESS in 17m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/054bef4ae793438abb7d82ad8bba6e82 : SUCCESS in 23m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b834f89d8af04da1afad4218b7d98aba : SUCCESS in 11m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/752b9d5b8b854802a3e68f3806d02cae : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/895576e86baf4db9a7363b4b5acfef7e : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6405440637124f2aa65cb80ce1582f5a : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b6938da764964e1b83276fd02407ca8c : SUCCESS in 4m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a169ba069dd4e36a2df471ea1a1fb18 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/816f8df59b8e4a0c8b6edd04eb132a35 : SUCCESS in 6m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/de6a35f2429d4bb59f6555eb154f3668 : SUCCESS in 27m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7348577b98f94b0fab664b2670ebeadf : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75841095deb44eb59a529dcdc5944f3e : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ec4c50f1396101be6d86975d424a75a4be736a4d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b0e17273b66e61ac0139fab8b0eb1d5f1ca8562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:33:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2c3d4136949c956f136437a44b78f2df4a6248c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:03:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d96ba77f3314798bca7da7cdc1d5807\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d791a2c8158b4ad5af405036ddd23f40 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811b29b42db844c3af05dadaa7db3460 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09b57e816dbf47c7b99485c49e17fce8 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df8f2336385c4597aad58cae2c040544 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c11ee31ebcee42a887e97c567357862f : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df24e442b01f4356a35f90a264c15fce : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72fd61dba2b746eaa1e12db2562058e7 : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7a9f085ae1a4429b4152da76d2727a9 : SUCCESS in 13m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/34241a04f29243ad8d8390245cef29b3 : SUCCESS in 13m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dd5691255e9e41bebf3d500dd344829f : SUCCESS in 5m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dc675195961e417caefbeea94f08b0c8 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2e860a8805f4f54819933a639e9402d : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3336b837cb4448dc88f7bde9c6a38428 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fe2f899b33f4a54a90824cd17d4a41b : SUCCESS in 9m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0ee91c9afa844e2ab420a5c748ca3c67 : SUCCESS in 10m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1dd9ecf207214286a0020f5d716ff734 : SUCCESS in 14m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/22398a2ea7554ba08f6b8c130bd6f35c : SUCCESS in 4m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6e043736fedc46168c274b2cb083a83d : SUCCESS in 9m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/12cae9be9a434a4f89acbbeb81fcfd1e : SUCCESS in 4m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d72cd78dd4e140cba208c07edde117cd : SUCCESS in 24m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1f177ae8732241eb83f6edfc0b25f10d : SUCCESS in 12m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b7c2bced028e45dcb4008dadc9626d90 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/569a6435a8a343ca9feca89f638d11ec : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2edf6f0b48e44c09a4c66ddbf7aed7c : SUCCESS in 22m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/457ccd693d7e45a58579cbfdb46da8c4 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7d990a97774b497897a1baead2e297d3 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd25264a553149e99f3fd92951b304f7 : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7d9afefcc8f4b2ea5b0d7085f02d5f3 : SUCCESS in 21m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa079d8ca48649829aea2737473eb348 : SUCCESS in 58m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a69ceea6ea14c498ce3698bd14bac0a : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0afd5ea52e63cc04cc02b4b7a9012c94c20d139a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:03:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9979f84b63c5c14db319919280245e50676a3163","revisions":{"9979f84b63c5c14db319919280245e50676a3163":{"kind":"REWORK","_number":1,"created":"2023-02-10 18:25:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/873424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/873424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/873424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/873424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/873424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/873424/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:25:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 18:25:08.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 5.1.0","message":"update constraint for oslo.middleware to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic5b6b774857e099860513bc3570ac6af707ded29\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I8cd989f416a973762440fe94fc7565d24e5a2b94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9979f84b63c5c14db319919280245e50676a3163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9979f84b63c5c14db319919280245e50676a3163"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
