)]}'
{"id":"openstack%2Frequirements~873598","triplet_id":"openstack%2Frequirements~master~I1d25eb3fc1d26ee3dfb6336f4687a7583d0fbed6","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-02-23 18:40:26.000000000","reason":"removed on reply"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-02-23 21:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d25eb3fc1d26ee3dfb6336f4687a7583d0fbed6","subject":"update constraint for openstacksdk to new release 1.0.1","status":"MERGED","created":"2023-02-13 16:06:20.000000000","updated":"2023-02-23 21:38:27.000000000","submitted":"2023-02-23 21:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873598-new-release","meta_rev_id":"6e0f6f1e554601e6ea44aa39d39d939bd89a73c4","_number":873598,"virtual_id_number":873598,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-23 21:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-23 15:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2023-02-23 14:21:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-23 18:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-23 14:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-02-23 18:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 16:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 17:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 15:45:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-02-15 18:08:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-16 09:00:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-02-17 09:43:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-23 18:40:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8ce89d01502f878736e4b3a0509840b54e0bcd84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-13 16:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d2524928e942c0606dae3a23b6f283e9a0ca86c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 16:12:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72503898e267470dbb2bc09282450044\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a89cf917690a45959d6fbb4d6fed0264 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0f966b8f33501118eaeaeae05cc37878829fc2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:42:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ef98abf31524da8b3b5c58053c0328b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7531798437824d86acd1a7231f1f71ce : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1c1897a470453581ee278aa4026fe1 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/832a6c442c8a499f9c4c946447a192ef : FAILURE in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d379b5cb5bcd443196c7b35c49a67c76 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd431f4f2c41418abc904738f4959ec3 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/637ab26c23a94e50a7df2291729c864e : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a2634b166584e4191df0dfe5ac656b8 : FAILURE in 12m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/094fd7cb7c9b42039e410982ae431794 : FAILURE in 11m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/04b1832e032e4ff9a3d8cb86c3a7fe1a : SUCCESS in 6m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f81a01ba6d974775b2924d6f65221b91 : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f4947e0eb575412b96be7634c1b24621 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8cbe7c4cd74d43608268b5ced275d832 : SUCCESS in 11m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b266a0b598c44f3eb54441b65f03ca77 : SUCCESS in 3m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ec9c222d2a28462a9953fac0acfb66e7 : SUCCESS in 4m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4a61e4d11c4241949ddd307e624f228c : SUCCESS in 13m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7503efde9747473d88c516516a85c10a : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3caaff8e18f445069cff9be05fc41889 : SUCCESS in 6m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/744dfe41128c4c51bc7b5648a9fca0b8 : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2498ab8c55f24a44a4b2a9ad2c99b3f8 : SUCCESS in 20m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/24ad66a5c7724015830d8021ed50f2f7 : FAILURE in 3m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a63ae3e1100e4ca9badb57a82fbb4fa1 : SUCCESS in 8m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/90586a2e0ced42c4b58007c45864e6ef : SUCCESS in 4m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e7a0e6195bf44932a5a776639232ce24 : SUCCESS in 8m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b023997c73f54253b2845f6af0103dbd : SUCCESS in 24m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ea36ac6103654217911ac8f233e2a774 : SUCCESS in 13m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0d815b0c80d64248a07628c8f9a30b9f : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91ed736616094e5b92ea513af4caaa02 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b090c23811ce483da8414628e8c49f33 : SUCCESS in 20m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/832cbabaaa834f86b877189534221115 : SUCCESS in 5m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6081ba09ee60466b9d3a1d1f61deb700 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/041ead3a1db9424ca0f32836060e4a08 : FAILURE in 4m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0c0e0c71dfd543c8a4f6ed81ff771068 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1316d8ca2f704669a401e48847161d5d : SUCCESS in 46m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcc1932b264443abbba0ccc7eafb86ee : SUCCESS in 1h 30m 28s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"6bf894881b868da867b8e1abe4a61028efdb757d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b2940f429b17f6dc06bc2e423ee5b69afce555b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:28:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28fa5658035b43d6ba39208ba68abfd8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9401675161534dd7a21b2b15b2fc8d61 : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e0b9be4ca3b80d8ebd830a8ce6c8d028ae2df678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:01:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a0ab4994f474447bde048567d72556a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8528e5296a64798b8b84cba3651cd95 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a337c0f2b345b4b0144978ad137d9d : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b6e4dc5ffd4091812dce82097179bf : FAILURE in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca4ffaf821aa4172a8c98f2c3eafb996 : FAILURE in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3d9666f85864f71976f7a70376533ba : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7146e3e1baa54122a6e8582b8c130d8d : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0abe03baec07454f85f92b5f5676b162 : FAILURE in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0bfdadea88114cbb90959173666b0de4 : FAILURE in 12m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bde0e6b4d42b4729b4d727b51eb9bc32 : SUCCESS in 7m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ffa6d0e95ba94995a35b8338dc9e7560 : SUCCESS in 5m 13s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/01efa75ed73e4a2fb5dde617d97400bc : SUCCESS in 5m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b6a390debf904a6ab4fe5e572b4d95a1 : SUCCESS in 10m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2e79e05f3cd64818b5fdaa02156754fc : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6157b5280b46456d92f28f31d55120e7 : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cc7b0aa3bb7f44a398a3abe8d98bbe31 : SUCCESS in 7m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/53a981254aa74d779c05c77edcc094b1 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bcfbdaad79d4261999d1913b7658367 : SUCCESS in 9m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2e68c776b3d24c4c8d4393482a90223b : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/036cbcd6a58549b1910752e023d05e12 : SUCCESS in 25m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/921be9926a5e4fe49706a44caa3f06e0 : FAILURE in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/20d5a20b1d744b0283f537fbca492bc9 : FAILURE in 19m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f4c25935922c48869bb61efb37e04ad8 : SUCCESS in 6m 33s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8e2d572b9c9d4202bdfd7adfcb480f19 : SUCCESS in 10m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a463f946607443c4b00bd82e21415bda : SUCCESS in 27m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/417fcc6496f1474ca177a11a49eb68d7 : SUCCESS in 15m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9c75ad2549d946dabca8db85ad6f5f9c : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc161b5d19bd4c8d972fef71d1e86a4d : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1bbc170119a14b86a86bb8ac2485fd1f : SUCCESS in 22m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8b1bacef6d3b41e0b7276684398e68ae : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3ec90e911b5040d28fc9997b60a19b4d : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a80a61926b9b46bca72cc9231fc00fc4 : FAILURE in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e76048e8c2b44ada2145e8cbd335a66 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/951b190e1194402cbc316a12aace1844 : SUCCESS in 47m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d314d27ea74473ba819145a14570b0f : SUCCESS in 1h 37m 55s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"17aaf22bc3098610a658bdaba9baa6f5cc853ca4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-15 15:45:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cd44705f01ba7245ab0b2091e302f5b6d53a787","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:08:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dae6d4088b2b21a5dc0f348fcb9a29d7f8f8d23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:22:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01a8c8d9356428184c8f1cc769fecb5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f58c5c36c924dd692f858ed0f35ab6c : FAILURE in 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"70c8c52f71f4cf19dffbfaa18fd9bc88b525fe8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:20:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9bff6d6661d4cb88193bea1dd211b62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6df903872646fdadafcca60a9fb71a : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728dbdb0ac214d3db21c23c9b7aa0aa1 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd99a840faf04c1bbda19e56d9e17f63 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f3975b7df97435fa8de44847f6438aa : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/549166b9ce4f44319d4edb161699e4b6 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2258485264154ff6a35a7ad522c1a3e4 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85faa007020a44c0ae94efc7baa19d76 : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69e756b4656842cba2a83f949b56a118 : SUCCESS in 10m 48s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bf239972497d4736a9ff18df88667125 : SUCCESS in 7m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/03fd01be27b74632b0ffaaa6e40abcc0 : SUCCESS in 6m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b0cc00606a0d4f20aee0bc4f71b5066d : SUCCESS in 6m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1f46e106662f49898ee6a0efb325bb46 : SUCCESS in 10m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8313c233af434d71935f7094480e4967 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/619b3e1e4e9342f6941fccbb2f47b403 : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4cf4301a9ceb4d43819d5808774667f9 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/17f30cbc6cbf4c1f9ad6b2e745d897ef : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b55b04ecb9de4fc18ee0fd5ffca2d54a : SUCCESS in 6m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7f10d651da824568b889e970b8a7dba0 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1b2c7f2adff246808165469465f76054 : SUCCESS in 10m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/373dca068aad4c3cb3e561311245f804 : FAILURE in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3fa6b62971ef4fc68566942da7488570 : SUCCESS in 10m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0546e623f9e34f9d87f90a56f478f2c7 : SUCCESS in 5m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ae26d66e2fdc4456a499e9f93773ab30 : SUCCESS in 12m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/aacc1b6310384322b48cf2d30be5cb6a : SUCCESS in 27m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b0205b94333e4c2290b7bfb010cb0541 : SUCCESS in 13m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/78e17af0788b405387fab81a346c73a4 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7870e9ce06f487d939eebd9d26f923f : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b3ad64317a845ea8d1c115a9dbbae0c : SUCCESS in 29m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d7224c3702b746ec80f957d8a8758c77 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/13066f4f18a145caa83804a49c11859b : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bef34d237b704515adec42ea6996427c : FAILURE in 4m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/18b4e8b33c814a7dbd103cafa96f44be : SUCCESS in 21m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/328fc3e3f4cf44f79140dd27028978de : SUCCESS in 1h 10m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1f66338922e49908e6a3bd35a51bb26 : SUCCESS in 1h 25m 06s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"ea8b7ed3238abaa391401e56d7b481947399c85c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4dccdaaf457cbb345c866b2de9db439985c5b3d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:34:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f82a072e48654c8b9dd2ffd6d815c0a9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1ce3cf211ff41e8884f9590e5324202 : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b366822e0d90621af9f1d822289644ce6cf9d316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:05:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4d1686fed3f446ea7cb4d88f0ab7b26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16723d8545b94d11b0b1cbdce1ec77cc : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67a09f62927409681ff9a34bec26388 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/854517e307e54270985186e2d9277329 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e5c6960bc074856b30982e1afc524ec : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b13f5bda1e2148a1813029eb48794f0b : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b156908ed4e4b4da8aeef4823380dca : SUCCESS in 5m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/131787db694b4c3ca0a2bf1304ffbe98 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9412470afa5842ebb8911cb97f9812c1 : SUCCESS in 13m 04s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/63f1a20ad3d049a4a0a0dd56de92016d : SUCCESS in 6m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/872be40fd9ae48bd97de47c28924bf70 : SUCCESS in 8m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1caacabee42740d28ab6dea1870795b6 : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0c5f248d6bbf4183b725a484c08714e9 : SUCCESS in 12m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cd865b0b572343f685e63bbbca300af1 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/eff28b51daf54227acb118f8fbc4280c : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d5dd9b3e4dbc4c7d806e07e52475f84c : SUCCESS in 14m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ad4e1c4c8ed046709a85682701e2e2fe : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/947da652c506486f86bc9ab6dae80ca9 : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6b505f3ba1c74b608cebcd7a8653f824 : SUCCESS in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1214d0f18e804113a167f34c0ccddca4 : SUCCESS in 30m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c0ee17efd55c4b4892e2506f015fab24 : FAILURE in 5m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a69e164bca86486fb97434d4a2318d4b : SUCCESS in 8m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7c23c1a54058477e9f9031473e1a1fe9 : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/01411137d53140f19f3daec0768050ff : SUCCESS in 10m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0884afbab56f4e2490e1dc6e8c428f75 : SUCCESS in 23m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ee19ed97e82f48d7b717901e9e1c4d4b : SUCCESS in 13m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/703150faeac145b6ab5ea678cca706b2 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1daeb620449f47ff8a4289484db54606 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d1d434b8ee24aefba6e5a8d35decb35 : SUCCESS in 23m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9bee139d7c0143349da0babaf58b8650 : SUCCESS in 6m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dd772570c5944ab2bd9293136ea7ddb0 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/284499a2edc041f78d6b94dbee1e1e07 : FAILURE in 4m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/433e8fa3af064f07815df73ee85e9c74 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ad559e1ed1f40998ac1de91365cd130 : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08897be5712443a395f49a4919b05280 : SUCCESS in 1h 35m 02s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"cd6b5a74431c9b09084d404d2a56d56994142d23","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-16 09:00:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"967f25fa914685b069a93f132e586e4bf68201d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f1e2eeebceddcfb170886a6b80e9c7af9b2e774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:33:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c498eb04bdb407fa65571f94798ef2f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25858c78228f4ba881b1252d3f05e0a4 : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"de4f9f549163479576de4378b666b6c5dc2096a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:59:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35bd818c4ed24383a75633f2b140e7da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56360c7590494e5caa83e2f4883319f5 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed62c55da3214381a2f05e15303d89e2 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46f82430b21b4fa88940c46368c55d8d : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d31f442bc36435d8dd086181b2dca0f : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c31c908c26e4548afdf4a391d16d8a8 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da3b47ad948b4bf09b94e976e8733ee1 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6dc030271684469cac4534cc25d9ea14 : SUCCESS in 7m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/edaac4e491ba4c849c1f756d4fbdefbc : SUCCESS in 11m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0d66491316b04447b0b25143f7329ddf : SUCCESS in 7m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/be3df8c76f45480885f86e2e8c2f81d8 : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0ca8075360bd49f6814d5d5f06724d8c : SUCCESS in 6m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8f4fd64a23384f84a427113c82eb34df : SUCCESS in 10m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/881015a8b16a482ab850055fe494d3b9 : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d6636be9473746cc918a3e43be470e24 : SUCCESS in 4m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ff41fad429d54897a2cd7292ceb7ef5a : SUCCESS in 8m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/34e1c802ed5c4faaa55ec84aba69204f : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80e4cb2001924a038e490c59174b8e3d : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7f5797e9ed90478e9e155ec2d5ed663e : SUCCESS in 9m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7ff7e597de444da8985905419886289 : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/09dffb4730fa4247aa1ce2f500eef2ad : FAILURE in 3m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bf2700da00a34f468d9515bb9dd46f63 : SUCCESS in 8m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9b5aa0eeba834967ac2a726bd9deb42e : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/967f2353f9e94868a8e7d592e1349f30 : SUCCESS in 10m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/01b325612eb14441acc3e43d152b97b8 : SUCCESS in 23m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc0175a8f9c541c881efbc6a56387a58 : SUCCESS in 16m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9d8a9605c16c4541861fc1fad0ae9b09 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4b4ccfbeb3e45a1a96eb5d42b46a493 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46d0b15c0f2541d7b118bb8b4811e36e : SUCCESS in 22m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0d782df92ea14fbd8b0835927cabbdd8 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/15bdc2f4fb424ae389349230bedbe6f0 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d554da6fc0c14f7f8e3b7d52be745566 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/33ba5895f90d48b09305c9b2de5c014a : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f73b3a8254e4d049c62593f682c311a : SUCCESS in 1h 18m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/062fc32c564b475f9ae2d5d0b5bbb9d7 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b7824331760a2f74bdf41a07deefe684b1ce01b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:44:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"44d367276d539e6890f84594145dd9c55dc2abc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 03:00:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767bc4b2b4c34153a54dcb476fa603c6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25ff6d04c331446b80a84e8b760af8e7 : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8a1eb2e9c5b99216e94a483f3ce8f22e520c84c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/471a53f2b9684c638f19681c7899c1b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea96e99808144eeadbc9ea392c9e1ea : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6744109937f84e529714c8f903f64582 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0203144ebb364e7788f8666f83ab38d1 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae6b7240747d4850a0479c53cdc8dfc8 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64083a38bf0340b3ae491d600264456e : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29c66461b7514dfe8c77fc5a42102ebc : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fc0538824974e709d5656a35f0fb8e8 : SUCCESS in 8m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7900a5d1136845e2952f9c0246a038f5 : SUCCESS in 13m 46s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/08e08ecc758d4c41bb125831a1b479de : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6ab0a002878d43f5bdc154158231eb61 : SUCCESS in 6m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/edf6511f746447e5b8cae3e12cf3aac9 : SUCCESS in 6m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0fd1194bfa7c4573a8d3c0385a68df3b : SUCCESS in 13m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/00b9046cb6c046abb466fc42640a2975 : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2ea56cbba6d24df7886b6e20e167cf15 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b83be4466b20433281bcb8c7424aab7b : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/629e8914d5444a1999061489b279e86a : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ea8fe8e40254a3db874849782fb7c30 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/62645d8a07c64a53b61dde37ef57800a : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ef7a5c1de8344b19ae6d6d2daa90f6a9 : SUCCESS in 10m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e88fc81704148d3855ad29a4894a130 : FAILURE in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/786ba3f9b1954a7489a5da278b2945c1 : SUCCESS in 17m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/79012e97be8344bf8d73a762067b91bf : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f521c11af48b4844ac7cbb69346cf494 : SUCCESS in 9m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/493cb3802b5a49f2b49651e74ef09f25 : SUCCESS in 22m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ecf0b36afc2a4c338e8593936cbbcda0 : SUCCESS in 19m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/48d54d89528f4500b8b76a6d85b2b916 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7400b2923ea44d7fa4418796c4f45f22 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6dd76c52d77f43569b088893bc18a5e2 : SUCCESS in 22m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/702f5f3f184a419e923c017d46704dd5 : SUCCESS in 7m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/addd22b70a43402fa3c6ef9836971f7e : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87064631e25a457daac827e0b6f5cdd6 : SUCCESS in 7m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8aa9d1d65bd644338f1205d798260c6a : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4ae18d25f2545b284c14a82028626e7 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0bbaf995c7948ecaa374cbe7efe10ec : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":2},{"id":"06239534b66a3a9741dc74afd4fc5f624a6ca29f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 09:05:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8548e66c140e1a29bf68088d13f90f6501ad1e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:11:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397285109f1247ce968066f3ab2270f6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5ec9748b021e4a929277d460bbd098ea : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ad9112507823f821ade0c411985ceb2ea49c58c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:43:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"498af74410fd24b4bebd6dddee480e755731eedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:48:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19b98bab1f634c91b16b8ad017254403\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a4f4736535640a09bf2852128040741 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb1b12b960c485c97fe91bc11fcc92d : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e02f05a05ab64e55bc8d06774d7c8c84 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d16258f66bc4bbcaa4aec9778f3168f : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a70540f963e049c49b8cf0fe24ada8a7 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e596f3185b7f406dadedd7699eece0cb : SUCCESS in 4m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8afbe9a5520430181bd5c2f70509f32 : SUCCESS in 7m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/156fea04f3c34f4782dfaabcd07417ba : SUCCESS in 10m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9736433f9d144014b5b0c6bfc1584086 : SUCCESS in 7m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/59c8015290004d64a7a5115723331027 : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e64d5a2be5754a2fad21cf8ea18bf3f9 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/43388bdca8fc47c592c3d615605f0225 : SUCCESS in 10m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3119635f9248464b84c30052da54b5cb : SUCCESS in 4m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dbfb752d85ef4b69b0d389185c301086 : SUCCESS in 7m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cbef6953f1fe41af96a86c246f64bc42 : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/548dfa4921994d648922710417f93f1e : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14d62782fbaf46cdb1170c3f9cc13bdc : SUCCESS in 8m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4a80a19a22e6477e9e88e5b713f87b69 : SUCCESS in 9m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c56d157a10d2474a8e231fcf5e6ec11e : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f6ca1e2eda0e4627826d2dde83cdf940 : FAILURE in 3m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c5acb39daf28496092f85a925904301b : SUCCESS in 9m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7fe84eeb5ebe4c51b01290b5d73b83e1 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ee43e2d7d7454c5dae37b69afbafd928 : SUCCESS in 10m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0fee914aba6848f68d4e40ce450c4d16 : SUCCESS in 23m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3631a323766a4f21aee1abd28fe80691 : SUCCESS in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8a69df596fe44651b6a98b4d14619393 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/26f1828a49564ddd82a378997e8b71fe : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1a21877aca04de59a9b196f91498571 : SUCCESS in 21m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4f5c4d77abb94530a878ab83acd3aeda : SUCCESS in 4m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ec820dd0d9304fa081acfef001b2ec9e : SUCCESS in 11m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d61af0dbe1b24d9aba713bfc0e83aa62 : SUCCESS in 8m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/17f771f5097b40dab163ab5396cbc43f : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3ee4a4014684c1d85ec47fc176b7d6d : SUCCESS in 1h 07m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/191ebf305cf54abfbb44841210cbb004 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4d02f2c0e261bf7c43b4389b5fa711b1380dadb0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 11:33:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f19b97d852111d25107cad594a1f1541b010d219","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:39:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfbdfbc932943d3bbaba9ba60c36520\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/668f2508b8654f94ae829d72887b26a5 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"47f05ce0751baa045ca2dfcafe7085f8fd24fa8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:01:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/329544c7aa8f4843961bb494faa17511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c8bea728924cd3ab55e586caa26c89 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ae3a4449ca429193e1f222ecfb53a9 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16926a3e863342459970f0312623a090 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9dbec5b6316476f819e0f6b94a1c085 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb72cb882b4b4e3b9007b6fa54a29575 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b57c80c931c343cc9df013714d2b23e5 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1023bf4ac5b3402591f582bb8f559977 : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cef3991b775d47a7a584fa5d04a116a7 : SUCCESS in 11m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/85e1a79897fd4a928f06d719b7885eba : SUCCESS in 6m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0b5d28500de8445f9efbac006b75593c : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/54b0dc9174854e3e94bb7f7e97edf1a4 : SUCCESS in 7m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5714ca94dfa0418eaecd2f7666939b9f : SUCCESS in 10m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f50eeee00caf43c687cb7a3969b2a290 : SUCCESS in 6m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/950d56ef20304ac5a447bbc6bd1723ce : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b04e5926a84e46afb4d7927b966ed235 : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7fe4e382b4ae46bfa090c85cae1805b0 : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5819db3e885b40bf88f3e36ad517aa7f : SUCCESS in 37m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/da77240726434a7da4b79a59f160371f : SUCCESS in 9m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0a528a5351d941858414ac55538fe5dc : SUCCESS in 11m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/72d4fe454d97446cb1f7d580c5e0d958 : FAILURE in 4m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/43f94e9012c7439ba35d969c0a8ad5a8 : SUCCESS in 14m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/65241db7f3db483f8177a86da177f94c : SUCCESS in 4m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1d3826c8c42944bf928d005d7ed5c52a : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/080a9713395a46729a5fcde79e29ee83 : SUCCESS in 23m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/318c07c254f841a495e1126497357b8f : SUCCESS in 12m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dd4550e195e948918ad4f4a7c900b7eb : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34a25b46b0c54c1fbd4b98b0bbd47531 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2376bf656f7343799554f017826a6841 : SUCCESS in 22m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e172bb45c9fa4d5eaad5f8bf1027a5f5 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dca75ff60c00476ab456b3f09f3b7153 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6aa4bdb9cba7461ab6df84524857af13 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d990b7df4e12457a82bcbfcb4476e7d2 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c139e7f51b84d74bb3de7086369e583 : SUCCESS in 51m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac295a5fa4df493387de97e7a1c161f2 : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ffd561aa489789433a9594111091a2a4feedf36c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 14:23:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c60eda78ceeac5a779102ba8894d5ebc41586f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 14:26:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d121a39f1fa534f6e710c5b48a6d1eca4e7cec9c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 14:31:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7ba6dc4510aae80e1b1c6b6c7394e6ff9b50cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-23 14:21:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"76f32c69ae6f04a034c0f518af4cfb79c99eddee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 14:37:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743194ad604c49f7ab48031f6a96b82a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c7e6ed960fc04d3c9210fada3e7a8274 : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ddc6c8fd86ffc1446630241b3f325748b2e46745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:40:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a70e89ee7fa48e1b13d839caf7060e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ae48de71e348029619c2081e959f59 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4208844ea2d24b73add6fd9a2da34115 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c00951e4b5a4f3ea7f2498ef7b0b6df : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/337f6bcfc6524cd5952f3d43e0122a4a : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d1d24823a884ae3aa2f08f5ce149e13 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88f01477b2874e8c8a2d00568fb589b3 : SUCCESS in 4m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/963f15abab464044bc1244ba4dfba2c0 : SUCCESS in 11m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/51a13bf758124a6da9256ec5df326cec : SUCCESS in 12m 12s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fda39684877d45e78cb6a1029afc9cea : SUCCESS in 7m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1e8dff1e98064b90bc9b14394047d1bb : SUCCESS in 5m 13s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/53bd2ca3bc1840fd9391faae4f86b411 : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/29d1cad3ee064d7db017d85751c2a857 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/511ad304342b46d6bef4d44945829aa4 : SUCCESS in 5m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/27349c74bd294e8495e559a5735ed347 : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2ded52010cc4da38c9c5d340ecf5ac0 : SUCCESS in 7m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/55a826881ab641d98f5707d85501c303 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/063ffcb5e2a7416f8ef7860b964a7a98 : SUCCESS in 6m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a4dbec92d3fe4a5d95b58fad644d8cd6 : SUCCESS in 8m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5ab23553f47d4b96bbad799b89a47da7 : SUCCESS in 10m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e458399cd26c4b6cb9a43b8e0d1d6c5c : FAILURE in 3m 02s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/66f53230a82248b9a135c98e15cd117f : SUCCESS in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e7693d9a199c4f068caf8388f82665ca : SUCCESS in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9d81ed7ed4ad4bf793f9e56e280e3cf4 : SUCCESS in 9m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7d86544aab5c44cdaaff29b7eeabaf4e : SUCCESS in 23m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a6f9552c7e3a4e33ae7ec47301708a37 : SUCCESS in 10m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c80c5f84a8a448159ad85830cf580b58 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/82851bf24fc1427197378c867097877e : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85ce62ee57b74a38a1186cfcb3ccda47 : FAILURE in 24m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7535d179b7454a7db94be2221688f0e1 : SUCCESS in 6m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/381afa2582bc4762a6fa90fc21cf8bc8 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b5803bc60e943feb12d9b9173fbc330 : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8400e14187cc4d75ba5c871fd5ced0bd : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05773a6a0eec439eb8d5669a5ba29967 : SUCCESS in 47m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18e456e33c9c4b95aafbd3857acc967a : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":3},{"id":"c33e460c7a802de72cf747e3ad261e279162e460","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-23 15:47:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8627e510d72f19b1ba9f98a5884072afa42ef029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:54:31.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ea49575bae4a8b8907336ccc4a4f00\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c2def38141a24b79b69ec17ac29aa767 : FAILURE in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ce461e911e4955f56df37da34ab6f2c5b8b319c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 17:36:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8b59916381476195ee824d813c7ac2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c307347ff71470c8ef790321cd48908 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b18bee4b71489989f5ab1e4c4aea4e : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69d465f60c484c94b2014195035a8eb2 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/597f1dd71cd8406388120f8fe761e863 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b07906539c1241d19e0013e188ea4a66 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/063d8fb9e6494a4c986e6a9003d9c0e7 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/647a8d1b6e47437cb80de0c692cdd9d1 : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60d1e028149d470884bb310ac46b9193 : SUCCESS in 15m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2a6ccffd986941cf9d879ee7f4033cd8 : SUCCESS in 8m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a80869f862fe4723906ecc02fd4143de : SUCCESS in 5m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4c7995fa4f594244b7912ce6cc25a9f1 : SUCCESS in 6m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/48f06741a9c84dcb8bfe2632e9f18e52 : SUCCESS in 9m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/120e36bf9b9c43e8bdb2270af0fd8ad9 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6dec4530a70d4328ac4afaf9d1ef1e2b : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d1eb2748e07642718e44fe0f82e7019a : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1b146fa1c91c465196a7f1618bd62abe : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4412f4971763443ebb4956da82eb5f91 : SUCCESS in 7m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e828b08279ea49f7966cdb4987b7d85a : SUCCESS in 9m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a9e745650a4f48c180f089118e7a0765 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/29f70b1a13b542d9bd4738fb09c18491 : FAILURE in 3m 16s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/341d14ac0c794a018c6e611cb72ae642 : SUCCESS in 8m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f3ac7028e61c41fbb41867b2b094dc4c : SUCCESS in 4m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/46b790c8ebe548de93ca09cd7b8d9e16 : SUCCESS in 9m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/40e00256390e448cadb1be9625d29351 : SUCCESS in 28m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/27ef6a235a844727855f6d8108fef36e : SUCCESS in 12m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/eed32fa3ef664014a6d28f54d973e655 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c87ad4ef87514916a7b178431e135b02 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0325acd4a10249e9bf9534be06e20a86 : SUCCESS in 24m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3ea37bec97fd43dbbc71f4453335c524 : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f2e034ca0c7746a0a5e6dfaf8f9ad987 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2780c39015b1498bb6cbe641e81a0b0f : SUCCESS in 5m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b1b8d4ad703e4f398992df088d83def7 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b35e28b8b744bffb107634fe951d082 : SUCCESS in 1h 10m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/644fa0c4133645f18710fcc53d48a4f6 : SUCCESS in 1h 40m 17s","accounts_in_message":[],"_revision_number":3},{"id":"831b065859eb3d976b3be8323660ee22acef5d42","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-23 18:40:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"030a3b433372e2a1f4a6bf36d68a60c0ebe0ecf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 18:41:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2aeba92ac02614b4b3bb70e0f899169ba8111dcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:37:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3bb0aea7b944eca1368427e0dd1de8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4b01d44cac49a6ab9385adb02f13ea : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0af771641fb4890b9a03aaff9c99e9a : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ceab463004a4f9b8ee50decf1e5a4d4 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7268271f069a408e868eced70ca3e0f9 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c16b46b093343d7889c8e07c390d45d : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6969347d8c64eb9b13d0801a8d8dd79 : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e29453b75c8b433295c9ca4242030d29 : SUCCESS in 8m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f2018c9f7f84c8cbc4a5d8312417127 : SUCCESS in 11m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/16272ab4d3db40f1a163af05830428e3 : SUCCESS in 11m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b77b22fc9e854cb9a008d8e69838007a : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9ced156727f34820805bee0c75396503 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/225208f1c0784742a63f6e1b9b55df60 : SUCCESS in 7m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f07c4e55638642008cfb964da95cd4c3 : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8407782fb1644a78a8ba4146b8cf22a6 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/68d36a2085a24cf48b73a54880e9b12d : SUCCESS in 7m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2e2c55e6c96b436d81a7cd93910cfc57 : SUCCESS in 10m 20s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4e9f942cc7a244759392c45ffa651bea : SUCCESS in 8m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/04913ba50e024addbbf7898ebfa66025 : SUCCESS in 4m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/53b245b7b5a04a989067a7726c9f9db4 : SUCCESS in 27m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c142035c1e9641d2af685680abb66e62 : SUCCESS in 13m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ae4e21a5ea6c411da63c95f3b46f9f74 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e08ac8da399c4db29b81141669cf53fa : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fbef0d3b2ed4d75989d5df94bbd68e9 : SUCCESS in 25m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ecc1fc79c17f44dbbc0fd981b0e12ced : SUCCESS in 4m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c41039d024e640daa1bc2a32c7b07a71 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4d332b1849f4f27b2233e7d27b813f2 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41609e370f1e46ff8522f287525f704d : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29f01c6cd0ba4a3d8903867550144485 : SUCCESS in 49m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c2e4e18d5f442d9ba4f76fab4bed848 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":3},{"id":"496131fe816f4f7e566f7fe8f8e9fe270ed07161","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6e0f6f1e554601e6ea44aa39d39d939bd89a73c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:38:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf47243fb9d4a8f823a29d4e5c20085\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e91db2f28a4417e8977a77b6b37d583 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ecc6722731d08223ec3ad20f202def8614245109","revisions":{"a2895318ad051c6e9fc9cffc373cb4c24bb81c68":{"kind":"REWORK","_number":1,"created":"2023-02-13 16:06:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/873598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/873598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/873598/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:06:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:06:17.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.0.1","message":"update constraint for openstacksdk to new release 1.0.1\n\nmeta: version: 1.0.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia1a7512067fb0cd7adcb250f6dcf477842987458\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1d25eb3fc1d26ee3dfb6336f4687a7583d0fbed6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2895318ad051c6e9fc9cffc373cb4c24bb81c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2895318ad051c6e9fc9cffc373cb4c24bb81c68"}]},"branch":"refs/heads/master"},"a3b2a63ddd423d98442fdaf5b99d04c3f9b1749e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:44:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/98/873598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/873598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/873598/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:06:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:44:46.000000000","tz":-360},"subject":"update constraint for openstacksdk to new release 1.0.1","message":"update constraint for openstacksdk to new release 1.0.1\n\nmeta: version: 1.0.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia1a7512067fb0cd7adcb250f6dcf477842987458\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1d25eb3fc1d26ee3dfb6336f4687a7583d0fbed6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b2a63ddd423d98442fdaf5b99d04c3f9b1749e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b2a63ddd423d98442fdaf5b99d04c3f9b1749e"}]},"branch":"refs/heads/master"},"ecc6722731d08223ec3ad20f202def8614245109":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-23 14:21:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/98/873598/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/873598/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/873598/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/873598/3"}}},"commit":{"parents":[{"commit":"d25bfc90763a1790ff8d27749c0c6650bae86c63","subject":"Make cross-kuryr job temporarily non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25bfc90763a1790ff8d27749c0c6650bae86c63"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:06:17.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-02-23 14:21:44.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.0.1","message":"update constraint for openstacksdk to new release 1.0.1\n\nmeta: version: 1.0.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia1a7512067fb0cd7adcb250f6dcf477842987458\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1d25eb3fc1d26ee3dfb6336f4687a7583d0fbed6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecc6722731d08223ec3ad20f202def8614245109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecc6722731d08223ec3ad20f202def8614245109"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
