)]}'
{"id":"openstack%2Frequirements~873600","triplet_id":"openstack%2Frequirements~master~I1868143456a9468f216bddb84c040855acd2a5d8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1868143456a9468f216bddb84c040855acd2a5d8","subject":"update constraint for keystoneauth1 to new release 5.1.2","status":"MERGED","created":"2023-02-13 16:10:29.000000000","updated":"2023-02-16 20:09:41.000000000","submitted":"2023-02-16 20:08:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873600-new-release","meta_rev_id":"16de34d723663f521c0e741bddc4770900eb0f49","_number":873600,"virtual_id_number":873600,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 20:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 16:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 18:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0c217d83768af2821605074fad6b19bb6f2a9e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-13 16:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb9c9ee57eef52a553c8008d7c68771c8019f41f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 16:15:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d279656b235e47ef892887a35b513869\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/53ddaf6607354493bbf0b1e7552d4a84 : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8dcbad9525c874e964dfc1639f995a3baecab13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:17:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/713bb9b0e855435dbf0d276d0a48e425\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd7456a430b418aa24eb2fff6b5de8b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f163d22d9249f7a8ca9eb4ded47d32 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7eb0038d2974dbe811dc19c8f06f70f : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd840041b18a428f8fd48c2bca1fed31 : FAILURE in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0892acf63a8406a93813002c9127ae5 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/931831deaee143458b30c31b7602d09c : SUCCESS in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56b0853e66ae48b39ae907952021d2cf : FAILURE in 8m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f565dc7125e944a6a83e822c316e250b : FAILURE in 13m 47s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0d89e8b67a8d4d71bfa46489a5c3e116 : SUCCESS in 7m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/426645868a2f441db9fc89f2ef13a219 : SUCCESS in 5m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/425c8aca0f724bc79b5344a630da8c15 : SUCCESS in 6m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ab2a5e24c18445d88da43b0d82a3ca72 : SUCCESS in 12m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/129a3dcf17c74a2e9b4e1a83fa1865ee : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c16581e5797444b58d4144e4c43536bf : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/211d0b5128054f9387cbbec2bfe88dd2 : SUCCESS in 10m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fd0ecfeb37c44a64a47c22e7bea63475 : SUCCESS in 8m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d9848fd4fae4bae96faff4714772b58 : SUCCESS in 7m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/79b1ae0ef2014ecfa93dff6541cea02e : SUCCESS in 7m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4e3ce727de5a4d89a97530674bef0c4a : SUCCESS in 12m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/29203cbcd6144b6396c2644905f3896d : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/775d8c3eb355422b847b52e0cfea6b54 : SUCCESS in 9m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/20c20df7445f4263bf1b003153012995 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8f5c5349368e445480e789a3ca5b9700 : SUCCESS in 9m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2f0dfb85f2f940a29b98e8f408c136db : SUCCESS in 27m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/77cf689244ba4113bec6d8c2f448796e : SUCCESS in 12m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/164666e8fb754b6abb5bb1e7f65bae71 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cad1618f406f4e94bc3a22941b2de165 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21bc28445f9b4fd5b5a93dd41fff8ef9 : SUCCESS in 22m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2eacc3d6e74245b9a17e87f2fd5ef040 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/acaa97637f2b41bbbe608c96efb3ab35 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2e7078abbdf438ab71ca667196c11b5 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90f7794c89544b71889835ad1ecad758 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff8cb46e065d45a2883611d4d9a179ab : SUCCESS in 55m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8792cd12f63a49d7a9b58d2fd55871a5 : SUCCESS in 2h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3fda1a82cadd9d79d40e7497472e6100d5aa0c64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aca495f06ef3dc951f40a82cf23ec5b01e11074","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:30:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac93b27bc2c41e9909fd14981f52ad1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f4bfa9ff856f497bb8b230e7ba127cd5 : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d0e0a51a7b30dfccd2ed7599ba584f0a81d0ef75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:39:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd38ae7d29424b65beb8265ae27c2823\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff22e0fbc3fd46538a1777f25731ba75 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15aa8a76ac384e7ea639ed88c75e1d21 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b933c05b0e914093b7088047e398ea5a : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27ec577c47bf468b883eb87a932b021a : FAILURE in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e008921f9a28499c9afd36f0a9dcebee : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77ecddbc31814f0786bace11aa5448c5 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/549303074f584fdb89a8f98ef99133c1 : FAILURE in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3568a44ffbda465583f34f42d4717eaa : FAILURE in 11m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7e718ea3cbcb41abb134f5cd7afdb3e8 : SUCCESS in 7m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3ae5961df3df4256b10ecd46e0d1d46c : SUCCESS in 5m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/420f702a533447e58e667ad6aac6ad8e : SUCCESS in 5m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a0f713c16b2f46fc92e4c3eda46d2398 : SUCCESS in 16m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8dd1cc43f1cc4d44b02fadaff15fcc31 : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1adbbb93bdb0467089a7ee8738e5e714 : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2d526356b660415993287bef4a236712 : SUCCESS in 8m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4ee87fdef0f4452c8d28419138c97dcb : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2008f8ce298e4223a8f1f05f41b4ef75 : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0c03879480c04e768b7d95710f4ed769 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/618084ef8e7542889b3c3dd14c9c0e8a : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3232f8cc176e4455a65a7d06971ae5dc : SUCCESS in 4m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/92ef3080d87144389a6d5f1b2b8db018 : SUCCESS in 9m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aa038ce944724f2ca8fa9ef9e149ebbc : SUCCESS in 3m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e2ac57469ceb4394b6387026517379a8 : SUCCESS in 10m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b5e768ded5c2432290cf187e5bc6f345 : SUCCESS in 28m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7ceead6943f24210b5586b7b9badea53 : SUCCESS in 11m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a2be088bd84b4b9f9bd00e35ae9e0d86 : SUCCESS in 7m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c5b7ffa804447f988725d7cd1d49b64 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26f7f04d2401416cae6c3e3f3e468c90 : SUCCESS in 23m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ca242d5b04ba4547a1b3a8076ba77c9a : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e7541c0dcb184a17b19407c93f5dce2f : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46657391c5ad451caa2cd04a104f3be8 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39f27053e84545e3ac9e6df843bc38c6 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c169327a0a1d420caf8b95436aedb506 : SUCCESS in 50m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4ad866e6c8743028ef8c295e0f871dc : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dff450b9434acc0a15989e4606d8567e36384db8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1b039d6fa1a4327dc53b66e86d33ebf541e24b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 03:23:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/401181fbbbd844ad8ee921eb526d6562\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/43872b89bc844547a85d0085ad5d1fc9 : FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"341847ca43315afa057fb290f803dfa27ae73d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:53:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ee1fef2bb44f93a89de7038157468f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cadeefe654f94a5ebe0540e6f64e405b : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51a3290a7ad4497b5d8d32135d572dc : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58ee1deaf1944d6fb7ab7c609a2f1a94 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e2e87b6f0d6436c9e9784d8d88bc2d8 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04c66bb92fdd4b4cb4c3e0c8848af17c : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b2bfe0fdf0c46c697369d1aff703198 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb8aef4d1e2b4a6085108f1385f7e17e : SUCCESS in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/426c07fe98784716aafd822b59115c11 : SUCCESS in 14m 36s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/52a358d6cd6e4a579ca4a43c7dfe9363 : SUCCESS in 7m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c45a7df9b7cc48a5882232c111619a8e : SUCCESS in 7m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/505220925aac498186610ef82e1ef07c : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/233fd2bfd55d41e6a38666bfe76cad19 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f65081eb18b148f880914e8b7dcced9a : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/408f728e85d9445e9ce1150b9719d360 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5280affb50ff49929c0da039897d9355 : SUCCESS in 9m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/84d91e8c1e754b129b29255db8fd7e31 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d49c02edeeb47b69533291f7f21e622 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bc780b24db4a4dca879240c6518b4a3e : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5a2523cc178649599a53925f72ad695b : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bca4a89e87ba478b8f8f3fb5bb0459dd : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/97151f8d9b2f4c12b8ca08d64d9d9dfa : SUCCESS in 9m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4052ca721d74423ba18bbaf1de4e7f6e : SUCCESS in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3a0908b1f5324aa68554c758f7856523 : SUCCESS in 9m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f03075f75853409487215f4156e9aa77 : SUCCESS in 32m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/10285aaf16eb4b10a3324681ebc43337 : SUCCESS in 12m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/155bfbd774604187ac8238f081735f12 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a413b8edd884c3b907df482166da1de : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2def8422ff9b4bda96b41654d8a058ef : SUCCESS in 24m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/facb6a2e26d74ce9b46ecafb0e1944a2 : SUCCESS in 5m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/46b383e4ccac4fd995f3563511818e14 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33ed45c469a741f2a818d26429fcc87c : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2d054405bae44d728f1ee36802a1385f : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2496525b00a464d99640a32875d3b6c : SUCCESS in 1h 05m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6b0ccdf84d14de6a090580464d45e63 : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"96cae74e531ae034cb49ecd15bcb6ab857d2307a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"136d49cd95822f6b19c4bb08ab119080d8d30028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:35:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6adaf0ebea5ded325ebc3226d51ac620890086e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:08:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f18ab81b6e4d0a80b35ce6156bcf88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d645f43f6c421fb0928a19db9cd370 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0156050782994e40b117fa3c4719bd8a : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c91e7b32115427f9529c0bff4d80a97 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ba28b42b7104ae8977801b79408f720 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d97dd26284545b8ae66c99833e9e13d : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39111ddd0e86442ab69d5f45bf9a503b : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eaeb2b11104246ca8f8d0eaf049b5ccc : SUCCESS in 8m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e09943f3feee4e7b9a1b858ffc37a60a : SUCCESS in 12m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/efb50367b01a46819143bf49c5186c6e : SUCCESS in 11m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f505d37b020e4698bcbba400c76d7db4 : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/253c438b70fe44229abcbcb8ae258fa1 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ed7f22387aa745df8ea370dcb3a2317c : SUCCESS in 7m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0898f925e46949b6901f5971aba214db : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f59f51e7b606453d96fd92b69cb5134e : SUCCESS in 9m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fb2fc99a7d7a42ee856b605b605c9256 : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8794f73d66de478b9449b10822d5a5ec : SUCCESS in 12m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/aa352aa9bd9d420488e936f3aa67722d : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/818fe00c40064c5dac53b64291b8f174 : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0e5ead07c8c84781a3575f7a06d9a324 : SUCCESS in 4m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9cacd11496d24bef991f3e345b307dae : SUCCESS in 23m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3ad97f021bd04a858fa4fd6c8db6055a : SUCCESS in 12m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a372d877f71b4a409aba71cc511b464f : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d428a41fcde4d4386aeceb09718423a : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c96f32a2ce141c0b90b18a18ca14505 : SUCCESS in 26m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5a21e876c78f428f93cba3388a234e7c : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/91ef15431d80471ea7ac8ea5212772d8 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6a84b54a9164cb1a3618fa1d5538471 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/31218844a2f04b44a2e4f16696ae326d : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86cef510ceb44996b058a5516e07c139 : SUCCESS in 58m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1ae1dd34ec74e9080157a6a6b9590a2 : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b054759e637cf4723b2c448a7efd26c5bbc5eafa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:08:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16de34d723663f521c0e741bddc4770900eb0f49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:09:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5988fae6fb4e249eee2d5dadf39344\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59047443fba04c13bd1616043be6f795 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94d634dfe2eb1f8e659d4b7c5677df47e01f5be0","revisions":{"94d634dfe2eb1f8e659d4b7c5677df47e01f5be0":{"kind":"REWORK","_number":1,"created":"2023-02-13 16:10:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/873600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/873600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/873600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/873600/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:10:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:10:28.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.1.2","message":"update constraint for keystoneauth1 to new release 5.1.2\n\nmeta: version: 5.1.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2ba2652c38dc45f2ac63f93a61a50faaf0f9788c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1868143456a9468f216bddb84c040855acd2a5d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94d634dfe2eb1f8e659d4b7c5677df47e01f5be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94d634dfe2eb1f8e659d4b7c5677df47e01f5be0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
