)]}'
{"id":"openstack%2Frequirements~873601","triplet_id":"openstack%2Frequirements~master~Idff4e511774ac4c628110fbc6ca7b6c61f997446","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idff4e511774ac4c628110fbc6ca7b6c61f997446","subject":"update constraint for tosca-parser to new release 2.8.0","status":"MERGED","created":"2023-02-13 16:10:53.000000000","updated":"2023-02-16 20:10:49.000000000","submitted":"2023-02-16 20:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873601-new-release","meta_rev_id":"ff2ed2540abc4d0f088609bc7ce3b6b48db4522c","_number":873601,"virtual_id_number":873601,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 20:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 16:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 17:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"46f191e8dfb5510a04fe6facb688bb7933d030d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-13 16:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71b1cd058ffeb22ce6225e6707a0b97337cb45c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 16:24:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629c52f958a44f188f62479b7a4461f5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/38f0618cc72d466e8c9aad856a567b8e : FAILURE in 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9ab1c2f1f91251d51f2c2a558a80e288e15bcd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:57:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d83a8582ab14355bd964c12f69372e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d78f9b4d2f4157993d1e0e92f0dcdd : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16440ce6c394633a6026541c6f7c4db : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cbccb5790a44635a51f6aa568047e5c : FAILURE in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6929e418dc974c6e90a02608b3a1dce3 : FAILURE in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8585d83084674762a3652bd261e41cc2 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df941fd7c35a45ffba5e2ee1bc592b55 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/478e8185a01d496f936d591854430f83 : FAILURE in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b1c8994b0d4e4868a6563f054994df1c : FAILURE in 12m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/21fc5d78cee549dd93ebc1cdc6297e93 : SUCCESS in 7m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a7fca49ac5234cf89fcb73515ea5386f : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ec685a3ade6844f0acf0d81320e57648 : SUCCESS in 5m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/47ac9f67ac374e569aa502e46ad28e1b : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d09eb15de305408cabbaceec81e34961 : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4d5c59f7bc694631b244f3d0073bd33f : SUCCESS in 4m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f700dc9ae5f44855aa7888798c238d8f : SUCCESS in 10m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/280d479b9a144376af53785398284074 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de48909f93b04e4b94cbee912125c552 : SUCCESS in 8m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8c85bc57efc847ad81cdfff06ea761dd : SUCCESS in 8m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e9b0ed79cfdd4f538139ce2acdd14a5c : SUCCESS in 18m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ab77213aacc44c5c8e871d2f426822d1 : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/668baad17de641e29b3fdf2b5b38a95b : SUCCESS in 9m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6fe537b4d0d741fabee4d34f4bcd4474 : SUCCESS in 6m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d0c4008d3e7b40a99b252bb6ec68f091 : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9cafd30b0f244dc9a0392cbf1c135a81 : SUCCESS in 26m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8571e000a244e9a8ccf0dd606eec7e2 : SUCCESS in 11m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9b98f10a14ed48ce86ecc79765a4ad42 : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e01230393b44144b375ffe5775af78a : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ce4ed08b1674996b77ca9156243aa2c : SUCCESS in 23m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e79ce994bc9e4813819d118029fbca1e : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/11ff34977b4e40e58cfe8416282ca01a : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96113ab5cfc743fbad37db22852ccca5 : SUCCESS in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8b24deab7aca49b6b4e5418f1fababb1 : SUCCESS in 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d48fdb33321444eb5c7338c53deb6dc : SUCCESS in 51m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb2f6d5b3aed459499d94f7548dc84d2 : SUCCESS in 1h 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"69de273561b67f243832197a427a5ed2035829f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"910a970088fff6beb219fa6a1ea28637f4ed125c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:30:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1388811922d848d0b81eb3130c56e8e3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7df7da1ed9f24f0a837bc65ce030c910 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"68fe262c16fcce0a1ded02f4b070f42ce1cffd1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:46:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4366164eba99413e88187ff0e4478b5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a99d253ea14dd1aeeb5c293886643b : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57233c6c85e84ce6b0cfe08e8ba732e6 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41ab93bf8e9486890bfcc2a869c7cd1 : FAILURE in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd642f6ee74440908243c33b6399ac33 : FAILURE in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22d74352a2844e9384dcb98367378a6b : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c4e383f285c4e55949c867930d2813b : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06e45c060f954114bce28aa640d1ae9e : FAILURE in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a03d461174524948a7dcf605f5cb4954 : FAILURE in 12m 23s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/aef9b584d69a40a79467644bd76cab2b : FAILURE in 6m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7b0f6a3af8d64417904f110198f90c43 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0992d79ed65e4fcf89838aad376c71ac : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8c05cd5b13b0400da1a3c455bcf6039a : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/330f3efcdf1a404fac7b57c850f6d8e9 : SUCCESS in 5m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ee0b154c8eb74c90bce874ae5ca445ea : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/58967c7747ed4028a64309663c8d384a : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5b124252046d44d599245156d539ffea : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70cb1cfc74f14b69976d59b2b26a633c : SUCCESS in 8m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d1599363ceef4e399801220c24c176bd : SUCCESS in 8m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d1a917f2df4d4bb1a740f189769d797e : SUCCESS in 10m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/08359fa58eff4907a31e64e8d496c8c1 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9864df993d59498aa827cb74e7294a0b : SUCCESS in 8m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c25bdf79ed924090af678c23c5268924 : SUCCESS in 6m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5a49aa3ca58144db8d8a1807cdcb4428 : SUCCESS in 10m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/19f16c25b9f64042a77cffccb7056d68 : SUCCESS in 28m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/edd7dfbfacc048cdb1f86060fc582032 : SUCCESS in 14m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/62a71d7a52dd49ce9269c3344eaf0b79 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b8e97aac9a4427d8185dde1b87edb0f : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa46e288bc534ed4a9f34842f754a1de : SUCCESS in 21m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5d83c3d5b8254388892ed39c8babd9a6 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b4298e9d7b2e4505b6198f8b89089507 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5d38f22e45d4b6190e0c3043de55ef8 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5882083d568d4cb2a6c431604267dc42 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6eeea4f09b864525b3e1cacde425c2dd : SUCCESS in 1h 05m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ebde09fc74c4f30a5977cd5c8370164 : SUCCESS in 1h 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d97c2308c63e10604164e99f1d16119749114eb7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06d95f85cf870b48c99d9b1a7c416faf3915b636","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:35:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66de28e8ab6742ef84fd185801db6347\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/821d4ccd8e1f4c81bb4fff020f5633d8 : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4164f952641a11b5f836409bc1912569a39be080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:12:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625ecb659c1247e6970f4eb71cebdcd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15c903a03ae4d9aaea0a9a555c17361 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85670ad4319149458e0d9042ce12bf9a : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1ded84dec04ae9ad20a0c7b6f9dba5 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24f9b1eee69c46e596c725cdf1a94b63 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b16bdd863a447f29aeafef25acbc79e : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34424170435344e49d93b8e5be7e2e01 : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f8128a9ade44041954136a47b316e5e : SUCCESS in 12m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/22058c5fd19d48f7bba04b56c5b14b8d : SUCCESS in 12m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e16a273bd9464e53a5c22daaf04fbf0b : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2afe3b0e84534dd7a466e8f7625d7aaf : SUCCESS in 5m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ed47b85b7e1a4da4b21e651148e9727d : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6508a64f545f447bbd113384ec11d24f : SUCCESS in 10m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/879143c5172f4f52aa74189fe6fb172b : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e777f64ef2764a93aad77cfdcc9f5f4a : SUCCESS in 6m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/27630d6f84fc43889984b9513ef1fedc : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/eea03118c2654e50b27d177209eaeff4 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a07cb5c12e774fbda3f343de19c511cf : SUCCESS in 8m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/78aa84ef4d594efba4b809bfd540ca1e : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7a7e853998e46299adb89c4a1c19ca5 : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4c054c165dfa40d5ab35e3ebed84170b : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/308423ac92e54ae5948dc58aea984a6f : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/af3e8da16c8342a98b684e5b96bf4d4d : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5136d4594cb7411b8f5f084f80d566e5 : SUCCESS in 11m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6a9d84191b4b4e64ab7064b55a9964ef : SUCCESS in 38m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6c0985ca51ca4f0bbaabf66e92f3b97e : SUCCESS in 11m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6721524b785e49cbb645ed2692ada959 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84ff693d7ddd43b0b88d72341837688d : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/989ab29edc834c5fae1322551526d4fa : SUCCESS in 22m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e0f4fd06e4404b6a9fed46b54b658bbd : SUCCESS in 6m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2800119eab1345f9bc11e2bc43e9a2b6 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/728eafc20ba543ff9c421cea2b9cb1db : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41f1e7501da245e3861cc94f3f08c398 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd94429716974eadbdaa00c812e6c243 : SUCCESS in 45m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/002ce47f16e84173ac528c3478dc810f : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1c3dec2971b3e57ae62d1222a2f7f7991af9ec09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f82a9a291c0bfe12a27669b9b2a57d6b2189ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:35:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee95755485b5dfff02d0bbc2c2a3ab3517c3f986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:08:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f7c37ad53794802a45bad4548bbbdc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51e2fafcbe94bf19b272c6e19c56fa1 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a823edf98e48f6b436fb4eaca5a17f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdb64f6e29e4c7abf4c1cd6f27d5905 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d821a33eb11743dd9cd18bf6bedd85f4 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/005fcc52ada9407ba588a27aea7b54c8 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f764fc7372e248aaa610c09a76bab084 : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39e468d9f1a44ffca6822ae7f9fc10d6 : SUCCESS in 8m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3db53411435149d0883748fae4cec11b : SUCCESS in 12m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e0feadb81cf84495897985f55c607672 : SUCCESS in 16m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cb8196c7cb0c4f79953e194a2dc9b5e2 : SUCCESS in 6m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d0891c36c9c34c419d025663b02475bc : SUCCESS in 5m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e4a66da7f3a5443693d5c0fb2cd630fb : SUCCESS in 7m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/56d173dd5c3d402b8b6be02d1dbcb3ea : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7dec487c9dda45a1982b7fc4a9f2f2b6 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cee9efc5070145258b9276843f7d404d : SUCCESS in 9m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dbfccd054d8042ab9787acdf36759a3a : SUCCESS in 12m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/32c3b0deb28e43748957d2c00550a172 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6693439d20f548c389c3aca1e254cf27 : SUCCESS in 9m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/43c8ff85a4e1464cb764397e29ddc4b2 : SUCCESS in 4m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4da9becf7590471593b09c851d66bc01 : SUCCESS in 23m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0973f795b58e491f81f0a6e8437bd478 : SUCCESS in 11m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ff31d00d66d042bab9b4939b3393bbe1 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ede2e005b5504d35b116ad53989f9555 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6d22b859f9541d6b0dc31dd71f31594 : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/90e2964859c144d98cfb15b50a43bd84 : SUCCESS in 5m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/00dfa45649ef494ca1e9de8af479584d : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6aa13c506cd54664944110a2991c7622 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8665e634e33c49379f9de3c4428d59a0 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df3af04373234dd9bed3964c30a3bfb7 : SUCCESS in 57m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38bf49806e7d485089f34a0bc1920bb5 : SUCCESS in 1h 44m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cb68c26f6c2faa17a979287f2efb6f1c719f9274","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff2ed2540abc4d0f088609bc7ce3b6b48db4522c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35fed325e164d69b7b408d49e50f3e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0a6f6452f640578d6b43d81413fd68 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3781f28a9ae7c445356e7e5b475fef82d3194dd4","revisions":{"3781f28a9ae7c445356e7e5b475fef82d3194dd4":{"kind":"REWORK","_number":1,"created":"2023-02-13 16:10:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/873601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/873601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/873601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/873601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/873601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/873601/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:10:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-13 16:10:52.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.8.0","message":"update constraint for tosca-parser to new release 2.8.0\n\nmeta: version: 2.8.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I84be1601f28ef146b0d8f3210a0e2652ce1d307f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Idff4e511774ac4c628110fbc6ca7b6c61f997446\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3781f28a9ae7c445356e7e5b475fef82d3194dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3781f28a9ae7c445356e7e5b475fef82d3194dd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
