)]}'
{"id":"openstack%2Frequirements~873620","triplet_id":"openstack%2Frequirements~master~Ia2ba3f38a83216abc430ed754bb7d8cbe8c564d3","project":"openstack/requirements","branch":"master","topic":"fix-requirements-gate","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-14 18:42:59.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-02-15 05:43:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2ba3f38a83216abc430ed754bb7d8cbe8c564d3","subject":"Workaround for unittests","status":"MERGED","created":"2023-02-13 16:53:37.000000000","updated":"2023-02-15 05:46:02.000000000","submitted":"2023-02-15 05:43:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873620-fix-requirements-gate","meta_rev_id":"6464902639fe1cff3b9e25e3a253523df3be40ae","_number":873620,"virtual_id_number":873620,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 05:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-13 19:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-02-13 17:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-14 15:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-13 19:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 17:12:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-13 17:46:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-13 18:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 19:15:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-02-14 15:07:44.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"b165ee0bcccda047fa00d08314a81e15f255f8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-13 16:53:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aba7c15e7e8e8f63a469b4f915d11447084415de","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-13 17:46:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43db16a57604611ecc433a0abf3ba8ec642e99d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:26:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dc24de543d8460398c83f8f63b3f056\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318edb18a6cd4133ac95858fa6d78cfe : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbd353f4a594fb9be52037a98397caf : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71261072b32416d82b6c8b8e91cd097 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd80400f12ca4fabb2d4f9a300e8c1da : SUCCESS in 2m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c367976b7c9f41e0b68f903c43fe2e8e : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c2b3ad38a8d4f35a5c986462637ff53 : SUCCESS in 55m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78e78a6ce41141b19dbfa76a83291826 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5f6bb82db2c895d44efc06df0560ee6a243b8931","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-13 19:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfd108ba7ac2ebb1360845d48b4f352eacdc26e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:16:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f31309ab06026a938089d9bd7526296837332ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:43:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55891bb3affe450fbdea5a9d1ecaacba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29887b5dc0ff4a7fbd013aa6127ef908 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed3d053002c4042b46477ecde07c8b4 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27351851c637451aa9896a6c85084a37 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c185c2d17aac4e60996de0e30d67b245 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0574978028a492e8a0b8ba706bc6bb5 : SUCCESS in 2m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf63b6bf442e4a419efbc1fc8633e036 : SUCCESS in 52m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7838e23a655340fca882857f0783e711 : FAILURE in 1h 52m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3029700def652df60a6d9576b4c72cb88a66c938","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 10:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94c5ab37272614bd50a0e3490a93e550f44c1c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:51:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8463963c4a9a4052b0a0542347e6feb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe217dd5717a4f509a3065315693e74e : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7152c86cd4475f9c408693bf38d9e9 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/824d49bfb412429a8d0594dbf67d9000 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2657516ca86c4118861c5143adba5c62 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffd46d5bb52744e3adf0b9eccb735258 : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/534f9cb850274f5e922b7dccf72e48cc : SUCCESS in 1h 00m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df11a55c2c3840c9a6d765676cfff4b7 : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"febaacb00f0855abe9f481b434daba8fcd2c0dfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:52:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29af85554ac3b6cda2f241a57da1d655bd2c3f37","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-14 15:07:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"11d21106a279a77fd4902535b5515386489a9f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:03:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9550fe9ed83e47db8bc27774ae6449d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99745edf5c543d9b34bac05a3342d80 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88233a138ecd4619b8d0857626b69291 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9637dbcccb8f4c849142328f4a741969 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdafa134e8a04a5faf300c2e1689915b : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/876d98e05b6744bbb2d1a5269521a5ce : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88200c98562d466f8ee7cd15ac017ae7 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbdf547854014603921b20ba736672fa : FAILURE in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"62788eb227a6004c09a0b3b7dc3aa079c72ba44f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 18:42:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbf6b0b623278a8a963d6550baa72711d3e9a086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:38:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16681cb0debf402891a61d5bba930ec6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0afb80551e7437d808cb7b12497a283 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db67d7a07d664554ac548980d5497880 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b54024973b341fe85826d12e141cdd5 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23ed01e0dcdd4cd68507ff68318220ec : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa2ec2ff8bfe43c79631e282916be95f : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f78dd4cfd844529aac384c6801244e0 : SUCCESS in 1h 08m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e2fe89885404aa5911002b17ac87798 : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3b65c4b98266c794273949e80c41774a668c5629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:39:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f58841579b95f757f34b433c41f1b54c55920da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:43:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e75ee44d5c4de88aab95c568577ced\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510f9b1d157a41c59450cef858dcb50b : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974d935eb1ca42bdb0b7b3ebdcb54c08 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ff320e0f634782b0f94c1d22c45cb0 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22dc5018f1614d73b22292248744f202 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4823d681a5ab421d8e2672bda57685be : SUCCESS in 3m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52281856db8544beb183dc48ce842cfb : SUCCESS in 53m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/815a263be8884bfe8a5bb532a77b631c : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6257dd28898cadf7c20f745c14574748a565d25c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:43:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6464902639fe1cff3b9e25e3a253523df3be40ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:46:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3535bea22e1c4832a03566270e7f7368\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4089854757104f1db53addb58e94646c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28248c2d01151b10c7dca8e373ca621228c57a86","revisions":{"28248c2d01151b10c7dca8e373ca621228c57a86":{"kind":"REWORK","_number":1,"created":"2023-02-13 16:53:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/20/873620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/873620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/873620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/873620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/873620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/873620/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-13 16:45:13.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-13 16:45:13.000000000","tz":60},"subject":"Workaround for unittests","message":"Workaround for unittests\n\nProbably with the latest virtualenv [1] (which bundles setuptools), the\nunit tests started to fail with this error:\n\n  File \"/home/zuul/src/opendev.org/openstack/requirements/.tox/py310/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/requirements.py\", line 37, in __init__\n    raise InvalidRequirement(str(e)) from e\npkg_resources.extern.packaging.requirements.InvalidRequirement: Expected end or semicolon (after version specifier)\n    lesscpy\u003e\u003d0.9h\n\nIt seems that the parser have become more strict and now it does not\naccept other than 0.9a, 0.9b (alpha, beta) releases when parsing\nconstraints. This patch updates the global requirements test file to\nnot break the parser.\n\nAlso the test_match_without_python3_markers test caused the same error\n(with: withmarker\u003e\u003d1.5\u0027) and needed an update.\n\n[1] virtualenv (20.19.0) released at Tue, 07 Feb 2023 20:16:53 GMT\n\nChange-Id: Ia2ba3f38a83216abc430ed754bb7d8cbe8c564d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
