)]}'
{"id":"openstack%2Frequirements~873906","triplet_id":"openstack%2Frequirements~master~I7a83d4647dab841abc4c1d5452f677170b5763d5","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-15 15:38:20.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2023-02-15 13:26:58.000000000","reason":"\u003cGERRIT_ACCOUNT_35125\u003e replied on the change","reason_account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-02-16 22:14:03.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-15 14:43:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I7a83d4647dab841abc4c1d5452f677170b5763d5","subject":"update constraint for python-barbicanclient to new release 5.5.0","status":"MERGED","created":"2023-02-15 13:18:03.000000000","updated":"2023-02-16 22:15:09.000000000","submitted":"2023-02-16 22:14:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873906-new-release","meta_rev_id":"3e89cfe14319b2b7cf9ed4afc919eae6fc653eee","_number":873906,"virtual_id_number":873906,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 22:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-02-16 12:07:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":1,"date":"2023-02-16 12:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2023-02-16 09:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":2,"date":"2023-02-16 09:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-02-16 12:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2023-02-16 09:49:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 13:23:04.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-02-15 13:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 13:26:48.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2023-02-15 13:49:20.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2023-02-15 13:54:04.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"REVIEWER"},{"updated":"2023-02-15 14:27:18.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-15 14:27:18.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-02-15 14:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 18:12:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-02-15 20:46:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-02-16 00:43:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-02-16 02:20:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 09:26:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-02-16 09:49:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-02-16 12:08:13.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-02-16 14:29:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0e8c02bba5eed12b12f2a250b5a67b4699e1d86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-15 13:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2b1ede0f211dd8fcfaebd4f7b5c21988fd6d03f","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 13:23:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"25cf439805aa58ba37d7226cbe13df1147209dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:23:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/690124a1fe104468b8c09b6634301970\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ba6e578b0804cca86473e8f9f1cba71 : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7dd820c891535a5a17118df8eef19ea66e7fd82c","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2023-02-15 13:26:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9cec0e63ea8279990adb05cc1d99c0b0f3348db0","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-15 13:49:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8557842310dca52c0d5bb594ddabd1d9f2942d87","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2023-02-15 13:54:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a35623488c26ba40ed8494713f000088ada18288","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-15 14:43:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07eeb6ac281c9685a5942c95b85ed48c9f953b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:53:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecd4e51cb62240398c5c0923d3caf49b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3e8b50fdc94c4389e0040dc6debca8 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f4366d08164ebe95e577bb5b02c910 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59b2431293e9493e9233b33babbc88c8 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbc37eab18204dd2a41a8f02ebbbf046 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ff198a9efca4719b265ae4ba7bc21a0 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6862667917b4177b1fb9ee8f657ebea : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e7a8a8f1cc6491a8318f5d16d080407 : FAILURE in 6m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eacb7f1371c44757803d9cec61469b51 : FAILURE in 11m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/edafa44cacde4117a9eeeac7a6b37bc5 : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a3eb6278f05347ff92c4dae819ab7b45 : SUCCESS in 5m 46s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7a3db1d044b141ea83c3a242d3c59e06 : SUCCESS in 4m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d1c6143bcbc04bb884fe5d5d45b747a6 : SUCCESS in 9m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eb061dffc2884854bb963970ddd0ce0e : SUCCESS in 4m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0736ac2ec4204ec590d879cb7f0e7c3b : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/572097dd9b2f4325a8d86974b60600c9 : FAILURE in 7m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d745f337bc6e4be2a1af1184fd13c7fb : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3148e4997a3541e1828e4911f916865b : SUCCESS in 6m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9e4fa6ad202d4b9f9bb0954a50ee54a2 : SUCCESS in 9m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8bfeccf1e38d4727861a01a6936af650 : SUCCESS in 10m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/434f88edb7254ac18b7cc7b66d8e324e : SUCCESS in 3m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e4a528c76698403b8895a608151fff19 : SUCCESS in 8m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f1c26e3b2ba6457ea9c0ec718fc0976c : SUCCESS in 4m 35s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a331326297674aeabd137278f6c9b9ac : SUCCESS in 8m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/38fb1963c706429fa56e2775ae1a5517 : SUCCESS in 24m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/28fe8259bb464e94ae5580cebc65d63c : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/811e6cc232364a9d94aec536b131d6c3 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e06cbe8b7c1b4309b154b52a8529813f : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98c51029e75a41ad8d551f50418250cc : SUCCESS in 22m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/569ae7d3b59a41b4b67232eddf874720 : SUCCESS in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ad6e1b1a7f154fc6bfb4c8c7fb77f052 : FAILURE in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07caf9146b0d4061a4c96b4282d6ffd4 : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6720a837cbd84f6da2731273a8e77355 : SUCCESS in 14m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4233ad049ee04e648fdef67f1ea10515 : POST_FAILURE in 8m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b406e8d813a4584a8253b567ff3ac7f : SUCCESS in 1h 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff49156a4c929a5639fd0b102babc52415f98125","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 15:36:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9760fe07157200f15fe323cc14a87af88eaeb2a0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-15 15:38:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"726f72acf06d00bd21764e8cd0a9e9ceb84b4b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 16:49:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0d2789d450ded27bf809bd2f43664cfbdc868fae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:10:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b9d5d746ca4fedbc23eb76237e9635\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/52fa201c8bf0484e8e69d3db267fe3a6 : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fab339185bae17463eb030464583e29f05692cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:48:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58b509bcabc44942aff1a3f3d58da216\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57837f986a09482fb782b3dd18390f48 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94674348d0b74401b94713871ff25274 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/586274e1f29344578cf2f0423af5d0a3 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73400855b6b44f558e1367b73e2f8a46 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c71785a22054a7d9d2003774f4897ab : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a83e2222104240939650c5f96434c3de : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/130b62aa74ef4225aaeaac531566fb5f : FAILURE in 7m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b6932f86a8574468a2866536b5da3eed : FAILURE in 12m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4ab6b67ffa49426fa16737421a8a3690 : SUCCESS in 6m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1bcdad96486947e89b23ee6b3aa49501 : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c03e185fa2264582bd967e00dd687e6b : SUCCESS in 6m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/768ea25eec194748be4ab3c9ddaa0aa9 : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9a1cb5d2148f47809d89ae077cb6df4e : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a7fa6ef18d944a27bc5eb2715dd18778 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/077eece47a294971a9c2187e9db713b2 : FAILURE in 6m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/312dfc7494a644baa482f07bb122d079 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91a0af4cf76c414ab4ffb115bdcb23ef : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/971d18590d2744119fbd71e2379410cd : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d59aa41cebdd4749b6b477b32aab01ae : SUCCESS in 14m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/871e37464a904dc2856fe9810b68538d : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/41437d6dbfda4255ba99afcb6f537944 : SUCCESS in 8m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fc125eca75604081a5576e61cebddd1a : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/182d0d61ab3e487886dda5d284a31d67 : SUCCESS in 9m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a9feb4c9c0cc44b6a936dd93faf875c4 : SUCCESS in 23m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/112b50120e38440f8288725e96822434 : SUCCESS in 11m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d10ed02614904738853a120bdfbeb10d : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f96134e821742c4b2e1bda746ac0f1d : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3773ec95872340bda208c1ed522d50b3 : SUCCESS in 21m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c713edfd876b45058b1cb0fdb77f42c7 : SUCCESS in 4m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ddcc5af740364acd8ed5531e440e8d08 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f43f4d340ec4577ac4e436fee76045a : SUCCESS in 5m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4d97ab9790a14f1396402457202f3caa : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b096c07568934170a1bc1d1a58f8f85d : SUCCESS in 1h 03m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dde54e47014494eb8ac4a68e6886a44 : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":2},{"id":"7476089d2d8026a898c2acea718f4178f2ffd243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 19:24:32.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"74b7309e2a68f2f9661f55189d2c8ff04e279119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:30:44.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa777676d85401e958058100cea8d5e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa349fb1506c47a6a6c0165bd85229bc : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"592eab7c6615a51c4f0e4ff9a3ad5a073c10bc69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:30:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98055e6b5cc24d15bfbad023679e6035\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3065f1b2b0944279b2f815bb22a9a0d : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93eef9acde414bdc9acb6521091e29d2 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03cda0fa869a400786e97f3022cec782 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24afdcacf8b847bd903559b8ff50dddf : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4676eaa3171f41d5ac524bc3b0c73252 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2153834bdf8342f4b59abb1789175a5a : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09ad7af52ea44787a8848ea918c5553f : SUCCESS in 7m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e5818c8ff3e43909b0ffa4681c3341c : SUCCESS in 9m 40s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d2c8a89f825a4f46b03ec8393510f74f : SUCCESS in 6m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/36f98fa8e99347d09b48e3387e3ead22 : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1eca83677a5943e6842dff5637e6fabb : SUCCESS in 5m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d821bf4570d149efbb592c0aaed688b8 : SUCCESS in 10m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fd86095f68af4e37b4b2532e18e9b4a7 : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e4c7c8048ece48038446e57da9d7addf : SUCCESS in 5m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c3fafac475fe4f39b439a424f484b4c6 : FAILURE in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/95162aa9c9624a98b5aa5667acf6e8da : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/182881f49fe84601822ae91d865f66fe : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ff52815fc84c4a8ebd9fd82d7f7e6b68 : SUCCESS in 7m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6274884fb00f4458b8d920c24fc0397a : SUCCESS in 10m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1e9de9a09e9b41ceaad27c789a2c02df : SUCCESS in 4m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/97550b0cdc674bf2a4120dbe7020a151 : SUCCESS in 7m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/960815bed08245a5959167f32f05a8e2 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a4909f048a564b539d1c8de380b1313a : SUCCESS in 9m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/df9bebddfae94f038590c8d4ac63c64c : SUCCESS in 28m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7ecdd43ec9194b5b9a3360ddf65a888d : SUCCESS in 11m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/96a1f63f212b4ed0aeea17904412d822 : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ac668b155c445a9a6d1bf487cf49c9d : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e63460487554cc7b2ff8fcc32c01cb1 : SUCCESS in 22m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f61c14966c19459d98932eee720cc236 : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4fc47f77b59648bf8d7ec76131edea0d : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/913f4d6471794e2f8f254bf7c80e1be8 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8fdb53eabb45425ea5241f7a887f461c : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b1041fcd0ba4eb792b6412d2be42bf6 : SUCCESS in 53m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afdcfc9c8ebc4200a6f1fd74593142b6 : FAILURE in 1h 02m 49s","accounts_in_message":[],"_revision_number":3},{"id":"0941edeffe642dfed372c92b183e27cb0433e87f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-15 20:46:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17cadfaf762e2b3a2a42147baec8b40383f13b46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 21:00:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/682dfc6185384d2884a84ffa63e144eb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4047ef285f0b48c6b156486619e86f6c : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fd6a671774334832aa58017b39bb96b6c3237435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 22:08:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e17b8772e71f40b2a4a5f06ef16b83fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94b4474fce546bb8dbfedea833447bc : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eeb08b9986247fe8d74a0cc96355670 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6321c9cca840099db14d4026b29721 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e37b95585fbc4a2dad5826f0535d5b09 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/880f23bfaaff44caa198d2df42d904a9 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/686b69b23d794c25af5241d320479651 : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f32e8a2cafe4f91a5274b9bb806624f : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0368d5cf2c544167b52b5d8d9ae5d8bd : SUCCESS in 12m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/aeec70b41e4d4073a74da4fa3b1eafa1 : SUCCESS in 8m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5402fbd3eadb44438f23ef6b19d1f231 : SUCCESS in 5m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/25937e91ab4a44778318a895d5cf2859 : SUCCESS in 6m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/69567edbf9b044a4890db0835faecc96 : SUCCESS in 19m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a6dd12a548044c75b499abe6cb008d4b : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d570658514194e8981c46f93298b38f4 : SUCCESS in 11m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b03520b1f69c4000b44f6b2903f5aa38 : FAILURE in 8m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/48706f7b97c2410a879b113c70d12634 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0ed394e075545c487bdef1c63cb27cc : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b6d612867f5a4433a585941a45ab3098 : SUCCESS in 12m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/01afcb3de7634797b35aa13c81748155 : SUCCESS in 15m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/46c564cda17140b1b06f922f01cf791e : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fe5c49d20f7d431ca5932c9d2d8e52be : SUCCESS in 12m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/dd3cdbf7c3b24816ac91201b562355e3 : SUCCESS in 6m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e8da30a49a854043b85cfaf5b8429cbc : SUCCESS in 10m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/01fc87d4d190448790ea864e3015b5b5 : SUCCESS in 34m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/451fe4d5da7942d6baea2d3760f3ede7 : SUCCESS in 13m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d8bc6d3bdfe8424ba72c198d9aadfbd9 : SUCCESS in 9m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f400553f0c1452ab41483ee2f0c14c8 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9746a4959ccd475a99c954e17085d300 : SUCCESS in 23m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9b1b9eb936c544cbb16d491f8455478f : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9a11a14f2d8544109e80645a94df081e : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4bd647ca092b4fa6b30076cfe6812f99 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/da824da7111b41cd82f00064420cf6e8 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8295f84953294962b8830da47f290e4e : SUCCESS in 59m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fdb61bd6d5d4e0b9fa7e8aeb9532a9b : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":3},{"id":"97b3b18249fe9dfe125c448f1566b64f75467f51","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-16 00:42:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12e18dbf5e5113f97183bac8d8b87d6486419f4c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-16 00:43:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65723e4e55782c9b01d0d53e808d5c5097ab2851","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7bd1adecff3482379016ed2ef20b8bba53b4c42a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:36:37.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65adf938329466288d1cc6d265b4ccc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/83ae6ee052e44e9486a9d3dab1eebfd8 : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"1866762c70de0c0448449e87047471ae81354f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:01:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c65af93537d2499092d646c4f9a8f5bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b40a4d89db44ae8c9694adc9c09c9e : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa97a4c5e2a43bf8d68c57610235893 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e119229aa314c239b6b2c1abad6d5e2 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42b4f64de7614426a9647cb2de6bc26b : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96f240429c38455eacd0633f84906ed1 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8bd192135ef44c78eebd5ade26d6a32 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af52393aeced4d34967f8e3259db951c : SUCCESS in 7m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/93cbc06fa0c64fc99e66bf54e6b9b1d3 : SUCCESS in 12m 08s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fc42c914144242b080fd212f82b26a1c : SUCCESS in 7m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2e0b0832e4ac45f9942ea9b3ae2f3474 : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f892e46fbbc6421e8d783487f753f96b : SUCCESS in 7m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fafcb8bac25c413bb7eaee51ff1bc20b : SUCCESS in 10m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2d4521646e3641a0a9e56189ce5e8fc6 : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f250e57ba4ee466ea028f78cd270d4de : SUCCESS in 6m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9bb14ac9c45e4863b450e0302a058d0b : FAILURE in 7m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/881948ccf978465082d6091294eb6b9b : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0cc26e99e6214a68a34012423f2793d6 : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ec32d6e7df5f4878b18e6da92a181dbd : SUCCESS in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f8120c3e1eae41439c42c607da12b6d1 : SUCCESS in 19m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/11c2f496f7d246fca363cbab548e727e : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7112204dfe734ad887cb3ac78061f6b6 : SUCCESS in 9m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/03e7214bd13d409592c5b39790c80993 : SUCCESS in 4m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/13b8eb90deee4a1b8349deb74922354a : SUCCESS in 11m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/898900d4ec5341caad825c6ef5311bde : SUCCESS in 27m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9e7f02fbee9c4c529e08cf7f50408c7e : SUCCESS in 12m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9ca1d3c12a3f449e8b03c7444e85ca85 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff75751a18b64c9f8a62f41870d42536 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/baa7481acaec43a89356b71d329b6fbd : SUCCESS in 23m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a43d16f3cf6044c189eb3d9ba62898a1 : SUCCESS in 6m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/03cb7a164da24400a78cb089419ab75e : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcf316a8653c4e2b9553e58353b63874 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/db7a099fc8cc41399db2d32b0d8de3e6 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f7db5b49f914ee2888498d71db71baf : SUCCESS in 53m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/794b1bffe5514926b098d1cdc1281bc2 : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":3},{"id":"84ecae8c4b75ceee674cedf292848991d41a7cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-16 07:11:21.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6ceb0313686837fb1ac70e7422898a04c47d46c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:17:29.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c041a7886e94b8899e09d809573cf77\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ce0035d693f4e36ab4ee2a20733ce8d : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"4e7b971aa909e942602c82a5a349f49498c40397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 08:58:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448e5d1b4d8c422db8433d883561432f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fc942552ff4696b3397e509cb9fa5e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b8c9b0a467a4e86869a4ce8061846e3 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e336d6f128b5497fa1eb05faf52e9b77 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81d2f7ead6484297be2b1de226d821f3 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/116882d7c50e4877a11487ebee83d1c8 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b22b6575d8243a3b8d13ff255832d43 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85df299f56af48e8aaa3b875b59e5962 : SUCCESS in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df6fb98bafc2426187bf51370e1f9646 : SUCCESS in 12m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c3057774e6f64d4e875e7d440bfdbbb8 : SUCCESS in 7m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8dc8c88292344f85b9258ac336270b12 : SUCCESS in 5m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cb0108d9c0c84f059bf0e6ce2632c9ee : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/36bfa61895034ce68bda449dd5e664f4 : SUCCESS in 12m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3d7ec3901fc543d7b0fff962ab643482 : SUCCESS in 4m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/778ec70cec36458f828b581e486ab92e : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b1092d3d5d844fdf953ecba1e0173081 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/68a140af47b84a8680cabb31faba6cab : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bdba0c2883243e8997dffc9064a27a0 : SUCCESS in 7m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3aa3bdd800694758b62cab6b82b1d810 : SUCCESS in 6m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/04e5c8e375c2473881c4c40cd2a1ffff : SUCCESS in 10m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ace38d58db274811b7b72ca58e15ac6b : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ac676f1674434ec4bd6aca8d3735cd98 : SUCCESS in 7m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/108dfd0ba97b449a9a628baff24b8e69 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f528984707044c7e9ac1ca3d66ffc6d9 : SUCCESS in 9m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/283fff9a987f4091af3852674436a5bf : SUCCESS in 23m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/299a222b7b72463d89a91ec8ba4f0b8d : SUCCESS in 12m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0432f3c64571441cba2cae3ad474aaea : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a64b4f24c0342118c9c087c78ee8c11 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00f2428399b340d7b034d20a53b3d466 : SUCCESS in 24m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/828f8259e1fb401c8b93375e267d0288 : SUCCESS in 5m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1a8242c837404bdeb22e8ec2ef2144e8 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac55b81931cf4269bc70ce5430053084 : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa13b4ebe6ab4a7da8e5cc52f38e5565 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f28da66638674d8d91a87fae204219b3 : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2112125d1e744d4a1e54a9873ad8e2e : SUCCESS in 1h 44m 41s","accounts_in_message":[],"_revision_number":4},{"id":"822422c5dd9ea9a3050b8a1cc6553a5ad07d6487","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-16 09:26:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d45478ef62c0e755f972abc2d9be74916c344b25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-16 09:47:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3f7d4ef63c5e9ce014339570e9981700c8b2eaaf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-16 09:49:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"512f1caa3e952cdab932a721a7d512b430923b50","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-16 12:07:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c169b6f772042158a66a2fb5e88502d8559b22b6","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2023-02-16 12:07:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"90b999dcf0219cbad6f30ea40cabbe65701b0feb","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-02-16 12:08:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"00863727974c4967cdf191936e4727ed459a8a59","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53a05e3bac1dd851c907396c1ad344e0a8c926d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:00:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"29bc69459da2353013dccc6fcd04accf36805272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ec8b6b429534a1fb454035cf727bc1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05d94bdb43044c38bdbe28bfe385236 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ca62ba28d44394a63fa3ad0b1be80a : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aac2316c32264c27a19482979d3318af : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76f36d9159bc4fcc866c02bc0286262b : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2a80f06f51b4501ad7f10cd5eff09e2 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdcc44140d8d4b788d1eba07404a5a66 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ee89edd3eaa4118b4b1291861316f27 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/490f8c211c384e44bd40233f5b38afec : SUCCESS in 12m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/53b8e1fb502342f296415aff7ea47c6e : SUCCESS in 11m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c89db11b86cf4af48ce3f713c8123982 : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e87dc623e3904ed7b75e9a8ff1bf5905 : SUCCESS in 6m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/36210e86f4e641b090d636dc4702c79d : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9f44520bdf874a878c940ad6d2b14e6f : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f93f70a2089e48579ec3c3a110d38d55 : SUCCESS in 8m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a5cf9b8ba2ca457cb80f145205f34d35 : SUCCESS in 10m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/13ce8687fda241fd94d5c63231744889 : SUCCESS in 15m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4473f1c283ea430bacee5e5430111017 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6180667b84ca478cb7ab68ab525a097f : SUCCESS in 15m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/89c5f334e19748f786c9ceb17ac8b120 : SUCCESS in 4m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/24982e5dc1994b3dac1c3133dde923c8 : SUCCESS in 23m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/15ba1c22c92242dc9be1d43e0be46c3a : SUCCESS in 15m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9829463c19f64ed0b01467cfce4aa98b : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6a1af41f0fa43cc80fd1307e48d2350 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e0417cfeb1f4cebb4459bcfd88708eb : SUCCESS in 38m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ac17be882ed843ec8d6acd407ccce41d : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9a1dce191412449cb0fe0591ced0f5e1 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4dbeacf79374b2998c592da917db47b : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c3614a8176b14d80b8f8e2ee67a7ff04 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28f8c115b0284ca8833cc65caf569b7d : SUCCESS in 1h 06m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/800b2a23e4324633b6a372d5be48c72d : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":4},{"id":"16ffc69dfa0173c06768eb56d4845be8828307b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3e89cfe14319b2b7cf9ed4afc919eae6fc653eee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:15:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad8e913ef54044bfa94a7fda88fb72d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487221ceb3074611b917a5203280f9b1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"055b5f5a52983f8dc9231588c6564ff494b92c94","revisions":{"3e47ed18c3b341a5003c86987ad9f4f223ac8621":{"kind":"REWORK","_number":1,"created":"2023-02-15 13:18:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/873906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/873906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/873906/1"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 13:18:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 13:18:03.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.5.0","message":"update constraint for python-barbicanclient to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Change-Id: I81ae50271bc866a425b079f25a86cc512fd924c4\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Ade Lee \u003calee@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7a83d4647dab841abc4c1d5452f677170b5763d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e47ed18c3b341a5003c86987ad9f4f223ac8621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e47ed18c3b341a5003c86987ad9f4f223ac8621"}]},"branch":"refs/heads/master"},"5bc3addf1470b268c7e7a09039189de1aba230bd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-15 16:49:39.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/06/873906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/873906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/873906/2"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 13:18:03.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-02-15 16:49:39.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.5.0","message":"update constraint for python-barbicanclient to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Change-Id: I81ae50271bc866a425b079f25a86cc512fd924c4\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Ade Lee \u003calee@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/873947\nChange-Id: I7a83d4647dab841abc4c1d5452f677170b5763d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bc3addf1470b268c7e7a09039189de1aba230bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bc3addf1470b268c7e7a09039189de1aba230bd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a93dfd5909a766bf65981b85b7fbf8de6a8a93a3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-15 19:24:32.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/06/873906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/873906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/873906/3"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 13:18:03.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-02-15 19:24:32.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.5.0","message":"update constraint for python-barbicanclient to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Change-Id: I81ae50271bc866a425b079f25a86cc512fd924c4\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Ade Lee \u003calee@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/873970\nChange-Id: I7a83d4647dab841abc4c1d5452f677170b5763d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a93dfd5909a766bf65981b85b7fbf8de6a8a93a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a93dfd5909a766bf65981b85b7fbf8de6a8a93a3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"055b5f5a52983f8dc9231588c6564ff494b92c94":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-02-16 07:11:21.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/06/873906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/873906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/873906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/873906/4"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 13:18:03.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-02-16 07:11:21.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.5.0","message":"update constraint for python-barbicanclient to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Change-Id: I81ae50271bc866a425b079f25a86cc512fd924c4\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+1: Ade Lee \u003calee@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/873970\nDepends-On: https://review.opendev.org/c/openstack/heat/+/874045\nChange-Id: I7a83d4647dab841abc4c1d5452f677170b5763d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/055b5f5a52983f8dc9231588c6564ff494b92c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/055b5f5a52983f8dc9231588c6564ff494b92c94"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
