)]}'
{"id":"openstack%2Frequirements~873933","triplet_id":"openstack%2Frequirements~master~I32ac6198721e58a513c6eecf3665e427cfaf9374","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I32ac6198721e58a513c6eecf3665e427cfaf9374","subject":"update constraint for python-watcherclient to new release 4.1.0","status":"MERGED","created":"2023-02-15 14:32:56.000000000","updated":"2023-02-16 22:14:07.000000000","submitted":"2023-02-16 22:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873933-new-release","meta_rev_id":"d2d78d7843e48178853e000023bbde751d183d99","_number":873933,"virtual_id_number":873933,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 22:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-15 14:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 14:42:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-15 14:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 16:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 02:20:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fb280c0472ec80c734ed19263bc3f5b9f789370c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-15 14:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ade6e9764d9f58bccb81ed0303d2d280970e83","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-15 14:42:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70d2b08e42b19ebe36943f80195e93bf57917ec5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:47:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f897f86e3a54a8d97f9fdb21437c3a8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e9dcd68cc26431e8960f84ae143774c : FAILURE in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9c3e6fe5fe6d8b72fb393762ede7414ff59f449a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:06:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a80653c5557d4a29aad81d61529f9444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38387497c0649e5ade50debdf859724 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337828deb3d54349a528823cf84e0824 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21474a87643347c1911b10888fbf6eea : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ed3ffb90c19467ab43c51b625d90858 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa2400635e4741bfaf8936d9ec6a4e72 : SUCCESS in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e949b0da77dc46f580369a48f09c9551 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf08f910c4924d9abb9d76cf9ac6a1ee : FAILURE in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33adca732752461aa9b18039349a338d : FAILURE in 15m 58s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e45767d9e08a4bf3bc3a1f0828dc9ceb : SUCCESS in 7m 24s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2a888a98845446b18fffcec9cae0350c : SUCCESS in 5m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/19bfd46a0afc4b8581969e85f483bb48 : SUCCESS in 7m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4ae08ae063c84e45a6e002436d4ca6c4 : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9b22ad919b33427fab6e6c0919c01f36 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3dbc3e81ba004f0098e09611f616c675 : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c68d5d64259446298cd3cf790fa88624 : SUCCESS in 7m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8799142063924cb2b87e89d47432b3cf : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729c006cd37d40aa8d5cbc9bdd83b80a : SUCCESS in 7m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f30039e197814003b29125cca8181f5a : SUCCESS in 8m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a63c3e30fab94b9fbba0f42a182a2a93 : SUCCESS in 12m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cefbf889a2f14fa19f2a5c1ee7fef67c : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9f8896ddbdb64c2eb173d055392f9bc5 : SUCCESS in 9m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/874764b1fdf34e9abeac21d503f43827 : SUCCESS in 5m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/267ef4eaa20d47e6a68e4785667f8e2a : SUCCESS in 9m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b788eb590f2e46d2972cf4f4b29edc62 : SUCCESS in 30m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b7317957bfd74f45a4a8f9f6276e7124 : SUCCESS in 11m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/79bf1925f59f41759a9ccaabf0258406 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8be498baec544a709230fc626e2c8412 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34203ebf43dd40239385bdc872636148 : FAILURE in 23m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0541f039e6bc424ba47dd6b6fb39a9b1 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f647fcd3a4f841849b447b6e28d6dfde : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3fadf5535ab4b198dd5170560e69a17 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0c2d9040efc4467eb52d0ee2bc74519b : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f68842a78df47b9bcd63b97d46e59a5 : SUCCESS in 1h 04m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf89b9c272ef4c478949b6fbcb3003cb : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e62b723a8b50e706f5e7900df1d86fa73f820a84","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3883f6ed5f7c63ce95550e09e1bcac4503b2539","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:36:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4638809d7b481bbeb5304ca48faa3d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ced2a0574f4a42b09794b16f61fac51b : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a7ef01bc5adbc5547614153c38130145719f186f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1373fd777e8f427a8d21a71010a8e49b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d78e25ad0744ad19d3778be88f5b9a8 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb7ff58dc4d4570b090142acdf3b292 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9e255c6361a452f897779f68a460f7b : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcef4ef7f64c4d7ca34599e33ce64d3b : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f6b109fb01d47fab13deb399169fac8 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61b7294f95ba4b6fbd39af2155201ae4 : SUCCESS in 3m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c54565e423614e1fafce99ecc5e1b2db : SUCCESS in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5e4125c1c62e4bfd9a49229b0f63f73b : SUCCESS in 12m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fd3e66f445ad41e19e2983547473991c : SUCCESS in 8m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/399bdc34dbdb4b3c91740585c7648197 : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2f540fef7e9c453bb327aec8eb586725 : SUCCESS in 6m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ca0a280571042a0be2fc442d51f8061 : SUCCESS in 12m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9127e3c8cd04f21a187e9bd3937327c : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4c17168bae044552b2dd34a9c6c140fc : SUCCESS in 6m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/38dbfa82ae984d61ba3ed47dec9a7364 : SUCCESS in 7m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/10892f71a5a5473a9087ef8fb6a3ad9a : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b94c8bf1e4f400bb31038acbc778857 : SUCCESS in 12m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/775f1acb937a445f847f5d7f2ecb456a : SUCCESS in 10m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/147d209500c4440c8ff8a20b677c15f2 : SUCCESS in 12m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bbb39e06608945939f9ca429e0e043f6 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2644aa1756944abf88047d97aee6e77d : SUCCESS in 8m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6abedadb1b9e42a6817efc2d07b1cf45 : SUCCESS in 4m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/42a57c2bbe904d02b34072d742d63369 : SUCCESS in 9m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4e5b2d60e19e4bd3887b72c7020471c4 : SUCCESS in 26m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/21c6a0d1726146b99ed52bce72adeb88 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/48a4419cc4c1471385f26e0c54260370 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04173bd9dc8d458cbcbe28372f4179f6 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a4c50e18cd2449aa56258489c2f2fb2 : SUCCESS in 25m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3d63157efac540228572aa67c88bca92 : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0962cadefb0248ccad0c4a83f1e5ebd3 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e50e35ba3c684f369ab9fb1c0b6a1d06 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f54d477811540ada7e52e555426ea3f : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c207be3b2f634a2894eba789ad0701f2 : SUCCESS in 43m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b805f153eaef4acd9eabdf2e6627a2eb : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cadabde5696a2c64d043a97cef55efaf53873502","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e122222041f0c8fdd63b868d2541b1e4de340264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:07:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b5e7f430d8c254a51e5ab2cd5b9e2317894c212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/999c4e0453e84b94b0899c87a90ff0e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0e187f63184abd93a92c34a71b7ed4 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc87423920b445a3b6b8c75b0f0fb1f9 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80e33cc188bc4641a035a135435183d7 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5147399853ed45dca4d77180d05735b1 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f518b61f5bf4f1bad87caf4b8307297 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29654cf0b1794308837acecf150c5542 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/326c0406f2a8470c9246c0611e3f1de0 : SUCCESS in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/689c407f48b54856833f7a233e8a98a0 : SUCCESS in 11m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aee2b418903843aa97702fb71d893b44 : SUCCESS in 10m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/356e8b6531534b8aada5f31d2ca61a35 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d58faa6fd6cd4f469ba595691993c584 : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8ea517461f9e49a9abe7e6d18601a73d : SUCCESS in 7m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/245a66a828f04deb965863c70633328b : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd57745ce2a04791bec06ffb64714659 : SUCCESS in 6m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b0ce022c7bc44d96836b7033de13b19f : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c9c6ad0f8b8642f1910cc2bc7083143d : SUCCESS in 12m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/79041e9f6056460198dd8aa51113b1bb : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6bf772bb4c0d4473bbc4734945bc77a9 : SUCCESS in 8m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/53706be40a9e4d8882e7ac657a1cfc38 : SUCCESS in 4m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2be572e484ce49b494b85d0113495346 : SUCCESS in 23m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/629f545cf9504e02a988ea85ae30d27c : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b4394f652d00402ba7e68086479c9900 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21945deadd2b4c9698966e6ad1b3aef6 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fc5d0fafd744ec7a617606089843c6b : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aa1cfa7b42aa42a9904fc85cfd5f928f : SUCCESS in 5m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7a98ecfdd0524491bfd9ae09a3385dbb : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee4e70570c32407f9f82c9dc280b1096 : SUCCESS in 6m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/418a46c2a36d4bec9275a6f812fe200d : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3517e630dd5c414ba6df9c5ea80ead0b : SUCCESS in 54m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77c983ebf2cf470bbb196c57b0e649f4 : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d2d78d7843e48178853e000023bbde751d183d99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d82ec4fbb7682d7ae20da7ecfa01d4d80a9048d","revisions":{"5d82ec4fbb7682d7ae20da7ecfa01d4d80a9048d":{"kind":"REWORK","_number":1,"created":"2023-02-15 14:32:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/873933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/873933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/873933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/873933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/873933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/873933/1"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 14:32:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 14:32:55.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 4.1.0","message":"update constraint for python-watcherclient to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: Id67790fd2c9185f23547710d9d52e528a5be9a3c\nmeta: release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I32ac6198721e58a513c6eecf3665e427cfaf9374\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d82ec4fbb7682d7ae20da7ecfa01d4d80a9048d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d82ec4fbb7682d7ae20da7ecfa01d4d80a9048d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
