)]}'
{"id":"openstack%2Frequirements~873935","triplet_id":"openstack%2Frequirements~master~Ia930f26f60ac720a21b832b2d01aac60c91ddec0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia930f26f60ac720a21b832b2d01aac60c91ddec0","subject":"update constraint for python-masakariclient to new release 8.1.0","status":"MERGED","created":"2023-02-15 14:49:54.000000000","updated":"2023-02-16 22:16:01.000000000","submitted":"2023-02-16 22:14:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873935-new-release","meta_rev_id":"a0722ccac1d6b41adf2e2dfd0411c79f4a60d0f5","_number":873935,"virtual_id_number":873935,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 22:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-15 15:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-16 14:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-02-16 14:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 14:57:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 15:44:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-02-15 16:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 02:20:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dd8be382d1f77e7bc1ca343990182480542d6659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-15 14:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02f13b70ed366c995f04ab147fce57b6d4586e62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:57:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a27513acde0544ff8d5fff86a94ebc89\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c4672d0f2fd84b8ab95da6e0fe14e243 : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5a7edd00a073e7adfba56a57b8fe13b9a614b6e4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-15 15:44:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c23d9ff7fff5b8ea7ed84ac0f68b2f5b3723355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:24:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c1fcbcbd8284561a834a847228d0234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3fa1e08bdfa4bc5a94cd4c2e169f8eb : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32a45d429f164a44bb6c45ddd62da2f7 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/024525cf69324037b19d5c79e879fc07 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c37e3d094c8472b99629299a6fc5ab5 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c13071062aa4201a79d07f525d52fbb : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69c9859db8924e48bebc5e8d55ec8f33 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29a24cca57784db598cdb8b010406fd6 : FAILURE in 8m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c5acb74571574f4ebf75fbb5a9175ef4 : FAILURE in 13m 51s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c5293c6637c4410db58fd0efa83e0532 : SUCCESS in 7m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/497e06351a234a5f9f2a437126ce6532 : SUCCESS in 4m 46s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/310f26d79210434b803507ea1300ff66 : SUCCESS in 5m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/559086c3e2d944fcbd13ba6c8663d835 : SUCCESS in 9m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f47dd392f8a74f3f90454444b2e9118a : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4d4bd01709ec49fba96ce8cbaa9827e9 : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/318417b704464df5bb59867f1806d6f8 : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7988ae44866345de9e1b65390e4d33d2 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d87d296c9e23424a82904c70507baafa : SUCCESS in 8m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2a997df063034a8f9aa4938f12b34288 : SUCCESS in 12m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/738a8d483e994a6080f2dda686581da0 : SUCCESS in 13m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/177f6ec8dbf24e9792e21bf0fc3a70a2 : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/15304e3145684e95844f94b5bf4376ef : SUCCESS in 12m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c77b00c250164017b5463eb99df70e38 : SUCCESS in 4m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f6387fd7110643319e4ecce12686bde8 : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5815b6941ca440d1bed76577073ecb54 : SUCCESS in 25m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c358b86347de44558ecc35e143887311 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ff5afbccf2c644c0ba2fb87401b09ac7 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe9aee2d08414bc19a8b9cb41426ed96 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7bd458f1023447f4b508a17d90263590 : SUCCESS in 23m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1cc81af20597418eba170659a3c26bd9 : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/860c68cdb88f4f359208fd14f9500b9d : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87de2e0b0d354d61b49de6d0c92227a3 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e00c9e9e8b6d44208bf57dc1be45a854 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d572f36394f84652bade995a995b1321 : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f39d73e2c864bfb93888ca09db031b2 : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9e11b5e4751490938f93dccdf91c77b2bc8f0382","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"223540de56523baf3ad381a24cec1da4f6e4bfb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:36:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639717f2dfe540b892e8044a674f1e37\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0435df7ecb0a4834a57ad2c21eccd401 : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b0b485aa8a58edb1ad2189620456c38274da633d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 07:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5aa5fb37d014901b5fce48ac474a3f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ba572079b848449f4b1c5f3f8cbe5f : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8addf2455ed24766a09000faba938d4a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5db34d88380345acb180a898877a1d98 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67204d9536bd4531b1afda47409c190d : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfdbc8af22d140ac90bf90bd0f406413 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66382b3af61f499db9f0278641af2614 : SUCCESS in 4m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9b7e20f303d43809937c884850a3266 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e0b0fcf0235c41e5aded10c515845eb6 : SUCCESS in 13m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e409e30db1874034bcce172d8c682340 : SUCCESS in 10m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/46dc7215ff1f42d2bde3bc9edb2fc550 : SUCCESS in 5m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8a97ff35adaa4e00b1cbff0439922c44 : SUCCESS in 7m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0f55932b66d140759591d1d10be1c013 : SUCCESS in 10m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7ed234d10ff44bc9927199324e084b75 : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a9e2c597a8734197a71ecf0e50a2a2e3 : SUCCESS in 5m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8084df37111546f485c06e8604e7808e : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/003a36dc0f8c44c68ea14c4b68b33c8e : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/383151670e7049fc9dd542950357a7de : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b28fa966739f47e6b1bacfab7c86f6f5 : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/238fe95d4e6a4e72aa5d35d48a0b6b55 : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/48ec084e52b24fe191cc7fa896ec2b74 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0ae02efcf6c242599c3492897b2f267b : SUCCESS in 9m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1c524f97371b4dfc83ddf33ab3ce5038 : SUCCESS in 5m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5d5e6b1a7020409ca8794cbc8cb242b3 : SUCCESS in 9m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b57f2e8b63e746a3b3e2791b7d055d8a : SUCCESS in 25m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f749383ebd0e4d18b4d73c7c51c2a68c : SUCCESS in 17m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ab91076885d64daa8e61fbb21d25e3d2 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d0a66b92c434c779c3d19dc6aedd491 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d53af8dcb791433884d04b1baa14f007 : SUCCESS in 20m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a229d4f40fd94e719ec5268f76e74638 : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7b1881de048b4ae9afa56ae7c427bcc1 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7026544b9b104840a42caa1b417b96c5 : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea3f3b8a37204767aecae7024930e113 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fe331013672483c9477744b6531982a : SUCCESS in 1h 07m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e84aa05f2f7498b9693d988a3596a6e : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7e96e49f360a5e3941497b7a493226ba567ed198","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c7ce67b5791fbf92bf02e88c7fde358fdff910a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:08:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"208fb5a573145ba152d50b1cc5318735c6c108f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11110d3a89a64d069b0d0fa8837cca03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4b0051dfc04ffe893c1cbca3e9cbe5 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926df6b16db74e5399597125cdbc7da3 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6c557cb489745f9b6f5ea3859d878b8 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/018e50e051e44a7f8864c8f206f3c00d : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ecbae7aecbc544dc8d7a3e6889ea716d : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2afe82a07e624669a00da0749fe010e0 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c96f94449724d43abe07f00f4231e83 : SUCCESS in 9m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b0fa530e7304b418b8468caff8006ce : SUCCESS in 22m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b29ff5843d86475887f4d70e6ab126e5 : SUCCESS in 9m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/207f41b5eaa04a7eac32efac95d39247 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/eb287e2c239d4768829b207048117e35 : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f8d233680a9a4b00b72175d69d1fa057 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b321dc137ee34eb29d3e5cf3e8d8215c : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10f1ee7edd6b421282fcc8c18f280170 : SUCCESS in 8m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/77760cf4bf9a4bf39e6d06d199fd588f : SUCCESS in 8m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c174dfddd6e471ca66ae55081ffc0ba : SUCCESS in 10m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0d924c3988254d2ea92068866a5e6ce2 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2efd4f795118459cae043c5a22d4a56b : SUCCESS in 11m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/484fa5bd96ca41838c9075920e82ec11 : SUCCESS in 4m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5c1a2a2d3e5d493583f93882df2f2649 : SUCCESS in 25m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9ebd2dca3fe24a4bbb98431855695aba : SUCCESS in 13m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2b81c7cc2de44a379021b51cff1c97cf : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6af01291c894cacb0ec6ddc3561da77 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b9a3c1d27ad41609d1bd7c3ecc03be4 : SUCCESS in 25m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e108923831a34549beb11a9f9a327f44 : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a2a9da098c2e45449557c558dd8be74e : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0952e9a70f94981b3616a7bf9ca9a3d : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/85df07e360554644af35d80a12766f75 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cf3d2769701404a948b6f57ba140e97 : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d8f490bf49e4b2f8e071e9e01e2c5c2 : SUCCESS in 1h 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6a243863778651a609f3fb52fd6ff9319117c575","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:14:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a0722ccac1d6b41adf2e2dfd0411c79f4a60d0f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:16:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b2a145cb764390b0c08b3d37e1346e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882bfc3392004a1daa8d5a30b96d2c9f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30d9a5beb9936a4ff8eb09711485b4285e174f9f","revisions":{"30d9a5beb9936a4ff8eb09711485b4285e174f9f":{"kind":"REWORK","_number":1,"created":"2023-02-15 14:49:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/873935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/873935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/873935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/873935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/873935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/873935/1"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 14:49:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-15 14:49:53.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 8.1.0","message":"update constraint for python-masakariclient to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: Ie8e4545e4a468cfff2efde2ffb1372e15a1ff739\nmeta: release:Code-Review+1: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia930f26f60ac720a21b832b2d01aac60c91ddec0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30d9a5beb9936a4ff8eb09711485b4285e174f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30d9a5beb9936a4ff8eb09711485b4285e174f9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
