)]}'
{"id":"openstack%2Frequirements~874161","triplet_id":"openstack%2Frequirements~master~I7c7f32b1196d10a627fc78848a554c6d03bc93b9","project":"openstack/requirements","branch":"master","topic":"uncap-virtualenv-setuptools","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-09-03 16:30:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7c7f32b1196d10a627fc78848a554c6d03bc93b9","subject":"WIP: Revert \"Cap virtualenv (\u003c20.18.0) to unblock gate\"","status":"ABANDONED","created":"2023-02-16 19:56:40.000000000","updated":"2023-09-03 16:30:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"87dcbe63fc840ef75b18946070a7a123e6308efc","_number":874161,"virtual_id_number":874161,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-16 21:31:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 21:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"beb2932d3d40be17281e11e945c167e1262bb751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-16 19:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5ccee9ae551a3c9ff40d2f515c6f74d6a3c43dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:31:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ec8942835b64b1fb55aa3a984e0e8c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923e587e414f498e852275ece6b6bf79 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c37cf7a7d704de4a19bbe03fb3f65b6 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76fcbec6e6e4ff999eee811eb63f865 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78a24d3b130c493caf913d693aa36322 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e849383797364de38f8b02f7abe20e05 : FAILURE in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2e5b6eff84f4a81a25202262e395995 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16b304d8b3d84f1baf4cd9249ad32469 : SUCCESS in 59m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b66adf5340234968b857df1a49928108 : SUCCESS in 1h 24m 18s","accounts_in_message":[],"_revision_number":1},{"id":"87dcbe63fc840ef75b18946070a7a123e6308efc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-03 16:30:55.000000000","message":"Abandoned\n\nbundled into https://review.opendev.org/c/openstack/requirements/+/893403","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35d4d6a04234cfa4c0899f9c9d1152b7b502b900","revisions":{"35d4d6a04234cfa4c0899f9c9d1152b7b502b900":{"kind":"REWORK","_number":1,"created":"2023-02-16 19:56:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/874161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/874161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/874161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/874161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/874161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/874161/1"}}},"commit":{"parents":[{"commit":"9e30202655931111ef3a0e0cbd2317403e413a6d","subject":"WIP: Fix compatibility with latest setuptools releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e30202655931111ef3a0e0cbd2317403e413a6d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-16 19:54:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-16 19:56:11.000000000","tz":60},"subject":"WIP: Revert \"Cap virtualenv (\u003c20.18.0) to unblock gate\"","message":"WIP: Revert \"Cap virtualenv (\u003c20.18.0) to unblock gate\"\n\nThis reverts commit 7c3cb297573b5dbec634e53829ec183ed7ba3068.\n\nChange-Id: I7c7f32b1196d10a627fc78848a554c6d03bc93b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35d4d6a04234cfa4c0899f9c9d1152b7b502b900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35d4d6a04234cfa4c0899f9c9d1152b7b502b900"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
