)]}'
{"id":"openstack%2Frequirements~874222","triplet_id":"openstack%2Frequirements~master~Ibe3fe3fa81dcc459fd68abf2ad1bbd6fa50cb1b0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibe3fe3fa81dcc459fd68abf2ad1bbd6fa50cb1b0","subject":"update constraint for python-neutronclient to new release 9.0.0","status":"MERGED","created":"2023-02-17 10:28:28.000000000","updated":"2023-02-18 04:33:31.000000000","submitted":"2023-02-18 04:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874222-new-release","meta_rev_id":"902e221c7d8d1cac585dd09ec1b15d4e850dad76","_number":874222,"virtual_id_number":874222,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-18 04:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 15:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-17 10:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 15:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 10:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 10:36:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-17 12:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 14:14:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-17 15:50:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e551072139ecdb48844cfca73953b47ad0218e28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 10:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc458413525e8f7bf79ae660c00322c130b899fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:34:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8243958f60c4cb78f7889da403219fe\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7f6f09ff08fb456bb4961e4ce8edb3b9 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"be8f62d4065db36755476d9d8ba50f096436539a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 10:36:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50864845773ae520e17612396f4783723f0e9826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:00:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c35178f24f434fb3586ec63fa473c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae18f6d776c483197a8571dc1a0c495 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955e5e7439534112a1bdf97fb76f6be3 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a51dd7a939814c46866e997d39657ab4 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9022bfb655c4921b9cbe00a1e3f7d36 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6fbf5230b614c0b8929a1171d82d315 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81cfa049096542199f879eb6d1266b2b : SUCCESS in 8m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9cd9e4e98e7497ba694fb67422128ca : SUCCESS in 9m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4e8ba51606dc4dfeb29f0b9e200ad33d : SUCCESS in 11m 52s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f355b17b883c41c499788ca555fcf5d6 : SUCCESS in 8m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cdd8ee0b0dd64c4cad7926727f3bb68f : SUCCESS in 5m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/676c9c8cd9304d87a4cabc46011823b7 : SUCCESS in 6m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0132f4f501b54febb0a4b21133fcc7c0 : SUCCESS in 13m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c15a21b0eba64414a14678e729657b09 : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/398a170ed4884971a19db5e930176f64 : SUCCESS in 8m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/60431dd443d9447f8a4c0678fd324b0e : SUCCESS in 9m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8cdad5a7a9ce4adea8631fec0f5c379f : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c449da138b9548a29f94a8861edc6fba : SUCCESS in 28m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e814713767774d6ebd9267db4b9e81f3 : SUCCESS in 11m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d15e95176ab645ed8443ca9a8535a5be : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2ae404ed6c334051820dad3e63a00251 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4494d896fa1943a9b34d29bf3df0882c : SUCCESS in 10m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c920d8db931e41dca4027014e4263664 : SUCCESS in 8m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ec3c2946f7e549dea2fabdcf069da6c8 : SUCCESS in 9m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/35c6c81a65c74091b149434f12408cd9 : SUCCESS in 35m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/606900c7083a484fb0cbee8cd778e7f5 : SUCCESS in 13m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3ace8ab207c94a26aeffc3f306e5fe47 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44b1b41ed1dd49d882dfe304ef70f1da : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ceb9a2f1e8546c5a20204f91c92b13e : SUCCESS in 22m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9327f3d512ea436b9e4db00acd958f0c : SUCCESS in 6m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6b8cf3d8974b4f559078d202e4d913f2 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90c726132aa142308a09a65597d1eac1 : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b377e9c54196418a8d598f8e152d438a : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e460413cb0d04a198afe824195247a92 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/867d246e134b47aeb664585183fd01b2 : FAILURE in 1h 08m 38s","accounts_in_message":[],"_revision_number":1},{"id":"723d74558444f80d26fa721cb0af834965fc4afb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 14:14:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25fc9127ff5e4f76034b8dfa54b9200ab0c2340b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:20:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a5e878a807947629bc63e1caa5939ae\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0f066211e7a64d759e1ee332fea58bd8 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3ec2d3f7bec356c27d942c4d26a092b6a1d3f08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:37:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca7819216cc4df8a2670b2ec89c0ae8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0100b8dec26c4292b8f77f8fbb4a813c : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f25c86fb9104cc58b02f55d7ef123fb : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92aa158eaa0b4ba0baa1e5c485d2110d : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2530199f1ec54ef084371c915ffc170a : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bd79b71dc664551b8509a0f05be3342 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b867c3b8c8f049b28edb4afb65fcd9e8 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32b91f3f0ce24b768e4a18a6ae651d87 : SUCCESS in 8m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c0d4ead562c24152bcda6bb826ffe1d4 : SUCCESS in 14m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7777974db6a34b359d722ce607242564 : SUCCESS in 6m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7d743b6766044e9f8dfbad3c55a08a4e : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/78faa87d9848455a88ccd72367397672 : SUCCESS in 6m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/43c79772d5764a459a5e74a8d5e6a378 : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f6b6dfd657f34916a9789df7e6363e9d : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2cc923a66ada445395bc4faffe477b61 : SUCCESS in 6m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2bbbdab4ff194a66a7a8635e7fc7bab9 : SUCCESS in 8m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/27795dddacba421c9e73bc881a126832 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4330947d1ebf428d80dde35cb6591eca : SUCCESS in 9m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/063a14953ecb4eba8fe10fa3b0d04884 : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/025015985bf640c6845e0ac65a5aca22 : SUCCESS in 14m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9198b14b36814eed86f63ff853e6930d : SUCCESS in 6m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b5e8148aa25c49019d944396a6acf7f9 : SUCCESS in 9m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b94d37c8fbf742e7a9e2a26e6e65b01e : SUCCESS in 4m 48s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/93ff1d1aaf8e4f0c89f8062aa6802e25 : SUCCESS in 11m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/09eacc31af214014805453a098e21966 : SUCCESS in 23m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8d94574b3194fa09b4e40487ab304cd : SUCCESS in 11m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9f96e51534494076a5549231ffd2dbb5 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/289d09f943594d1c98442595466d350a : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d337456ce7244c1a9506fcac2decf52a : SUCCESS in 22m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dbb43f40b04d4b22ab6dc963553fe261 : SUCCESS in 4m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8ea1a9a867014312968416f41ea0cb23 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd18e2603e914ac1aa20858109022361 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e7675f6c2b04060bde1d9ceb9abf7d9 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88289829a2e44bc9a08627258464e9f6 : SUCCESS in 1h 01m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74cba6eb7b984dafaecc3cba94fbde42 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a475ac5cb046385fe70ef911cbc60a6a7a466c5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 15:50:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77fe2de94fd099bf234dde69215017dc29ee8c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:51:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aefe8e03457b58647b9b4907fd70fdc73b5b6ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:05:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc8d798515bb4b83ac5ed16af84faaf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef30dea67524d378e52ead986c6e453 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f68da1ceae4befb034faad8d42060d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18c2afad0a244ee183c22c967fcd6e2d : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f73fb232cb7d468c9ad90c3f8cef301f : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a7147589dcd4e5fa4cea300643fe381 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd9a64e0be16439682d7f2e893eae318 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1bb6e8999b1f4dd4a3adcd62b75be6d3 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/676c3d64b65042829d2eddf951697d47 : SUCCESS in 15m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/02105428a8b2495d98a03f8d47335205 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9c630fa9423f4f3cbc9c4fff3cb37a71 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9d3ab76492744f2b8800ff1db66e56a0 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c360cbfd5d0b43f2b23d90499df9384c : SUCCESS in 10m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c748acfe6f9d470e88d2f2d25d7f0a9d : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2134a8a846a24941a04fd0e3a0643216 : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bf2b618a7b1f40949bda296ee4fa4a86 : SUCCESS in 9m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/35e2d2b578894c4a8581f4edbdbe5578 : SUCCESS in 10m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c460d51f74a9448eb9fd61927d016e17 : SUCCESS in 5m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/11ac57b2323c4af49f0caa17b01ac402 : FAILURE in 19m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5de6280fbabb4fbd8f9201c249e62f86 : SUCCESS in 5m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e7a2311b9044409ab7727afdf128b4e7 : SUCCESS in 34m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fe90adf6cf2e4762ba1affcf1c368056 : SUCCESS in 12m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8f79c306680447978c0ce32d4524f7a1 : SUCCESS in 7m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7c5ea6054f547a6b9e2375e5ce3fdc9 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21f0ec4fa5df4a93a8f7433bd741068f : SUCCESS in 24m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d6cc71f184494a50b71caec0f3bdde8c : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bc9d0801e6164d589564893ee781cee3 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cf5c3a2a50c4e5ba4df08828e8c28ba : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3e462a6ffa20416b84b08cd1e5525491 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3216f9491042441a8afb121dbc22e3d7 : SUCCESS in 1h 02m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c05c88a6252743f89bfe29e63036fe6d : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b27ca7323f93ae150ab08a0e4e0c2822c5e36b49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 00:36:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e47862a38e037e41ea6941ab901bd25fbbebd44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:44:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec9a2015ca94117acb9d4f40fd990d2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13b884654c96412bae2a3d139e81c13d : FAILURE in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1d918ff30a1ab0f6bcfdf73a41afb05bd40e01d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 02:02:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec64713f147941b39853dd31e3f9c1a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec39fb5aab304297a0cc5b8e8eedbd41 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c32deb0f0e4d43b91edb4759598871 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d89f380a3cf44f08dbdab9a0e32cf65 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f221d0aaeefe415db791bbad7080c315 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48299a42bf304ce3b43508d37e6aa6ba : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ff5fab4c4c7458a8221ebd8ee643ac9 : SUCCESS in 5m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/484d246a83c846e2b74c3377f5757f88 : SUCCESS in 8m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/224e6053f2174b6d8bc682ad0836866b : SUCCESS in 12m 40s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9f7ba58bf69047c0a21b0b76425275e9 : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/290631918d8a4f569c1eadf275ded2db : SUCCESS in 5m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0b5594057d8641f3adf2ccd516915712 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c6226ce5438d4c95a8292bacc269e5a8 : SUCCESS in 10m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4499c4fc5f394c97868f647f1419e3fe : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a496fc00f96b40fba8155147c77404ae : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f20de3faeb8b481cba257c1e188de0dd : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/33dd8c4958764a96bddd7c3f64aae38a : SUCCESS in 7m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/492a6ca39b694ad0b1f7d2c7d9268c66 : SUCCESS in 8m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/da317a495d2349f28b7ba0600188a9aa : SUCCESS in 8m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0c9bf892935d48cd9f78f3ac0cecb87b : SUCCESS in 13m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/26f9fd1a8112496da044c64c93fb23a0 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3d00187b1ff14b40a0a02e627a97bd43 : SUCCESS in 8m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/939d07fda8214c07ac979eeff211a6b1 : SUCCESS in 4m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/78e12825974747128c4f36fc13fd1b3d : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/07321738391f44cb9397e008c07dd246 : SUCCESS in 25m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b9c091ef94394696bd5cafe83915fcf3 : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a2aefc46e87f47e59a2c5247204d64b4 : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f68cf10eeb2b4a0487dccb6b092cd89b : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d64f401382124c58bd3a5ec8c0b82255 : SUCCESS in 25m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dd127e44c2224603baacd9991a492205 : SUCCESS in 6m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3705eb8e3f6241398ba4b532b85f12f7 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b0b8de33c8e483ba4fc3fb8df7632e6 : SUCCESS in 5m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8f70e754c1694af8891563e1e17cdaab : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09aad06ddafb4445b33703505b2d959f : SUCCESS in 53m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bc876cf06d24c09b19ce00598d82071 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d696232fc8add03c53ab3259e24ada3e049c9e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 02:04:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31ca2aa6b95654ef96514a5875b37258f6ae71ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:32:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9902d2c1807d49fc945f4fb3ca6a93fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80471d36b634cf88c431646f419f1dd : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3fe97415c7b42eea007e0b722f20b18 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfabcdd422dd4ef5898d4f5569321823 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71e43e2b66d9489893b0b70b9a681f99 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af7859eb6083411898b274f7fedfeb55 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/067b7cbb81544f6abadc44599e972125 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/26a80e8ce7a645eea1e7b4acd8dd25cc : SUCCESS in 8m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f649832291534fe2b42504e1662578fa : SUCCESS in 13m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/718b54a26c6b4e2b8146e3e7749c321c : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/caf775ce957b47d398155540b180a51d : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/377926abe6644ab8a8ce10a82afade29 : SUCCESS in 5m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/14277aca33484379a0720e881c3c1182 : SUCCESS in 10m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/152f44f969774498bbce925f384ae55f : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d3f6bb8f2bd469897241a1e1cf9f6b2 : SUCCESS in 9m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/75a93de0732245d8a498bb42763e12d1 : SUCCESS in 10m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a97482337230454f841282bc1ca4356e : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4fb6f3f9787548fdaf9b5c4b271c0b8a : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e4d9aa62322e4e90b063b26b9e182058 : SUCCESS in 12m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/733fdcaf9a134b68bec87d55d2525f02 : SUCCESS in 5m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/252de4deee234b859ec3b4604c5fd225 : SUCCESS in 22m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d9d0562b1de84e138df905eb5cf1f952 : SUCCESS in 11m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ee20ce12830d455c8e26f72bdb0a4cbf : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de295ee8676a4060907c3a1474a22443 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b858878869444c98925486a6c0845e5 : SUCCESS in 29m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d0a68f1539e34efd80440f18d36e7dbb : SUCCESS in 5m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/31e87d9fd8464d50b4959096cd2035a7 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb43f644e39949edb3dc8db96ccf89b0 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e06f9b27bd34e659da972b6015410b2 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68cf335ad4c9470ea9a4b545bd35aaa6 : SUCCESS in 54m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de6dc81bf2ab49cdb7b7709fbe232b28 : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e3727a3220ebb21f35dab98cbdbec68d021fe71b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:32:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"902e221c7d8d1cac585dd09ec1b15d4e850dad76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:33:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77172100fde34feda22462379a0f3c62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f94a5abc49f4960b6ecdea3cbb7262b : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff4f83783926aec2965787c9125ceaba78b8c99f","revisions":{"ff4f83783926aec2965787c9125ceaba78b8c99f":{"kind":"REWORK","_number":1,"created":"2023-02-17 10:28:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/874222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/874222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/874222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/874222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/874222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/874222/1"}}},"commit":{"parents":[{"commit":"260beef025c92cad2fe17201fe700d35f4b47692","subject":"Merge \"update constraint for os-win to new release 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/260beef025c92cad2fe17201fe700d35f4b47692"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 10:28:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 10:28:27.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 9.0.0","message":"update constraint for python-neutronclient to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Icc639a5bcc9eaf32c0c074b05b611568e5153791\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ibe3fe3fa81dcc459fd68abf2ad1bbd6fa50cb1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff4f83783926aec2965787c9125ceaba78b8c99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff4f83783926aec2965787c9125ceaba78b8c99f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
