)]}'
{"id":"openstack%2Frequirements~874224","triplet_id":"openstack%2Frequirements~master~I5e33e1b9bcfbb05bbee9e1b6b5bd02fcdbbbca8d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e33e1b9bcfbb05bbee9e1b6b5bd02fcdbbbca8d","subject":"update constraint for python-magnumclient to new release 4.1.0","status":"MERGED","created":"2023-02-17 10:29:33.000000000","updated":"2023-02-17 16:29:51.000000000","submitted":"2023-02-17 16:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874224-new-release","meta_rev_id":"ac1e44075e816c7e5967bb935e2775d7a9de4488","_number":874224,"virtual_id_number":874224,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 16:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 14:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-17 10:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 14:14:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 10:35:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-17 10:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 12:01:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 14:14:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bb48671d81fcaf8359b883a1130d017013ebbd63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 10:29:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d056f7f9a15e829c57153233b7baa06e96bb78","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 10:35:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ea58bb875e1132d7b091afc555f882bc106c1c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:45:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8276914dde24263bcac3a4897cb890f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a590cc31f9de48c1b4964c2eeb45c412 : FAILURE in 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"432e6c5b1892a258977953d0898f852f72d0dc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:01:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb95b70a54a6457cb58dde5362116f49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c3959ea4c240e4ad759734e943e215 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d174932a5c2422da59b027d5b7cbd76 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53268165c31488c92aca6e33665f01c : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e3040d310f94da9b365d251e8822c69 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bdbcb3059374f688a6525d003ae6e4d : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3439857c4914322971a253cd6bbb539 : SUCCESS in 4m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8cc8de6f4f9c4c2686512b02605d4f48 : SUCCESS in 7m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e1a61d5e724b4ea6a96e157feb5ff7f0 : SUCCESS in 12m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3a03c28881714804988d1cf909670dbf : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ea4368e8999b416eb8ccc69cebbae7df : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/eef110ad47fe465d8979368f8f2eb367 : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e9ca7d5f9c024998a40d47d4c9846455 : SUCCESS in 16m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/61dace65b4bb492f9fa195d6a00bd7d3 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/08c290b4b6a14d66a6d234a4bd9d3ff2 : SUCCESS in 5m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/80ac1b93082a4848a332e42f4c3f616c : SUCCESS in 7m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/da50ca13f64e448d999303359eaccfd8 : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3613e63a58be417891c6a8b4b49ed761 : SUCCESS in 35m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b988c0a9057048ccad89e31bf3b9757b : SUCCESS in 8m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/af5293f6404b4356bd784cf66b8a7525 : SUCCESS in 10m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/948a5467db5a421c9de15bd9e76a0373 : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/766f73c737174fab81d7eb0b94bf6769 : SUCCESS in 9m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e24aa335f14f459e85e6eecbcf1fb3a4 : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b3f46bbe83bd4f069740d424db0044cb : SUCCESS in 10m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6583cef3337b4a3091cae3f898235148 : SUCCESS in 22m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/02b111b5344d4ebb83f372df5369e4b1 : SUCCESS in 10m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6c931d19648849a59f33352a4869fd92 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8adcd52e966480298a78570ac55805a : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de3b581dc81d4acaa6b40530571eff03 : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5f8c134e814f4459a63452b7d1195a43 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/10b9918b257a4890a93c360065d906f8 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ad0ab1ea8a14e59a394aed681921348 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8552c9f6ce85453f8d250265165d9a04 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e104169979dc41df9f7c1caf9df7d5e6 : SUCCESS in 53m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38bfd95bfa6640649e06883fbd68480a : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f99ddc21dd404ceb5d1c620dd4692c9a3c98c8ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 14:14:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddda5cfde7bea2c43917f7b950cbb5555f2a6f7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e808724ad035d0c60f55c8d99c12db1c77f7bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:29:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ffe71fc618441cc8c47462bcc5e928c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b1bf64440d4928997a896c9ad2d4d3 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9340acc6d4241b5bf171d36a2238082 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eff844be1613448c807c6b99499d25e4 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c1137553044217a88ddce7f4e866d3 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85b2222a7d24407098eca5fbe479b631 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/726fb26afaaa4c5d88eb42129d8e4df0 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5354339fbc744e3ea08691a40582a6bd : SUCCESS in 7m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5665c7c661d4728a823c1c3fa6e2920 : SUCCESS in 12m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3bbec8c679214c72bd3620090df1ca41 : SUCCESS in 18m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2860703cde7c4347995c0bf1aa66e082 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d979a034ad714b358ada52f2e17a1fbb : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2183124386484c2fa6803b2ff90ecc66 : SUCCESS in 8m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5f8afeb8fe7842c0a14f4a6f30f739d8 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e38bed962b8e4873a90054147d54619e : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0af6162649e24fdb90b40ebaa0f8686e : SUCCESS in 9m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/292b15ea7f2243c79b7e411642603646 : SUCCESS in 10m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8cdcaa1a822140b195281fd26ccfd4ac : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c851c04f383649c2b30a22f5e0cf44fd : SUCCESS in 9m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6d4ee8685ed14981a4b611ddfca3f078 : SUCCESS in 4m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f6903b0140464b67953fc8f84b033989 : SUCCESS in 22m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/75b7810f58d647d799c28780c711daa2 : SUCCESS in 13m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5eb1412a8df849ab88f99e5023b6104a : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/250284e9329b45ac83731b4295614a42 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07fc3d7fb0de471baff892a87faf3717 : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8f6b02c83665446483c10b61e054b402 : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c4919fb43051490e93e57631481f0f6d : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/373c24c86c9e424e94098add6f52cda3 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/214992e46903454fa86eea6c97e0b52c : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65bee1fddf0a4cb7862b90eaa7df935d : SUCCESS in 1h 04m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc4ec82e1b544e3eaa375ddc4effa625 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ac1e44075e816c7e5967bb935e2775d7a9de4488","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:29:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1f553835171b7088f4e692103eddb245861097f","revisions":{"d1f553835171b7088f4e692103eddb245861097f":{"kind":"REWORK","_number":1,"created":"2023-02-17 10:29:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/874224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/874224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/874224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/874224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/874224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/874224/1"}}},"commit":{"parents":[{"commit":"260beef025c92cad2fe17201fe700d35f4b47692","subject":"Merge \"update constraint for os-win to new release 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/260beef025c92cad2fe17201fe700d35f4b47692"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 10:29:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 10:29:32.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.1.0","message":"update constraint for python-magnumclient to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Jake Yip \u003cjake.yip@ardc.edu.au\u003e\nmeta: release:Change-Id: If3cc417a3f3b8d70e995913c61d5b71872a88f71\nmeta: release:Code-Review+1: Jake Yip \u003cjake.yip@unimelb.edu.au\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5e33e1b9bcfbb05bbee9e1b6b5bd02fcdbbbca8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1f553835171b7088f4e692103eddb245861097f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1f553835171b7088f4e692103eddb245861097f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
