)]}'
{"id":"openstack%2Frequirements~874236","triplet_id":"openstack%2Frequirements~master~I3e6fc61419b8c608c43ef06fd6860a14eec99dbc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3e6fc61419b8c608c43ef06fd6860a14eec99dbc","subject":"update constraint for python-designateclient to new release 5.2.0","status":"MERGED","created":"2023-02-17 12:28:58.000000000","updated":"2023-02-17 15:51:57.000000000","submitted":"2023-02-17 15:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874236-new-release","meta_rev_id":"d4db6243fda4eef05a407dec2611cdb41f85dbda","_number":874236,"virtual_id_number":874236,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 15:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 14:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 14:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 12:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 13:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 14:14:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5154c5e621928af42bc74ce5b1b05a5674df38f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 12:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dcc9482b3ffa0608ddd25bb7d8fd36ed431b695","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:35:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ca66051d9c46e694e8765a89e6f560\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ba749229896a4bce9ce15f42b4331c52 : FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7d8f72fe1814f52329908b376cc8ebeb7564929d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2fd0d6e2a04e87ac1e623bee302c53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7a1d32047d442ebec07ba557bcdadb : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7c80a90a0345a9abb4910b15efab63 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d611673234d94d5498f989e26c990657 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4ce7b3872df4d0aac45ee929ac23ecf : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/faa004b8897744709c292e51fe4ed46b : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af5b43633b844dd39dabab7e22b14ab1 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/651dd91f8f294e5fb281810bbc4217da : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d64406f8449748e5b78b4cae70c0f01f : SUCCESS in 15m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5cf780597a284449a8e4e270c1bcc689 : SUCCESS in 7m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/aa3d09519d014194afed538e02be9ad1 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bfebe68f0936475e88d8ac0c85e32b90 : SUCCESS in 6m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b032bb096d6043ce9df004b70eee2d3c : SUCCESS in 10m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6b3c8cc61d9b4580a3394a490b598c0e : SUCCESS in 4m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/92577f4722534cb4bb9e81da3829bf04 : SUCCESS in 6m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8df4925778fa4e4aa224d613ec3f365f : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9eaf70bd79e846aabf81a77c97d86de6 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/951a0d1e763c47fca9f309a2541265b6 : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f2bc159692844f05bb399ac5126d70e8 : SUCCESS in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5df189333ddc422899510662831174a2 : SUCCESS in 10m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f0a214ca652a4b48ae3475c0d8217233 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2adbe3e125b14869a39ba00b765216ac : SUCCESS in 12m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6c798afb018e4340b1e85d400e9e8840 : SUCCESS in 4m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5b217ace3f984ae7b7f3b02e0ea4e0f6 : SUCCESS in 9m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/df1119623c65473a9c7f7388e6dfbc03 : SUCCESS in 31m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7e9a04482d2f40f49d7f4557d543b375 : SUCCESS in 14m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e40bce76863e4a7fb8c771a4f690fda4 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44d95b577c634b50a82cceff05841aed : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd88600e0fd747119f5ed6d1bb3c45c2 : SUCCESS in 24m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/76782a06c10c4eed922f11529b6fc28d : SUCCESS in 6m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a446daaa693c4b9a94a87d5b54a29730 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7aed38d796f94fbebffa8192fbee0f41 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0afa36fdc594451f88ff2117630189d7 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1226de9b7a664b99bbc5f43d5f4ebf27 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6eed2e06ec9c4174a8c368d9dea7bdba : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"98cb1bd563300a84126849d15f1635039953499d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 14:14:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d42c77fa5989202ef5b47182897e15c16b5d5b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:14:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cbeda0468ac573640c73037c5282aceda2442f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:50:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a6487584464388b98d1adcb7ae87ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b6b8a48fc940b0a8fb7b43ea060c6f : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab65a47dd9a4d00a42609dd9e535942 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f7fc6cc83ae493b8dc2dcd8e973b065 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2d15effbe4743828423a56cfee891e6 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4ea76fde1ad47e68d99b4cf5a96c3c2 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1430884a796e485db326d3352bfafe3e : SUCCESS in 4m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b37d3399c3fa41159ce2b01bb5ddfab3 : SUCCESS in 8m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e73747d7f732419c89e3e8d1c83a91e8 : SUCCESS in 11m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/44cf0fbc9b5e4d4abdcc8554266fcb4d : SUCCESS in 12m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a723d2b80f874a168130a415610d74b1 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/529d3318ea0d440ba5d26ba0596705bb : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aeb3589501c540bdaa30b5bccc213fb1 : SUCCESS in 7m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/71e639f490f946cf9a29c742b157804f : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/096c2b09e9884afba8f516d4c73e3337 : SUCCESS in 11m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/642632ed527f480dabada277fcf1054a : SUCCESS in 8m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3ed8540a020a465ca7f97b4662715b5f : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1ed4aec23c7a42e9b6728d4d53f0e12b : SUCCESS in 4m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4fd9ec80149a420f9065ce7929d48b5c : SUCCESS in 9m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8efb001948d848fc8f04c7a04000052b : SUCCESS in 4m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a63763ba75ee4097816269e0d4a06bee : SUCCESS in 26m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1d442c8d21a642c6aeee31fb06d0c376 : SUCCESS in 12m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7abda7ccb95144358b29c2c9446dec9a : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a7acbf5ecfd4254b70f0441b797e292 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99aea22744f844b48df8375119a55889 : SUCCESS in 23m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2dc67685a899459d8d7ac6621f451aeb : SUCCESS in 7m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ce4b71a378954a43bf1be0bdc4c53f37 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/927e3f29f10d42eaa28f2cd3225e024a : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/366ffea9d8a042c7a4bc5f05a1fb3f51 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/431ef4c75b764cceba67c75cd514a0a0 : SUCCESS in 56m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bdec766ca6d45d494b4ee5c857ab25a : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"93a6d132f073ed6559297ad354dfcb9d71bbc655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4db6243fda4eef05a407dec2611cdb41f85dbda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:51:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae14b4a4ed5440fa07c40da639c76c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9914b483d876498d8a308adb2f2a475f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b454cfa85b11974156c8354ac91c2fe9f88c779f","revisions":{"b454cfa85b11974156c8354ac91c2fe9f88c779f":{"kind":"REWORK","_number":1,"created":"2023-02-17 12:28:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/874236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/874236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/874236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/874236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/874236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/874236/1"}}},"commit":{"parents":[{"commit":"6fb0d2954a1610a1be6237d4192d44b5820e3e7f","subject":"Merge \"update constraint for ironic-lib to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb0d2954a1610a1be6237d4192d44b5820e3e7f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 12:28:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 12:28:57.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 5.2.0","message":"update constraint for python-designateclient to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: Ibe848c4a2704f509629e404cd8bc65dc12260e22\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3e6fc61419b8c608c43ef06fd6860a14eec99dbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b454cfa85b11974156c8354ac91c2fe9f88c779f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b454cfa85b11974156c8354ac91c2fe9f88c779f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
