)]}'
{"id":"openstack%2Frequirements~874243","triplet_id":"openstack%2Frequirements~master~I56daa8ad481c41994679933a8c7d89a61132fae4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I56daa8ad481c41994679933a8c7d89a61132fae4","subject":"update constraint for python-manilaclient to new release 4.3.0","status":"MERGED","created":"2023-02-17 12:58:03.000000000","updated":"2023-02-18 04:54:42.000000000","submitted":"2023-02-18 04:53:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874243-new-release","meta_rev_id":"d0fe9be44c8b5e2167fa27df8ffba3061d093d6e","_number":874243,"virtual_id_number":874243,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-18 04:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-18 03:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-18 03:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 13:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 14:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-18 00:36:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-18 03:15:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3253652aa5048801094c1a5957ef13aab33101c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 12:58:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22808ecb22774187c250c3067ec59514e353fe03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:04:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b12a55eed14ee9a41dc08566c1faf7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b4c839726e294bb9874c7ec21120d761 : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"35e0084d98a1f0a7863373d8b07ffd4f89791ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c7b02aff53c463ca12be7a09f1b314d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f0a532182b4285b2a3abcbba2745a7 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752fa024d0a14ea8aeb731192d9e128e : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d09df32400f6420bb7aaa0907da26d6b : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55b0891cb4eb4293bc8c5b708de09dd5 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d42af2092b7442d0af681c43c8152fa3 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07274124622141bc82161845554917c7 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f093a46b4168405cb26b04cf6b338e68 : SUCCESS in 12m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/55159aa11479482d87c7c18407cc422c : SUCCESS in 10m 01s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1e7d53dce4244910b9b776c3241b62a8 : SUCCESS in 9m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c0a907b55e8a4a54a26224463096ae93 : SUCCESS in 6m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1f3ff62bcc674fd9abfef303f5670c07 : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/263698da9b5245f9a2d137400ada4cb6 : SUCCESS in 11m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dfdd12ab12214e239f2df86065ddc96c : SUCCESS in 5m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2de5967444e448b3b4d7f0d18fb63799 : SUCCESS in 7m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4ef6263789a24d579953541a91756ce9 : SUCCESS in 8m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f278e8b6a9184f75820835711db975b2 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5d334984f204118be2958daad9464bf : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/06398bd6698543b2858eee11393b0ac2 : SUCCESS in 9m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/79d5bf6423874bb58fe3ebb4cdb06782 : SUCCESS in 13m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/62e7f4aa2bdb4cfcb24ee51938154cf1 : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/45c232935d5f473b8ccb2f60d95c06e6 : FAILURE in 17m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/254ec6984e6e4288baf07b774864b115 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1200e82f6a8e447098bbd4d9b85436e1 : SUCCESS in 11m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/791ef72285124209a3d32b8de265f7c2 : SUCCESS in 23m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b754789c53f7481287d5dd9c8080123a : SUCCESS in 13m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7536695f36684e8492cd5e9912c5a8e9 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/348ce2d2b8b6429eb78e82926d50883f : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00cef602a7d24b519aaf3e6ee0aab9c7 : SUCCESS in 22m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5da1bda142f7420e955298f07ce8f5bd : SUCCESS in 6m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/efaea2dca8ae4ef7be66d70e20208dab : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/772cdca3a732435bb49b7bb62b0ad48b : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/398d820969964a2b9f28e71fbdfc673d : SUCCESS in 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a06ff36bf7e496ea9fbf2fd0798bcec : SUCCESS in 59m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e4310430cdd48a6a9464a1f71f8c14a : SUCCESS in 1h 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e2ff4b1b6822aba4f399391cefadc9c90f5d0def","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 00:36:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dff0df07ae60aa1b357110a0eb42ec9c098ef57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:44:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3a88870505471684cf033d514f0a37\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/14967cd687d34293b829ed882ff0ff31 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4b5864d157ce3b1cfe673f5e08f3cf521fb2e4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 01:55:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24141da6f76d4cca91620729bc0da996\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c10b80b9efd497eab90cdc266158f7a : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cd56a43061423d965c05b0ef097ecd : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86e6f3fc0c7344bb9615f679bdf13a1f : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ceb51cc2f304c1f92e328b4a7a58d72 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abaca79cbd3147259667c7028cf1b019 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5301a088180042d7b67d8c81edadcdb3 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39902c837ad14ac6a1e7e4711525c063 : SUCCESS in 8m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/584e036e9ac543acba1f1d5aacd8b46f : SUCCESS in 11m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5aedf56165474c45a0fbad8082807af3 : SUCCESS in 10m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c341c751745948c0876ff238257d706a : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f60d4b261c7244888204e501de4bc228 : SUCCESS in 6m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/39a82c595615419fb53465514a61b481 : SUCCESS in 10m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f6db9f4e19dc40e38aa80d7e60c46470 : SUCCESS in 4m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3e4b8dc852314e798dcd5443253d4cc0 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7043c002af714fef86705238f3e3a841 : SUCCESS in 9m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/68654a15cc094db3bad4a848c3978521 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a7c36d8f7c54188ac0609a0f9b104dd : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3dd4fa9b74a647f1b138ab59190927ab : SUCCESS in 10m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b8888ac5ce2e45908a72cbf1d561c1b7 : SUCCESS in 11m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/68b9a9939f6f4f4fa76608d0d9378246 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/31c618d940f14c6c9f814eb4c05b1fe9 : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7c40aa0edfd94955b7cbdb164c543024 : SUCCESS in 4m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/293bf385bad94b9fbf622403733eefeb : SUCCESS in 10m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d0f7ce658a8e41309be1fc262e46c0fe : SUCCESS in 24m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/36ce23cde5a540e58ff08700ebca77ca : SUCCESS in 13m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/707fd7e4298e457f9704a96cf419ddaa : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7fc7d7f0ed94bfcb73968720d28a71e : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c542707de9e416aa315604228e70a03 : SUCCESS in 23m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/12d8fc5df8f24acf8a2e345efa4c6641 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aefe853f785940cf8cce36dfb0a67c04 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e51d54edc334d998a64007e1c6c5206 : SUCCESS in 5m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6ae39301383b45a2a4672c9f87bbb07c : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/930992d5cf7746cda18b273fe70f1ad2 : SUCCESS in 1h 02m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a09f59f4b5948b0a843b86727aa06c7 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9b0de8aefbb0f028a3faf173d09b51541897e16f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 03:15:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b619c6893f720a44fd95f922a5ab47f918f0ec8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 03:16:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be653b1288e6cfa694b77a098952261f8dcf7994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:53:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d3fd6f5c832424fa4b31ee275a80c8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44dfab4fd4384556a77a1da015a9042e : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9253142f9731417a9001c7d8648c0012 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3c8738fbc4c4796b85c6abfd654bb18 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60e08ff9b9904b3cb6ff416a00de474c : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0948eed510174588acfae4bfdb9a4f95 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b426f9ad7974fb88e8949c8662d5455 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9aab67f9dd9f4cffa7fdad9e6331d289 : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4dbeb156de234f9d9195f1789f79e423 : SUCCESS in 13m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2b0c164f50454a50812638384944ada5 : SUCCESS in 12m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c2a9a5867ff343eba8a8a66345e29056 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a2b09631ac3f4416821242237a5f8155 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c392e0e18ad143248c32ebadbfaee1a7 : SUCCESS in 7m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e4a7b39df6d74364a29f3a95db1e1d4f : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c4efd3975cd49e487cf49985b4f0ad9 : SUCCESS in 8m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f88bcda13a59453babc088e509270f62 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6a2f72fa36dd4d3e8764ccd344f983b7 : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ca295a67eee24202adf8319a36b4572a : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/247f3003f6b0484a9fef19602e1758fc : SUCCESS in 10m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0718078f815049ec96551df57867d076 : SUCCESS in 4m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/31f2053035d1452ba08f33f82395de70 : SUCCESS in 27m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f1b34ac87480410ea47b36e1fa2ccfdb : SUCCESS in 17m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2043ae47d59741e4ba12f7099adeee24 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b89b3e14f5149a5b7bce93ff982f822 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab26388d10e84deb8d7cabf338e49299 : SUCCESS in 24m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/454a318cc89947858a69a53a81dae986 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2b892cd8222f43f59118006722cd0c1b : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d563ce6b184462988a282c3ed26610d : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c9a5307854a84366b36b8d7fd0dd2e81 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7e98bd1d8054ac5929feb5c15c18ba0 : SUCCESS in 52m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cf13fb7197c4cae91dcc8ead80b3259 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fed72742d502ab5792d9f5a63ecf194fa047c129","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:53:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d0fe9be44c8b5e2167fa27df8ffba3061d093d6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:54:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eddac9e16a8461886766aabd6a5d262\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0d013c570f4f78a21f7bb60a3468c5 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c5e703fa80af8b3761326925bdfeb864e2b000c","revisions":{"4c5e703fa80af8b3761326925bdfeb864e2b000c":{"kind":"REWORK","_number":1,"created":"2023-02-17 12:58:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/874243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/874243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/874243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/874243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/874243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/874243/1"}}},"commit":{"parents":[{"commit":"6fb0d2954a1610a1be6237d4192d44b5820e3e7f","subject":"Merge \"update constraint for ironic-lib to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb0d2954a1610a1be6237d4192d44b5820e3e7f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 12:58:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 12:58:02.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.3.0","message":"update constraint for python-manilaclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I3f0857e681efb816ac871ede4e1b2608e629cad0\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I56daa8ad481c41994679933a8c7d89a61132fae4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c5e703fa80af8b3761326925bdfeb864e2b000c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c5e703fa80af8b3761326925bdfeb864e2b000c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
