)]}'
{"id":"openstack%2Frequirements~874255","triplet_id":"openstack%2Frequirements~master~Ibb0591c33ed59da20728455f8fd72c2e509b16b8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibb0591c33ed59da20728455f8fd72c2e509b16b8","subject":"update constraint for python-zaqarclient to new release 2.5.1","status":"MERGED","created":"2023-02-17 14:49:22.000000000","updated":"2023-02-18 04:55:29.000000000","submitted":"2023-02-18 04:53:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874255-new-release","meta_rev_id":"403b948df58f52a9657f4e65500ef2842bf0c118","_number":874255,"virtual_id_number":874255,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-18 04:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-18 03:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-18 03:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 14:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 16:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-18 00:36:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-18 03:15:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2fb13fc0bb62b4cacc79435286bf92e88496c13e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 14:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"058b94dd11137d9f06f6c06ad7ed4529819c0e3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:55:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2247592d274afbaa20ceb31a8c9e45\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74704f5909c946e18e2ccaf8d6f3288f : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0a3d0f77af913f836ba6dab02694d72e8b0cdc96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:23:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ec2ac5ac2364cf6b32bf8ae4873eeee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40713f295f0412a9feb798b71470de7 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4110cf9946dd446cae7c4b90cc934049 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fcde41b07784bd7ba29f9e0e622f540 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a43422e90cf45a09b6778024daa7104 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e63cc424607942caad73df7839cf3465 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01ad6be2287c4d00a6fc706c86034584 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ede7e307016c433582c1e6408722ccb9 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/855992eeb1c04eceae54159e362d99ce : SUCCESS in 13m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9ac6ce06faa341f58eac46b9c2a91270 : SUCCESS in 7m 22s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/05b9511223ac4cf0b8079369ef800eec : SUCCESS in 5m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/898b034e371942c6b900e113c39a96ec : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/66231a0f296a470f9579467e1e759a53 : SUCCESS in 12m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/985767755fcd487baf8882df6366485d : SUCCESS in 5m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b9b71e9b2adc4bd68964acfdb5e04c59 : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eaad458368094ae9b546e2bf8bfad5bd : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/86ddd7428b0740f28fcbd8cff6300339 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2804d9f28a14efba5698b09bf19f6b4 : SUCCESS in 8m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c4f617b14fd2452bb0c5da002c984cfd : SUCCESS in 9m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6f08bb7f7b984a77ab90e57d0d594f74 : SUCCESS in 22m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ff0d03fda8f245549388ad1acab200ec : SUCCESS in 4m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/061a2905e8e0478e9b7832aba41bc5fb : FAILURE in 17m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1053413660f34d6db902a6a2d140407d : SUCCESS in 6m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a8173f8b818d4648b724124324a57cdb : SUCCESS in 9m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5ccab3de89054696a5e6aabcc968567d : SUCCESS in 28m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fa7a9a06d8f84c74a5b0d3188f94e57c : SUCCESS in 16m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/821131cab54244acaeee8f247dfcd96e : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2a8bfaa0f594ac78ac133227621e7f2 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24bd18cd84d8418aa9e8d3cd1010ab55 : SUCCESS in 21m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9939695cc96043a9ab52efc95a5ad80d : SUCCESS in 7m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/297e3bce1cc04831b30d38a56e6f17d6 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92e3eeadeffe479798ca7b89efec1b92 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4845df1793724bb9b8e3a98f9e95da27 : SUCCESS in 18m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e03cd8976e24a7099e71d0f0075afc3 : SUCCESS in 55m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2d9de51415148fdb74a2095654d194b : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d7de111322a2c86b7a490e8adc8e5a621a4279dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 00:36:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf78da5358f9aa53d68abc223d2bea0f6ce07bfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:43:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0acbcbad01d43fcadae36bd1c127f14\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/82d710f5aad447a6952798bdff0e04cd : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2a68cf240c507697e0c3fb10b4688ff80767b54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 02:18:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4472a96a0154b71960984ba5adc2f8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7127d5d417ed403cba9294b75ce495ce : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f580fbeeb54bda95d0599fe9ffb3fd : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a2ca5737e204400b771d158a46084ae : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fefbbc043469452cb42aed4cc8ad3019 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0bf8d4491d544cb8ed57b9bfa5a78dd : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ac51bff252941079b8af07824758f89 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/103561351c3e4b8c897b9d9f9cb3dae2 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3e41bc5b45ee40e6874f1b95aabe0333 : SUCCESS in 12m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2b4f3030225a4052a84dbbfedda324a3 : SUCCESS in 7m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/49c900f999f0403bbfca852d255c0539 : SUCCESS in 4m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/825c495f5e1d4ef4a821b4a97e7c2a56 : SUCCESS in 5m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dfb4ce5a4c7b4efebba8d616272d7d98 : SUCCESS in 10m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6e5eefc3befa41929f6c32233a115e3d : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c752c52667d7435283b58d73be38a006 : SUCCESS in 7m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5a68e196cde34294a41c56998842a146 : SUCCESS in 10m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1ba61fdb459f4e6aa70cceb333de128c : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26b719be7314416dbae20c85ff27c3c7 : SUCCESS in 7m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a2190cf4617c476babcc1841ce91f0d5 : SUCCESS in 7m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d9b0728114ea4cbbb5f8d6d36999dd7c : SUCCESS in 8m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c8e05365f4af4199a2859a97a99cac9c : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fac3f393bd904f12a801ac298dfd7ec4 : SUCCESS in 9m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3ce912b78def466d8b4d564036084655 : SUCCESS in 5m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ef056c8b337843daa2881ad76e7045f5 : SUCCESS in 10m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ddcd21fd706b49ef922c0ee7f4e30217 : SUCCESS in 27m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8ecde67319a44b8cba7a00cd6e1bb39a : SUCCESS in 14m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/81eed8cdb3d6473abb9061b3f0d8a65d : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25471d29ae7c43039741f775b97e6afb : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/271e8d6e67184816be649fdea9fef6ea : SUCCESS in 25m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/34e2091277ac468daea9385a2a13b7f4 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ad74c9222ace449f8e462cd8d418b880 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48dfe208c19c4f27986321d12a1ac22e : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2e3e54555e7b41eba5f81437717acc7d : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e527ccec3990419cb91b111b578793df : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0fe3c6cdb094e2480dae08b8145959f : SUCCESS in 1h 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4651e55749f2afd02ecc36d8602d972bd9c9d4f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 03:15:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2cf2cd8017a98893d7d8ecba3a37cd1ff8b25fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 03:16:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f0177e3fe5142df6cb3d8d73f69ab8d47ce0c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:53:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89f550fbede748febaa9ff10779afd8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a7a544c0794fbc9bfbe6e19cf17514 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c436893f5c42559cc399a4f7be4fe7 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/819d9dcffd82482e99aeb09fba9b95ee : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d53a9ee812148f6817fa602433f484b : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07e8ec6931904ea3b9353239ff56da9f : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1564ded242a24f84933feeed2166c26f : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/beb7af9bce68464f92afe91f958e4cc2 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56e44b65cda54fc084a177957902e332 : SUCCESS in 12m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/791e5834067849ef83e94711935f1c30 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/021f4bd55b974adf96e969803eab2ecc : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3941a64ee0b94d299f5197e59929cfb4 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3682e627614d48b290951e6b7e30b1e1 : SUCCESS in 8m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/de99f2f049814da39915cbc9fc049243 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29f14af80d5a4739bd11f4ca7faec8ac : SUCCESS in 8m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c125f327e08740e7afd04fa85e3399de : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dfe67e8829be4af9bfef9b2f8d4879c1 : SUCCESS in 11m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/945f409004b04267ad60e30ca78e8df8 : SUCCESS in 6m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/faef2d71207a47cfa42cb1c973f3e633 : SUCCESS in 9m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e64850aa0aa947f8a630878f506c48c7 : SUCCESS in 4m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4515430f3c244da5ab006ff26e801cde : SUCCESS in 22m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e409fc816b104de0923a5ea5481e7211 : SUCCESS in 14m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d932fe4905704768a180a892e1bd7891 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5235ad57c7844f7b92c3ec796fd2e4f : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb313fa11e5e4cbeacf68b73d206307a : SUCCESS in 21m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a2e9c9dc40d64c6f9d06179c4e7aa593 : SUCCESS in 6m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dca447854d474798a244019689da0b90 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a04ab5259e34f0cb0b94ce6be3eea66 : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/20ccc868c1c1497f9f2dbb3c6d80c295 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da3e21f12eab4681addd2d04bf23974f : SUCCESS in 48m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8faf964d870c40a7b5697c8087172192 : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"24e00013590171fcbf4c15d48783e372e17c8508","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:53:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"403b948df58f52a9657f4e65500ef2842bf0c118","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 04:55:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d466dddd06e4677bb2864648d0b1949\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07d01e70f9e433096e4ad0eea71b531 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5a23a04a159c55ca7d70c49c249bc1e82fca40b","revisions":{"f5a23a04a159c55ca7d70c49c249bc1e82fca40b":{"kind":"REWORK","_number":1,"created":"2023-02-17 14:49:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/874255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/874255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/874255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/874255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/874255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/874255/1"}}},"commit":{"parents":[{"commit":"6fb0d2954a1610a1be6237d4192d44b5820e3e7f","subject":"Merge \"update constraint for ironic-lib to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb0d2954a1610a1be6237d4192d44b5820e3e7f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 14:49:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 14:49:21.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.5.1","message":"update constraint for python-zaqarclient to new release 2.5.1\n\nmeta: version: 2.5.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4899926fc6ad5a4c6c4d867b0e4572feb922b99f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ibb0591c33ed59da20728455f8fd72c2e509b16b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5a23a04a159c55ca7d70c49c249bc1e82fca40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5a23a04a159c55ca7d70c49c249bc1e82fca40b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
