)]}'
{"id":"openstack%2Frequirements~874265","triplet_id":"openstack%2Frequirements~master~I378e9014433efcd28730cdbb46e4830b5f617a08","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I378e9014433efcd28730cdbb46e4830b5f617a08","subject":"update constraint for python-troveclient to new release 8.1.0","status":"MERGED","created":"2023-02-17 15:30:42.000000000","updated":"2023-02-18 18:08:40.000000000","submitted":"2023-02-18 18:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874265-new-release","meta_rev_id":"fdd58d4d200c1e666bf9fe4b1054a3227786e143","_number":874265,"virtual_id_number":874265,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-18 18:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-18 15:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-18 15:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 15:36:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 17:12:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-18 00:36:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-18 15:54:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fb59762d5b1641f9f09ca76d119ce2a92066c6ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 15:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a63039db33de5bc64938e23d403b91024e7c8b68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:36:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220cf393cf0548529dfeefa826c1a458\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cea8bad4febf4b2c9ef97e31da89d035 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f3c5b0c59e7110b92d17c58841105f0e02d84087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 17:12:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d145c5eb394f2eb01f626861d68fe2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce222757a0648c1a16a6012cbed1fbd : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa82011aa4d34863a4514331959343b6 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdae9e6ff8c849198f279f118f74131e : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/392edffc05cc4e3fa73c4cb7410f42d8 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51c00db450234de3ba022b928c400532 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/524139c3817d476b996c445c45797a33 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24f3e039d0364c8cb8123d9fc7820b9d : SUCCESS in 8m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e2a32077d54442f79dc8be2dbd5f9fad : SUCCESS in 11m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/08b9e148c39f404eaa330655ff621904 : SUCCESS in 6m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5af5f93f299b48fd85ecfc158b442a74 : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d9f98be4c3514022973ef50a636d9ddf : SUCCESS in 7m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cb43c03ff3ec4e779eb832677ed347e3 : SUCCESS in 12m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/837f2a78e39344b2bf6c083595175439 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d133762534ed46b2a87b48501b596381 : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1bbbd0da1eb342b2af6e6f27ccc11b84 : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8472e3defdce4665ada913f5f5f66f93 : SUCCESS in 15m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a01a979830d544ebaebc838aae2f40f1 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5fa5aa0fb5974d8095379c7521380cc0 : FAILURE in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d8e1c15109044d0ea9b00b2d453587fa : SUCCESS in 11m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cc6161e49a3d473cbb2ef2614f25eecf : SUCCESS in 5m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b0730d8570474f28b73ed8a4d386a820 : SUCCESS in 9m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bb31bc37475d44e3b400beda5261d4af : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/730821ab89bd4b0faf92d46333ff9430 : SUCCESS in 16m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6125876893de4f208f5924cb3bc4d253 : SUCCESS in 26m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e754a37795464d3085f2db455f9273bc : SUCCESS in 13m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/002136d239614790a141b83a1cbab574 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a51bd2d401d44054952aa26508dbe16c : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73266c74cccf4913ae6de07e6e0c25d3 : SUCCESS in 21m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aaf31c6f37394982a7a4d3c4534c7b0b : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/15be09ecbae245589ac5ebec2c4fd878 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d54ea61e6d5f4ac29ea5405876740ef7 : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d6066b442365456ba51676899efedbfa : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ed6052fb1504a8e96d7945a8e804f5d : SUCCESS in 1h 03m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c3be81108c8462eb363eb6c982cf54c : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c59694740d7525ad30baf44dfb5955d6fe809eee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 00:36:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a3432eee31abcf1ff12b4d6a86ef4653f6815af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:44:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14b1b79d04f48aab1f829d382cdee29\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5f9d3e3726694b4cbf6686d3e0e355db : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"282bcfd987044a389c4eaa54543e26f7cb967f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 02:27:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/121dd878ddb1438e8abe4b663b6d5488\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7710cd2a864cb085510682c46a6004 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0f715a91804b57a0e7d7687d3a4277 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58ddbf577b24429b93a44ba952dd98a5 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf402e1528844c6c99a8246d5440c606 : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/077d812ed03d40ff87920baa53204350 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9f32c5fe1ab4477b3828c62a157eb29 : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00008550753f42b0a2cede365313f74b : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/677cc08963e54f008282187b8d429841 : SUCCESS in 11m 13s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/264f7026d7a44cb9b837a46353535bc0 : SUCCESS in 6m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/83d4662d23a8456d90bc13fa38d4dfa0 : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/22665dc6b7514db991312853c155ec00 : SUCCESS in 6m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/da9bc7e55f9f4deab62c1f8d72c2cb34 : SUCCESS in 11m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f989d5e5323f494db5358190a39b4da5 : SUCCESS in 4m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2cdf9421e33b46b299300e93636bdcd9 : SUCCESS in 6m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/078900aeceda4168a30d9d579e5fb30e : SUCCESS in 8m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b4d0b68e73f041eba8ce650aaf5e6587 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd22b30809a449c097bf7ab16ed632fc : SUCCESS in 10m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d93499fe833e49e9bac2bb488c5f8ca3 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e339e8f9c2c848b0905d4c28d255e3de : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c3547dc88b2d437da0f43ce529f58ebe : SUCCESS in 3m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/998c3f03c8964505a703e3822f062045 : SUCCESS in 14m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2b14a35587f0420ba20e0ff568d14ded : SUCCESS in 5m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cfecde6278114537b256bd6820575834 : SUCCESS in 13m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/db2e3074b5c2437b807266bd3a595144 : SUCCESS in 27m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6512d8a86ec54c958e2d3ee174537d14 : SUCCESS in 13m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1808c1e657e2498f8fe246473fc80040 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93b69ce85853490881770ce79f76c75e : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b897180bce44d53ad974ae7506a3421 : FAILURE in 37m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d63be4b234514af1afe9bc198a8591a9 : SUCCESS in 4m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c52b9c1474f84c269597120bc7914c12 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d092b54a977440f898110dd1f024636 : SUCCESS in 5m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/36366ca14dc343b8a8a3000c48b972f7 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da17c13e4aa2449fa562ff2be0d0a3af : SUCCESS in 1h 16m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1674a1231c8b40d48cf9e473dce00891 : SUCCESS in 1h 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"727dbdf8f2a1a1a516593542c335b30bd8731f30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 15:54:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"402251aec2957e95cbd66cc8fc949956d995cbef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 16:00:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd5facfdd864066b7d938c2ae69634d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af6db63ec2b846f4af41628cbbd9d321 : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fc4810109ec50b2818f172609240b16b74263dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 17:08:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32b7c08139574af0a10c3cd9afa304ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12ecb2d3aef4e0aa7ce361e376f89bd : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fb161fa6e14cbbbee7c446158a408a : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea1c45b98814272a3a382c9c08b0ad7 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40b592cfbbca498b9ada441f47cd82ba : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6312ede67ac444068513c2dec4b7ce30 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d43e2f663ac4422aa99c5852c48c50dd : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67e238c82db1417a9eaec64ad540672c : SUCCESS in 14m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f169887f5d664775b529ba5fdcaee2a2 : SUCCESS in 10m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/264c4fa5223f4b43b2ae456d1d001e6c : SUCCESS in 11m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a873506173364ddbb758e6573c78cffa : SUCCESS in 5m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8fd3dd5ec4104e19ab12c00d10c16e38 : SUCCESS in 5m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0681d0bbf128434490b1fe4aeeba2edf : SUCCESS in 12m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f06269ac75fa4b74bbdcce8424bf6db1 : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/946595198ddd45d5907dfd058f5bc161 : SUCCESS in 4m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2b6d518c76074bf29e411d4ef37e53c9 : SUCCESS in 10m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3874c144df66477f9a2e63cf62f9c410 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44b742401df64dbe82e84c96d02d7840 : SUCCESS in 8m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/aa69c430f68e4547953d1d8b0dc35d56 : SUCCESS in 11m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8fff426ef2384c16b393ff6370dc57f6 : SUCCESS in 14m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d3422c421c754ac789295caf991bf522 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/32ad414bc9fe4c50ba145b0436fa1f16 : SUCCESS in 8m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/01c2601323184d139c384ad279f2fd07 : SUCCESS in 4m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eaad824f87db41c0ba5b06f97b012cf9 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f9ff826f4af447f58a59cea1414ec0eb : SUCCESS in 25m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a9bb2817cd424db7bdb781535f5986fb : SUCCESS in 12m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c3e5550ab3cf4830b84f4ce759558df9 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ce47ef806584fa388f7de26f930acd8 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d3b337c3f50434bb0ffd037fe3d064e : SUCCESS in 23m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e8510132b89a411781bec7994e271768 : SUCCESS in 8m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/48f2916459f94979aed412b340ba69be : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52c3b2a08ec3443a93a03936f9e4fa57 : SUCCESS in 5m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/85be2923bcb6449087d40bf65b659fcd : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb8528ff40744cd0a3a136a1853b0b8a : SUCCESS in 53m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6339d08a21c44e13b4658eedaf7ba8f5 : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"715d6f6b481a7ae7e3b0abb3b08680037bdef8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 17:09:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"385e1f64df09b7851f131412ed1f6e9c96373fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 18:06:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/425eaafd63744c7492358f261ebbc95f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833eab60bc7c4e8cbd4847a5c683d33e : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31894425acf64c3daae2750d2b096013 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/941eecfdf5ed4f94b313202837885497 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89beb82cb634353b78db2e7d2bf9572 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/994e4342df5f4a5cbb08348199c62efe : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c344a8714e794a6ab373545aa00b26e7 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17f2656738674f1ab778c1bd56adc7ca : SUCCESS in 6m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4b1072f44b034248b47715ca3d32319a : SUCCESS in 9m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8ccfb32b2d1b4ccdb6e57709d86643d7 : SUCCESS in 10m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/64cc51d6285d476dae950c0fe6ac619a : SUCCESS in 3m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/579e57faaa2d41ac99b49d376795b989 : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ad3b56ae7b7c4d3ca5c5eab7e7ce3bbd : SUCCESS in 6m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0d8c06b0bc1a46819fd79ad0900db5b8 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be1fc3ac9a7547febb42622b6a68933b : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4984c8d019b9460e8cd3a5eaa56ac922 : SUCCESS in 6m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/612399c4deb54f139d34cef422150dfb : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/42e960a8a7b0492588c5b3b797946c6f : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/43b1326625e24f199fafd36b9c746a1d : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0ef087b8695d4423b516a9fbb64f71ec : SUCCESS in 4m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/009656ba6c0d4ea6a4d418665cb14702 : SUCCESS in 26m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc5b966a28e2492b935155642f260f22 : SUCCESS in 13m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5d09432867ec416b9d30253ce1b68de9 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3242581ddb64ba38405a76a34947bc0 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92dcf71af76a4632859bd62b4cf6541f : SUCCESS in 23m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7c2ff22088304f4ca7588968392f7fbb : SUCCESS in 4m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/611df7f7a64e4e0fa3f595096afab4d8 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aae2fea1e5874745b3a6fafd2d0f2874 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/87f8034f0c9b4690b9225dd3583be87e : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09bd26b1d3cc4ce4baac3e1fe8cc8467 : SUCCESS in 37m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90d88aed34f24950bf44c8c295071fc1 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2206ea4ba63a181c61baee054fb042c6b9f38c61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 18:06:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fdd58d4d200c1e666bf9fe4b1054a3227786e143","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 18:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4447aab917472da163bb73a72f8e75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c9c5da61af44cbaf1c6d52dae9de2f : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"994c9858589a98394f368ae032988b998204583f","revisions":{"994c9858589a98394f368ae032988b998204583f":{"kind":"REWORK","_number":1,"created":"2023-02-17 15:30:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/874265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/874265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/874265/1"}}},"commit":{"parents":[{"commit":"6fb0d2954a1610a1be6237d4192d44b5820e3e7f","subject":"Merge \"update constraint for ironic-lib to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb0d2954a1610a1be6237d4192d44b5820e3e7f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 15:30:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 15:30:41.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 8.1.0","message":"update constraint for python-troveclient to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I0e4d298370379d7ed76289f4345fcc5be3efb28b\nmeta: release:Code-Review+1: wu.chunyang \u003cwchy1001@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I378e9014433efcd28730cdbb46e4830b5f617a08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/994c9858589a98394f368ae032988b998204583f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/994c9858589a98394f368ae032988b998204583f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
