)]}'
{"id":"openstack%2Frequirements~874274","triplet_id":"openstack%2Frequirements~master~I826650f736acda90216c6b0ae5d5f678755aad1e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I826650f736acda90216c6b0ae5d5f678755aad1e","subject":"update constraint for python-zunclient to new release 4.6.0","status":"MERGED","created":"2023-02-17 16:23:05.000000000","updated":"2023-02-18 00:19:51.000000000","submitted":"2023-02-18 00:17:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874274-new-release","meta_rev_id":"58c4176d6f895f2a129bd2d0dc22928714c19302","_number":874274,"virtual_id_number":874274,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-18 00:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 18:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 18:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 16:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 18:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 18:36:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d1824b9ecb3fe75171a055e354e58fceec9c138b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-17 16:23:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c8028cfe09844e032c9d39ecff3eec7828b5d90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:45:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb22ebddef44b37897103eb90109036\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91b6a063e5f34538885611e2e454da15 : FAILURE in 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e832c17295569937d950c2d20ac8a78fdc5273c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:13:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe31aa494b7a47218015baf480670986\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477e8362560c444f8f14e01f63eb7c76 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb5f685b977498fa199efc0142582e6 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f42f0e1329c4fb4b8139eb562111f04 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4babf297a6ab42a0ba2006513d8b1ba5 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e6e7d4d9f3240d182626e35a59cf430 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cbf4e854d434ba5be9f79df2cc51f4d : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8de6fffa712e40d9b7c6327321c49674 : SUCCESS in 7m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c62b9bbf0f0468ea823cb582779361e : SUCCESS in 12m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/55be8211feb748f39df73989894e8a77 : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7bcdfef0f24544888d23e4d5da42ba15 : SUCCESS in 5m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b0decce7f2904584a3fd2abfca69c002 : SUCCESS in 6m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/810cf66523314a5ca906644ac652a814 : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c61f230f7bca4755b2e311f408594341 : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/eadb0a63393244f0837d437e488cf1c9 : SUCCESS in 8m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/afb5400fbfcd431590efee8456d6ad3a : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/556f2640d32347f6947d4140116fcee4 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f09cf0434d314810b4e4fe2184e25593 : SUCCESS in 6m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/495a013019bb401485394f62230edb55 : SUCCESS in 7m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/decf07b52a8b440c80d732495f2928cc : SUCCESS in 11m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c68c70a4f1f64fb9b3d2920b71ba9579 : SUCCESS in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1985ae225784fc19791ee8c0c2b9589 : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/08f1b47351e24c46a6f1c82afe485730 : SUCCESS in 9m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/330af6be4ac445748e07cae5c1b4edeb : SUCCESS in 13m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3c9003ac71e342738dfb9cfc8744d019 : SUCCESS in 46m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/23ead9b9dc374b339826f9e9385cd9cd : SUCCESS in 12m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e052664cee0e4b9c92eab6b6fe4ac7ac : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49cfaa84d6d447b99c6265185b5927ad : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb08f2a6422f4e0eaecc66de75a8903d : SUCCESS in 23m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8f31a472fafa4b6892d0832ac17b167b : SUCCESS in 7m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1b6d11eb2f284c6d832ec69502e33664 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6cb7e4ec2a514f02a3829dc290f5dd68 : SUCCESS in 10m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bbaf943f6ddd4ae7abc18331bafa2499 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e14db0d66c5649fb9bd58147eda4d691 : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9a84bfdd8e94b0a8229935d0b1bee18 : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8ef702327dcf237dcde52c14b500587173c8907b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 18:36:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e6876519480f58c8d99806dcd935bc8f8209a86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:37:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5fde63f7a063c015dd97faef3fd96af70e1bccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:17:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb11533ed92241cda98a148d5d3f65f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4cbd33ebd145f5a21546e4fa13bd9c : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b394064fe4a7401b87352065b9c4d8d6 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43f084fceb1c4acc9b43ddf0f87a9022 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12e92e8aa097400db4833e128a047f3f : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d02d5fc846c4f4db7518e097077c05d : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a7e76f7f28042ecb306185d201bab44 : SUCCESS in 4m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/898c4563652c46e4a44bbaa1d98b41ae : SUCCESS in 7m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b02bc443f24047a983fb7fc3789c6375 : SUCCESS in 12m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b76f6909775242a6b4ead23199c05e21 : SUCCESS in 13m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6115ddc6a9864614932af6127da72f3a : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/32a13704febf4a2f87d9979adf87eeea : SUCCESS in 5m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1b510d74656747cdb407f80803117144 : SUCCESS in 7m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/03c11e0b53fe46deb45cf2941df68bf5 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d638f8a3afd4824a1467a86e5584656 : SUCCESS in 8m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3de96e440f554c8591f82c4e72acfac8 : SUCCESS in 9m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bd248366da364a26b72786667d4035da : SUCCESS in 16m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5b2c162c73b644b19ec3f3d987232d3e : SUCCESS in 4m 20s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0109b4fac5e1419db790721905a65d58 : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/89201244bfe24a6ea3531b79a621893f : SUCCESS in 5m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0f24a7e6cfd843bda424ff9719e7de51 : SUCCESS in 22m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5e31e85269234e38a866c18f1f76f164 : SUCCESS in 12m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d7bad837979c47e98480d452693ba618 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e06d818f44c94cdca1a9ea3569f08d33 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecffec2bd4f74e6c9a1268885422243b : SUCCESS in 21m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7795b40f11ce491894489acdf61d000f : SUCCESS in 6m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/342b259faa9b45f88da5594777c9ef19 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69566c71642e42cc822408de8f1fcea5 : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/77bfe8a24f1a41f0986662740a4f18a3 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2fa373084dc4e21957abe4b046b109f : SUCCESS in 1h 10m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f8dea87412444f91161bf614f6cf4a : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"63b55c146b79e2d5c2f6f069517be0f512a92c84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:17:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58c4176d6f895f2a129bd2d0dc22928714c19302","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:19:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9238f8b87414b0e9f4af67d564d9fbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e9194fa5d9430bb20ae7ddf6b55ceb : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b326e3f5a997bd6cda8dad9c7894ee4316791de4","revisions":{"b326e3f5a997bd6cda8dad9c7894ee4316791de4":{"kind":"REWORK","_number":1,"created":"2023-02-17 16:23:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/874274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/874274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/874274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/874274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/874274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/874274/1"}}},"commit":{"parents":[{"commit":"b454cfa85b11974156c8354ac91c2fe9f88c779f","subject":"update constraint for python-designateclient to new release 5.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b454cfa85b11974156c8354ac91c2fe9f88c779f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 16:23:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-17 16:23:04.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 4.6.0","message":"update constraint for python-zunclient to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I2eed7365e759e94a08c14df29d7ade98b968bf9d\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I826650f736acda90216c6b0ae5d5f678755aad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b326e3f5a997bd6cda8dad9c7894ee4316791de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b326e3f5a997bd6cda8dad9c7894ee4316791de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
