)]}'
{"id":"openstack%2Frequirements~874552","triplet_id":"openstack%2Frequirements~master~Ie0f6c80abc41629917c768222c3308d8a73e0da6","project":"openstack/requirements","branch":"master","topic":"python-subunit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-02-21 21:42:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0f6c80abc41629917c768222c3308d8a73e0da6","subject":"Bump python-subunit to 0.14.2","status":"MERGED","created":"2023-02-21 10:59:57.000000000","updated":"2023-02-21 21:43:47.000000000","submitted":"2023-02-21 21:42:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874552-python-subunit","meta_rev_id":"9697c54c460e24ee04c6913c3da835df44866431","_number":874552,"virtual_id_number":874552,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 21:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-21 17:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-21 17:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 11:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 12:55:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 17:43:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"060e9b25a0f2e52768b1d06e57f38d83eea4ad6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-21 10:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3781a850039646c4bc88a597e975b711efaa54d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-21 11:00:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"763815cc5eea1f52ecce4b43cc4a93ce5af93b5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 11:06:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0c08bf0c30400285c3b0f400b4f11b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/08d7122556ad4c4292ee2f9fbd2fbce2 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"70dd55c730f137c8ecf2286893f1a8aa1210347e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:55:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a04d117b5d4ea29369c99b36a823da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dff539b1f1e4e21a69897290312bc14 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc5575792224013b00d1c346a32417f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a874cd00491a4fc5bbfd8d75e4b8b496 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33907b6162724b9eb2d91bbfbd92a7f6 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2ff5159081d432688a95ebf5b533b53 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68a6dd16907844798f222edcab4f4598 : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/749c6a717486472abafd07089e2db7c4 : SUCCESS in 9m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4133d5ce6c2344c0bf0eb9b579fff1c9 : SUCCESS in 16m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ef368b9f3d1a4c4bb838ce8ce1e626eb : SUCCESS in 6m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/480370da7cbe46d7a3ca26323c3b87f8 : SUCCESS in 4m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/922f93811f7b4d2b8b3f833f1063ae57 : SUCCESS in 5m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6751d152d0d3472096f40835c2888e5a : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c23532cbfdb0405ba0b06d3365a81a80 : SUCCESS in 4m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8b323fabb80f400bbfd19e9fab91e1bb : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9de6714f85344a0aa360c61e0c26e41c : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d9a0cbe2e3174ef88cc8223f3e9fbf86 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/178f19fd68ad422c99daf126f9b12b03 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/27c066547b3f4ef2a650f2a07415be65 : SUCCESS in 9m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e4476d55ed824bf98e1deacca9db1f5f : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1eb3a8412f13419db520276a21cd3756 : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cc2d2d0bd9d44c2b810974c94fb6ae86 : SUCCESS in 7m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/859abae47dce47029143f0d6016be3e9 : SUCCESS in 4m 39s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8f126d4c2c9f48b6a097e079f0f7ce9f : SUCCESS in 8m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f26568c96f2f4c46a60ac57951460978 : SUCCESS in 19m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bfe6c33f16674a138584fd8e0c36c0ec : SUCCESS in 13m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2a6e1b8198b947c5965b14f77905cdc6 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85033408b4f943028df5eb443d15bc5a : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1365fb0eb76842d5acefecd2e3cec350 : SUCCESS in 22m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/da65db52f5a24d91b981f61c242b3265 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/608a065e287f40e897ee267a106d9a6d : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e00209633e14756bdaa36f630d9f575 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9911aa492798420fb69472713f441bf3 : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b81e502ec360420ea09cd31ac0e83628 : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01d25fe6aaf041dcab8a175b0d2d0be8 : SUCCESS in 1h 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3ddc3c02ca871b9237687f65e044d1306409ffd6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 17:43:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e21ef1d189a0ebf9902033e49c623154f43ef45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:44:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71a86750477b6b681d9eaea527b7f98f2b80a9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:42:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5b487da5c248cd9cbfc374e5be9e55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7945b8b412d24e2588eeaec4907fa358 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b35f0656c6c49988febca3bd6abc8e2 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f04b1543e94540b287837efa85d726b5 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6b880e4267042f9b1bbd44099f6f05d : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93c1d16d31054a9a860aa20c81e694f6 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a059f25f1e1d46a08a39f7a94f62ba2e : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/663837020994490a91ac03601a4277f0 : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/27b553a6ed934c1eb787907313c19a27 : SUCCESS in 10m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/df5c4139e86840128c5ce36fa6599858 : SUCCESS in 10m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fa364ca58f0444ebabd7019d2d7da842 : SUCCESS in 5m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d60fe4349e874fb48c9c0aceb4795b7d : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8ef4a5cba3a34e2184651c62db8d2f92 : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4ff05f3c7b8c4fa0812d8fb40a96a4ce : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b4b2c1820f747c1a6ca36f37dcff8b7 : SUCCESS in 13m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/76c45b68c886443783f8128ab4b23e14 : SUCCESS in 8m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/daca581f8b68422eb5a4020723bd404d : SUCCESS in 11m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0cb64ed585fa4a1993a35b8f90d1f2cd : SUCCESS in 4m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bad814a8dae54d39be6ed2500bb76de5 : SUCCESS in 7m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2bcd73d518274debbf15e50cbff0c7e6 : SUCCESS in 6m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/238039b42bfb44fdb3b379583e96f627 : SUCCESS in 20m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1f7fdbab37994a0db98d57755086f277 : SUCCESS in 14m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/849eb60bd3d64a2b9e8c2c943e96427c : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/514b20b789d04b039cb1c7cf723f95fa : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34e192e8600b49d285d80526352a7e35 : SUCCESS in 22m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/11adcea19577429a9b43cf199ad36e02 : SUCCESS in 5m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/74ba8dc4eeef44ccb1af0ddc627ad130 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a264369d11c1470082d21a4ad0343f69 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0bb8b2fe6bfd40b9a7d0991874aa9de6 : SUCCESS in 27m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34c255c9acab451f95f26560627285c8 : SUCCESS in 1h 05m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2baa8b0afe34f84804a7ad3b369ffee : SUCCESS in 2h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"13053880816ad4f191c941104a880cb53b624667","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:42:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9697c54c460e24ee04c6913c3da835df44866431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:43:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e8b5938da74a0dbb21fb4d252fe3c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0803bdc3f14315a7bfbf9082be661b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17c3b94984ee9cc3a9f9b9daaff60c8e1ab2cb32","revisions":{"17c3b94984ee9cc3a9f9b9daaff60c8e1ab2cb32":{"kind":"REWORK","_number":1,"created":"2023-02-21 10:59:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/874552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/874552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/874552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/874552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/874552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/874552/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-21 10:57:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-21 10:57:27.000000000","tz":0},"subject":"Bump python-subunit to 0.14.2","message":"Bump python-subunit to 0.14.2\n\nThis resolves \"invalid escape sequence\" warnings [1] among many other\nthings.\n\n[1] https://github.com/testing-cabal/subunit/commit/fbd95162279\n\nChange-Id: Ie0f6c80abc41629917c768222c3308d8a73e0da6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17c3b94984ee9cc3a9f9b9daaff60c8e1ab2cb32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17c3b94984ee9cc3a9f9b9daaff60c8e1ab2cb32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
