)]}'
{"id":"openstack%2Frequirements~874670","triplet_id":"openstack%2Frequirements~master~Ia716311c51fd219e3ec620ecb43cfd2c47012939","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia716311c51fd219e3ec620ecb43cfd2c47012939","subject":"update constraint for oslo.config to new release 9.1.1","status":"MERGED","created":"2023-02-21 17:30:00.000000000","updated":"2023-02-21 21:42:19.000000000","submitted":"2023-02-21 21:42:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874670-new-release","meta_rev_id":"31a7c0786e44d511c00ffd659ca792be2382e8df","_number":874670,"virtual_id_number":874670,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 21:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-21 19:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-21 17:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-21 19:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 17:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 17:47:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-21 19:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 19:38:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"76b8b1665a765af8504987701137dc161d333ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-21 17:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b65420427fb32b7bd2e1a2172920001ebec7be3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:44:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad8bc1c22524a709704598d8fe05808\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/da4e173b10604420b09dc340c516d821 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"98fcbc8585cb2c029c06183670a2a45fc5ff1ff3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 17:47:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97763678afaa0169303d400540e8c376a55baf2c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 18:02:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175b3dd116cec8417b752f4a036fbc026de4ea3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ff179db62b46a39d420d5c37911630\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6362d1d235134c03853fb15c1c261539 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389a61a35d934ac1bdd925d5d2d050d6 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e00dc1c40cc4852b4feeca049f9a493 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87bcbd5e50b41f1a1f99bba1b34d256 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f9d1a65995342e9b7fae8d248f6cbec : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7533760da2c3445b9d984726692c75cf : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b0221558c714ad89e714c0c1ed58aaf : SUCCESS in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7686031b71354ea58097acf083f70933 : SUCCESS in 13m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/47e7eb35a22f49f2b9fce829c7962b44 : SUCCESS in 8m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f0227fdafab54e719409638257e1b839 : SUCCESS in 5m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3c5899f714a44d28b9d3baaab1674c88 : SUCCESS in 6m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bf1af35e6062415d9fb76db5f15a2468 : SUCCESS in 10m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/670cf5df70b24f9a88a58d460b7b072b : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ed779146bef1471d817699601258a783 : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d98ab28972a14650a431e1f6862b8b3f : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4be4344a8cb340aeb531d1772f4436d8 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c72b2567b941499583240a64da2a0f0b : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9b36537d465d492895343eda217eb7ce : SUCCESS in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/453f4006017e4fff8728e330d780234a : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/40df84eb4cbc433cb5f6ef092d635aea : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/06235a3f1c104ba3ad191f1340b0e03c : SUCCESS in 9m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4dad830e511047bcba0eb9b151f1331d : SUCCESS in 4m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0b1f0dbf7f614f5588ca66712c34c586 : SUCCESS in 10m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/54b96733a2924d66bbea46b612405d60 : SUCCESS in 23m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5d00be5346884af7a13a50daf688d03b : SUCCESS in 12m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fba8d14ac23f446c8b9e3c94476c77d2 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b4416e025e19495b9a5243459dc236d4 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a56969733ad74069ab290e1764bb339d : SUCCESS in 21m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/72bc993906ac4e86981b2396794345ea : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ac9eb3ce3d8543f0a3681af0ca4dac56 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b4fd7df99bc4b52bce074b5e83067fe : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a3e85e21697946ef8b4a1236ccdf7594 : SUCCESS in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94f418098199404ba87d5178fe2b9f06 : SUCCESS in 47m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18ef81c398784befafa5696b117e8226 : SUCCESS in 1h 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f41229d94b0dad708e36f8d5122b3ec9e2f8b7ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 19:38:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45e9515aa978cd57e4819bf8bd972926b3bba433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:39:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83c4067f0b50a8247a25d5ac6977a9175bb15d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:42:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d82f25ad6ef4d0b8b709c89a554b4f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4abb5b5db34ec1b315fc0b3b6ad920 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a325109c6fab4ca180ae499a3adaf04f : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c649ba53bef74864af455d116f5d6f17 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b442376707b44fdbdccbbbba0572203 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df6821e3ab9c4e79be5b0ddf19c40d99 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc3e297121324baaac26553bca1b8219 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b3df900a55a4e3797a1e612d9af1b0b : SUCCESS in 9m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5927716471d94f24869dddb78b6198a5 : SUCCESS in 11m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f881366192704c6da0c8d42e279705c7 : SUCCESS in 11m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6edbae9c7aca4a48b30b10c0b1347925 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2cd9df5eae6747c68cd84b8d9d761e33 : SUCCESS in 4m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dde50e294444417f8cf76b5f66e09a66 : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/729a6b0fa0564e168c90a262919d22f2 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/532bca06624b4b688f981b3d8684c921 : SUCCESS in 26m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fbaa87a35d6749e8a90cdd6478e949fc : SUCCESS in 9m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9eee21df58f34081be375aa1505b1c32 : SUCCESS in 12m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6a5e191f319a4fe38b92c910d21c0849 : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4f53f2db85b744699cf45f1b0b10e0ff : SUCCESS in 8m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5f34ce64588d4397a72660217b4fa8b7 : SUCCESS in 4m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/eb333193d7a94980ad8b87874bd9cfa5 : SUCCESS in 23m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14128eadb30b4616abd195ee705ee66b : SUCCESS in 12m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a0b634431e8443c8afe1e42a6af58f4b : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b18e3801ea794ac1872b3e4e38e6d011 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed670afe7e4b4a909d7300eb9971c869 : SUCCESS in 24m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b790ea3f5e0e4ffca1619c99a63498aa : SUCCESS in 5m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e733178487304cb6bf0e07b62910c44c : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d04cdcf13e3e4e52b0cdbe5475278fe3 : SUCCESS in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/825974b070004be69767b9e13fa53147 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40224d63942747cfb08bd07628929b43 : SUCCESS in 50m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36ccae89de4b435aadee6e1343066724 : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"31a7c0786e44d511c00ffd659ca792be2382e8df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:42:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ba68f0cdfdc672a333fa67f71ba2d004bc09fe3","revisions":{"0ba68f0cdfdc672a333fa67f71ba2d004bc09fe3":{"kind":"REWORK","_number":1,"created":"2023-02-21 17:30:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/874670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/874670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/874670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/874670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/874670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/874670/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:29:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:29:59.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.1.1","message":"update constraint for oslo.config to new release 9.1.1\n\nmeta: version: 9.1.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4a971a18d490b465b7aafad9c8b8f10e0bc96ebe\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia716311c51fd219e3ec620ecb43cfd2c47012939\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ba68f0cdfdc672a333fa67f71ba2d004bc09fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ba68f0cdfdc672a333fa67f71ba2d004bc09fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
