)]}'
{"id":"openstack%2Frequirements~874671","triplet_id":"openstack%2Frequirements~master~I7b88f7ca22f048ac3fc533e4cccdf0457462565f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7b88f7ca22f048ac3fc533e4cccdf0457462565f","subject":"update constraint for oslo.cache to new release 3.3.1","status":"MERGED","created":"2023-02-21 17:30:07.000000000","updated":"2023-02-23 21:41:37.000000000","submitted":"2023-02-23 21:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874671-new-release","meta_rev_id":"b2d9dc7231e80732eccf6ad6d2453d404624b8d6","_number":874671,"virtual_id_number":874671,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-23 21:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-22 00:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-21 17:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-22 00:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 17:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 17:47:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-21 18:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 21:56:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-22 00:35:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"26cd8b8625e6f964e1bd024e132159a7a0478955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-21 17:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17ad558e40e95796cd9424ffd0c606823a859e37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:40:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e597db04812645a6a81ed49657617d71\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/93631927e8e948c4b5ebf0d245b072ac : FAILURE in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"70bee4069faa24e769783d5707cce9673105cc27","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 17:47:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"427c02d032b7c82e613f100b8be25ed58ddc1887","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 18:02:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7754f6c39f204d463afc6d859b4d1578ba29ab0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:55:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b489dd4dedcd468b92871b161c6844fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b56ed6fefe45b6a5052ad845b39036 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778021ba19b141e6ba8a3ae6eab84bfa : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ce8eea4389478ba78f0b35e752d554 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47a14fa06201469da66b7a9ca0b7202c : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfe9fe37e3444da5b3824e99f24e68f1 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d02e37b70874da3951a453ac1cf5d1c : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f40feb2661ce43fabc1319cec67d88b2 : SUCCESS in 10m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e9b8ffb5d0654dfc940f319a516409b8 : SUCCESS in 11m 55s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9d9df717496f4d05b16c4656f6d029bc : SUCCESS in 16m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/890d34b0ec6a431d8e2bd9aadb1ca8e3 : SUCCESS in 5m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6cdab7f80d2a41e086631cbe3f58fa4a : SUCCESS in 6m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dc2efa9acc6243c0a6334dcfbf273085 : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b8fecacde2ba453e8b4fd9494de54d4a : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4f35695a11334d029ad03c209ee2aea5 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/849e068e6c8e41998f3a07bc5d308812 : SUCCESS in 9m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/38d94e629eb345489fd3e9aa1e20113b : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2be573090dc425c9fcfe3f75414ef51 : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c9d04585a7cd432f9099957c24cc029a : SUCCESS in 10m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/73803d12c09b45ba95f44fef763fc1aa : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7790b05877a640ed8260dfeb1c3f439e : SUCCESS in 4m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b268e54dadb2465d933c90b0363420a9 : SUCCESS in 9m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0347ec79675b4a2991dc5726d1124255 : SUCCESS in 5m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e07463ad01734499a92df548f2d1d680 : SUCCESS in 10m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0b84b72642334f508236ad12610d9be8 : SUCCESS in 23m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fb56b18c9f094d69a94d32a646e268bc : SUCCESS in 12m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8fe625d8e95445adb437b7d6f5be27f1 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4709c053a904be798343834eb61a73d : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af4fa51e457e41e4a6ad31957fc24271 : SUCCESS in 24m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1865aa1c612a4b84b12bc0ca9fa9dbaa : FAILURE in 4m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/48d015208a0142d2bf6687efa8997ad2 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc417d0a0713418eb0be3c1b4770cfff : SUCCESS in 5m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/45935efa6aef43c2829c5323b3454291 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8d0f22f09ac4620b21c9bba04565d5e : SUCCESS in 56m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d66153d475de47c0a7c4b4df54012431 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e97897a57c3f9e9cc0544c09238446506fcb00f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 21:56:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ca06bc8a212daf5f639f76928ad66ccc2f6e20c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:01:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/598fac5efac44870a89938243f28d5b2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5615cac2d7f34223b4857f118a7af0aa : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6970ad83db2a2bfab08d9e13b40a2f848ff7ef74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 23:14:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/550d47d31ea04b1a94a50e30ad5c979f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4940aa1b5886430baf8afffbe50a65f3 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2430562f39b4adc8bf313d09991c032 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2303050ca66e486db5b06cf929a7377c : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2c4485c5ab64abe84ae0e699840345a : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68f12eddb1dd499a9a16e2f1a240aff9 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2408c0d39c243e7b743a855eedbfe84 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37214eee15dc42ec91b00f8b5031ae52 : SUCCESS in 10m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad4a8fde1a4840afb85fe89bab943432 : SUCCESS in 15m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/60117cc34bb0434b9ed410dd081b1610 : SUCCESS in 6m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/caa75cc54a49415cac91ec14da3b62c4 : SUCCESS in 5m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3ca77b1ffc6745d394df572ff94e0b69 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5d94db68d12d474ebdb798b1e911988f : SUCCESS in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/83409adafb85402c86cdb825656b0038 : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c87b4dd1a89e4ee7a68faec2190dcee9 : SUCCESS in 6m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a3d2e07c934a420eab89a8b438aaa9c2 : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/18df4ec685b14f688b8fc68a38e58a2a : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f55708d67db4644a38edfb6ccfa63a3 : SUCCESS in 9m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/af395c79a3a74cf49e6be32b81c99b19 : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/15c754a9408b4601b4d1e5e7e9e79f47 : SUCCESS in 12m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/be4d447d0f5b4efe8ab82fe0883a0057 : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8ca8e911297c4ae09a7ead5b561f2457 : SUCCESS in 10m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/adc44997a9904410b7aa55a73c8e4774 : SUCCESS in 5m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/df5c436e5f3d4bb89e746eaca63ad448 : SUCCESS in 9m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/afd7d017bf6646a0ad8e20ba551af38b : SUCCESS in 24m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/57d9224f4f5548cfb609454b051dc5d9 : SUCCESS in 15m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c93d5384f3f946d3a3cf10e31da875f8 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a40f9f76b0b4774b814ff75cae844e7 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6331561164a34f0d8f7e43cd6e23ea80 : SUCCESS in 21m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/226f060bd7454962af5019271b78f9fb : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/843b64b951614fd38edcc963bda9de38 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab391a1479ea483e88d3a87182e2b2ae : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1ff81acb48cb4e0281ef7faa2a5181a5 : SUCCESS in 16m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86cdfe4e6c034bafb5e71acac3516675 : SUCCESS in 1h 01m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4b21347750148b1a1177dde42067585 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"006668212e3daa45266bcc1c3e45aa54ea65e061","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 00:35:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6677e1ba067f67afd3c017eebecab1dac2073731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:36:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cba6ea9d83360915d7a2967c1438eb72b549c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 02:10:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe3afa95913c4ee494b78085518ce091\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d84587eca34481aca4f82a24770b0b : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6229881038ae453b8f4b8e6bd04b1042 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c598fa63660d4a0ea7bc4c77ad44cccf : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e84be36256ad498cbc13208adfec967a : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a66f72731c1d49b3a40252ca6ac75ef5 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4cf15404c65407fb2da34ddf452d28b : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6729403e87b14995bed684edef577a9f : SUCCESS in 9m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/883be58106d34e10aa01b60f523188c6 : SUCCESS in 10m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fdad40f16be7453f90147456f6659c49 : SUCCESS in 11m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1007b3b5fddb437797ff880156d719ec : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ccb5e9ed23d145c7a6d8bee7eced212b : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/07b625202bff4496bd7aa98ba8fb7942 : SUCCESS in 12m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/27c952272faa4eefba650454c80f4cf7 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/408c0dcd0d84457faf9a809bc0087969 : SUCCESS in 7m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bfd1eff4a07a48dd9e599293a05c990f : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8c4bcfc886d9481ba9c138f4298c24b5 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/04cb803277c74f6ea9ecbce6f29bf228 : SUCCESS in 6m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/058edd493c4e4f94aec9d3864c9526a0 : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8adce0edd4bd494e86e2a6fed08c237c : SUCCESS in 5m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/543b511ead6a4d85b1bb580e9d958158 : SUCCESS in 30m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0bb03fb116c247ce8ee8f405dfc11393 : SUCCESS in 14m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/14777dbe88934f32b54b871659f748b0 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/292cbb6c0e46415ba7716414918c0430 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/069b2c3fed8042e3be1cae4e9ba0dd2f : TIMED_OUT in 40m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1375e0b21c394a39a2cca20c8491dd8d : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e4a151afc38c422bb152ef675a74c7f8 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b256f8d5f05411abcf879fb0cc31e6d : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/74c8f52f4e614ac3abca18e9ab00f434 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf902a7f7e7c4b25be38e7f3f5280998 : SUCCESS in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c74569a511b442c1870388b3c66829e6 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b19d7ba95fb75794f752e02d7b72bc565dcdf8eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 04:29:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e084d1632e9e0bc4ff5ad49064799b9fed3592f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 04:35:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2119667ca20b4ae0b378d1303bf4fc07\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dd4f629de24d49cc81d31c719423b92d : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"88000b4993a0efb15bae451bbc6e37918326c164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:03:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13646d239e1245028d04cacf1bb71398\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9463a521700c447f94e4be6277863fc7 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72d89932cff4689a334d91dcb7e1146 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b637b8a011840b8a744b2d301afd43d : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3e3ad9f937c49b4a7766218b5d889b2 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fca594874bb0431cb9d513bc6d925e4a : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e19e4e806e943c3902a587b41bc36ca : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c7227f1b3b8a4650ac80b5adb3a8f3eb : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c0167e68ab4448bbb290d47aa002718 : SUCCESS in 15m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/27c234fd0660438d988524b31be4ec2e : SUCCESS in 8m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4c87d5baa3cc4e769fcfebdce604a96d : SUCCESS in 6m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8fc94c65841443a19bde1867eacf2510 : SUCCESS in 6m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/daa7655471e1421b85fca90157420119 : SUCCESS in 11m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4c4d32cc4c394352935b3ea739224ddc : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b4c0d12be2854d53a3754d36fbdfb827 : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/89319d5fa67c4e64b0e8414adc8e7f27 : SUCCESS in 8m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1361b680290840da9b0e963fef8a1810 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d29e4d756374215809d94f687362607 : SUCCESS in 28m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8b3b817ac2d9497590c6147be90ef76c : SUCCESS in 10m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/17ca71a828c04c7697ff263a245d30d9 : SUCCESS in 13m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/719b400a50344d58a4f7b26dc8f88e06 : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/467518bc05244b41bdb01876deb327ad : SUCCESS in 10m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e3eb797cbe2e454798b6d922a188e21b : SUCCESS in 5m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/27af3412127442a69e021f9b746fc1be : FAILURE in 10m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/09f2c4b4a35f436cb389fc65751a5fa9 : SUCCESS in 22m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8ce39e06dbf468d8a45c5c3fc1fbf89 : SUCCESS in 13m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/62276791623b4363b58921e85c403a4f : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6c0a2078c3d647b2bb4d2bc7470b27a8 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65f3bb5016214f889bdccb5c673ff742 : SUCCESS in 24m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f927f838e1e14ce0973ea4289045926c : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/27393ed537674be88d9aea89c88d9235 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/483c75dd61d14024b2a90e3a1ef242b9 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5b1d898752494e12b9f7b8265cc2bfad : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/528a94a6a926464894884bb997277e7b : SUCCESS in 50m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8063e44668c04098a4be592301498711 : SUCCESS in 1h 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d162a803a26345b35e732ab336eac0015070e6f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 16:02:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d570cac25d87f9c33446ae7a3c6dff645f7a3702","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:09:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2123fdf5a114e4d80d985535bd79888\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/085d0bd621364c0e8034f9ea7d039ee8 : FAILURE in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c0df149279f491b0be4ceb10a1ba7273c5182305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:30:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f4ffc30b29e4444a533dd48d6c3496b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e5931bbebb436f90bc3101e39cbec6 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c78cdbab05447e6b60b32d34d22beee : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f137a8effd412fa66e20994ce4ffbe : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c5fb24af8e4dd0b961ead78e33626d : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59f29b21767c4b94b6068fe770f3678e : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b70952186f5143c98dc77733792fa1eb : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19a280fb75a342fdb822e2c650ec8e90 : SUCCESS in 11m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/078862d2d9eb4c4399f002310ee61194 : SUCCESS in 14m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d1e01c53325941ce9e3d5f664cca4906 : SUCCESS in 13m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a35929ea5767479ea2a52b0d385db467 : SUCCESS in 5m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a30ed8b3205a47c087579d635ff940d5 : SUCCESS in 6m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/798337ec0102468ab253b72bbefa219a : SUCCESS in 11m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b99668f62fea41f1b2c5a37fe4f480ab : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cbf5a5ec6fd34b2bad07e2b5c1b8cfa8 : SUCCESS in 7m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f9e9319da40043648019bb21f181a894 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b1bb392455c247b6a2ed4c0eb5d30be0 : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fed2cc6c502742d2b20c12792b2a7d93 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5348b65012d544fa9af3c8bc53933d07 : SUCCESS in 10m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5810aaf9fabc40b6bbcdab5436899042 : SUCCESS in 10m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/137988d656c648a5a5bd4203b804bc98 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/52ec1a9e451f46fc954a27c783269fe7 : SUCCESS in 9m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/82cec1daa5fd49a291ed4e31b82982b3 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4205e57904694221be377615767d72d7 : SUCCESS in 9m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4d1f3a2830fc4d7ead93e5cc52037c63 : SUCCESS in 22m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a9fa29cbdd1a4f51b7147440598c8e19 : SUCCESS in 15m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/15ecca443af84257a382c57d13954d13 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40f88c1011fa4e43a81480e208fd78bf : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a155886e61649cc85c340cb03fc4099 : FAILURE in 30m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/566d95f10cb94ec281f4f912dbb500b8 : SUCCESS in 6m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/06f14411b3ec44e4a74a31c4e33efd8b : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fba8dd2a3b84052828ce4ad6fc466f5 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/74597f7f57104e7dbffdebf23f29e3ba : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aeee3322ac0c4ffbaa96f553cd8f7b58 : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9655032118c74dc3b3029b5738d02d58 : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"283c8485ded0a034cc78c8f36511fb050695a61a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-23 05:46:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cd24fe52833ac17d0f8c93f54916127f08e4775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 05:53:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8e7f03d2ec444796a998fd6d146be5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4406336d0b5543aaaee87fcf8a1bc5ea : FAILURE in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0ed570025dd13d98ebe3e98dfc69db759e4e934b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 07:28:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/178aad24e2814b9fa1032694a2e19de9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c47024f8c334e32806e27668c86e45d : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba4d7daa68af4e7fbecceb6b88e99247 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f560b66ee345b3a0ec32731a744d65 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35fc6f8d103743a99dca17682db2fcb6 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9ee43bfdbd54de797120b1426872042 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/19109753d8d0409a85b598e54359e126 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/610462a1c3614adf8eafe1dc61bc887b : SUCCESS in 7m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/414c60dbf4ab49a0bf39db45ee5f526f : SUCCESS in 12m 47s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/35c5171cf1964e8bb4db4f3581d28d1e : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8bc7d3078b4a44e4bb1fe112ac75a1d1 : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3633ab1c8fc44894bc2a2854ad8272b7 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0e64e109bae34c79b6702b4cb8623db9 : SUCCESS in 13m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c4d271c765774339a4f9a0477dffc17c : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/42fb86fdc136435ea56b8e168c1a4a58 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4660ee0133604649a294d62f740ef81c : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/09309657f6444a229e36f993e8a93a95 : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b3f59109e774ee9988e6b10fe60c513 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7676822a003a4b3bbd2b53b61bbdad94 : SUCCESS in 7m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d3804502923a497e82fbc4a18f3fa251 : SUCCESS in 9m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/085fdf739af941dcad651219514ec733 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/09bb65ddbb5746fb82430ed4c3d216b7 : SUCCESS in 8m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/25ef32e6549843cea328e6ad98bdf092 : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/28a0038f7d7c4628a95861b7f9c9b807 : SUCCESS in 9m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/60ecd85e7fc34f9da512dab13dce64c5 : SUCCESS in 18m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8288388c972b480388230984f220e6ea : SUCCESS in 12m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/91f8bc74594a44b48b0518b97aee4bbd : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f23d4f066c6e4ef6a69c3529128e49dc : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ee3c14d1a3c4863892ab4c7556b8ae5 : SUCCESS in 19m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6a68898d9f7b41f097f3c3be1f5aad35 : SUCCESS in 4m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/32a591c044ec429e9f5ea9290f9a0c1c : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fb64b5c8dd44f4c9271ee08931f822f : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2c96ca6bb7b74f2b81858567a619a96e : SUCCESS in 20m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b7f928746224ce097c02e6b25f606b3 : SUCCESS in 43m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a951d9cee5b409aaf5541fc9ca36175 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"10866eb986cb1f61ec42a7a9ca1faa307b4fd300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 07:29:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31cc4aac84360d965a20f29d0ad5141f7dce1c6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 09:57:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13629279354a4662bbe76d96c99e4e31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba465944b84f1aa7d2f07b379f2d60 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bee773c51b4ecd8a1ea270e185d6dd : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4216d18dd9ca43d183be976b092f8784 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e196a3f92634f8293844726dd1aa31f : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53725860b96946d68e734349b94e9a2d : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44defa124c7045f6996d8df84ed47a10 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/135fde23d4054518b535e0c53a07e2e4 : SUCCESS in 8m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/176ddded4a1e478b960ce4652dd69a11 : POST_FAILURE in 15m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/11f0dfc0b8eb4bbfafc9e5eb44365e7b : SUCCESS in 14m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/83bb6215725d4990bae2297ae7102af7 : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d01f3a2e3a524a9fbd566121aa4d0579 : SUCCESS in 5m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/240a3da466694b029f6c994369a0b6dc : SUCCESS in 9m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/23033b8784064195880dad7d4f62d6d2 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46bd26bff7d94a788afec3afdc3362e9 : SUCCESS in 8m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7a1effbb325e465eb4835b3b4bb04304 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/66cbbf1192b74dcd9529640eadc80ee6 : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/80910c5b80d84266b99b5322aae767ea : SUCCESS in 4m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/62a64d3d74a6405e95dbea333dc57586 : SUCCESS in 10m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/db06fe9495ee4aeb8a102f5a89d9babe : SUCCESS in 4m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b1b8fae529e04f01b3c1016eef8a344e : SUCCESS in 21m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e3174b5a649a44e0a0b75c4f7376e3c9 : SUCCESS in 12m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d98ea782535d436792f343080655c417 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f8f8e7022cc431b90590c7554536003 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f55a470c1a949c788b59b50e94231a7 : POST_FAILURE in 27m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/00d22fc9e68041019e6f86a2a04407e0 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/de9080471c4c4f408098ea9574dd119a : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91e0b979b00a477599f96e791f938a7b : SUCCESS in 5m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/477eaed9c9f148afb77efa4903aa4821 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26246f509bc34249a8fe2d27e4575553 : SUCCESS in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44a4aeb6456b4e02b79edad3b68fb7f7 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"40a9abd303eed95953a8029912ae4122ddee80be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-23 18:41:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdcc3ae27c769c3a10d16f35183277fbfffeea61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 18:55:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3234ddc3636e47b3bc15f05afe6e4c1c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a61da04555e049ed899e4e58d4148d1c : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d916fefb4b9ab8605e9a3b7b3dad0b1d679357ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ab6c7ef12b4be4926b8957d01d06b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb8d30a9dfa4dab8e022319c149fbf3 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434721aa7c914f5ebb8611f9257565e1 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afab5acce217480e860c5e5df9b15860 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c0a8d5d61f6497db03bc9a88f53e4df : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85c174021cc04965ab70aab115eb0db9 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da003c1738984634ae964f4c27b81309 : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33e619244af14c399acd3ce9e4d93c36 : SUCCESS in 7m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8b6b0b243a9d444c91027852de463d88 : SUCCESS in 13m 46s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7c6afd0ce9ef4232b7dde3f9a001e478 : SUCCESS in 7m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/af2bd7c4d96a4f1db30a0aaf0980bb08 : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/232381e08f5e40e9b6bc51c88603a377 : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7ccce923773649389461727e88cfe703 : SUCCESS in 11m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4e2c0b17a479473eab03fe490d0b70b7 : SUCCESS in 5m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cd876a43add649b9b528ce70696aeb0d : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c6ff1c787fe34a19acfd9cbeaaa7d53d : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/68f58f72300b4a4a88a02fde17b4c31f : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4bb8815ec984f7d9372b8267b452bc6 : SUCCESS in 7m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2ccc671d3a4c4cc9bfc312ec0216f8ab : SUCCESS in 12m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b4c09eb3f4a24764ac4ce650616d88fd : SUCCESS in 11m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/20106f4dd971480e88c8c08998c3674c : SUCCESS in 3m 39s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/145736f7db7e43e08282bdf4887ab931 : SUCCESS in 9m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6b009f7665604249bc58c668976e724f : SUCCESS in 4m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6375b880863841e8b0e52745173dd099 : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/415c1df01c7648ca9f94540800d69e75 : SUCCESS in 22m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c491d3f8c6074708b4fae4d721f8052f : SUCCESS in 13m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d96f273b367048c9b85b4caec396d704 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a318a3432f643038e8689f7c0d53000 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cedb06cf4cff4698bd5a6da04b84f439 : SUCCESS in 19m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3a526b95b35a42e5b96d33756a57919c : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/876dc6056c79449fafe71af629503ab7 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27e44930def04b95aa5a4a3e77736512 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e977a685448843d4913ba5447e6d67a2 : SUCCESS in 9m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bb0f5b03f5a4fe7afb90f926d323b26 : SUCCESS in 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc2e72fd407d4c109ed584a3a6d508ef : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"36b54eb983a7c98f4611b849669e87a90616122b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:01:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"816f3d7e46a064f0767028ad353f7eb1d5eb2953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:40:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39bfa333c5564919975fd0c1dc099872\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c939cf2a63846d88030543e8b0f6a47 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea825a19de2446a7aa2a94e212595129 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55eb1f273ba34ac9b872354898a65e02 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d84e71fe6cf14ec2a4d7b684c989010e : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18978f8abc414462a714a06dcfab6ea0 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58389129935c422ca4e1106ef84d8679 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/485b31355ff3455085dead6a6cf30b77 : SUCCESS in 7m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/78985c8bd81549fba25a89cfb1223282 : SUCCESS in 10m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1b7ee95397944ea5ba31605acfaa5d98 : SUCCESS in 12m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6aed7f14f558401eb2ea997a2ea2b257 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/717c317a4b4845599831701e890113bd : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/81399299f27d43d798fa0a6e45e1b65c : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b5177dbae0a14bbe8e8ab92a3c837455 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6f57fed84c344a4a59b93edf0600bb9 : SUCCESS in 7m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/80312dd4db9c4699b5d40599d658a407 : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4ae8eb9285494d46a3a01593dad31a37 : SUCCESS in 11m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/079f7f6e8de540eaa10d796ab0ee1b57 : SUCCESS in 7m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/eb794f1ae4444d14bd6fba3aaf1e0c29 : SUCCESS in 4m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4cfbed4477234f1ea41f9e1f743a3339 : SUCCESS in 23m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3927baac53be4a8db0b2f6226ee40a6e : SUCCESS in 14m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4646a0b6cf1b434bbc262a84192a76de : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61d3de81042c4a42ab42f754f35013b1 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95003ce45d4b4fa296d5d67d5408653a : SUCCESS in 27m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/82fb284e9ca445048b4c14552ab332bd : SUCCESS in 6m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6e9f7c2d7379462b8f19cf13002d2635 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68351ab43841490c8821d7257bf19a88 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/10c553c87f46499490222b8bc19b053f : SUCCESS in 9m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b29c5ba8c2ed42bfbc1735eb89ffec4d : SUCCESS in 1h 03m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6f02dbe2423483db859fb8dbc903a0f : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b10eeec864f50ec0a8351ab89ffbadb77b1cfa57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:40:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b2d9dc7231e80732eccf6ad6d2453d404624b8d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:41:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1837cf0f88b452ba4e3952a63988463\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e014a47a735a4c498d7ad1cef1cd9ef4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43c1e9b7a2177d59c1be93f5b473b305a9194932","revisions":{"43c1e9b7a2177d59c1be93f5b473b305a9194932":{"kind":"REWORK","_number":1,"created":"2023-02-21 17:30:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/874671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/874671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/874671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/874671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/874671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/874671/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:30:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:30:06.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.3.1","message":"update constraint for oslo.cache to new release 3.3.1\n\nmeta: version: 3.3.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4a971a18d490b465b7aafad9c8b8f10e0bc96ebe\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7b88f7ca22f048ac3fc533e4cccdf0457462565f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c1e9b7a2177d59c1be93f5b473b305a9194932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c1e9b7a2177d59c1be93f5b473b305a9194932"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
