)]}'
{"id":"openstack%2Frequirements~874673","triplet_id":"openstack%2Frequirements~master~Idae309d6ad7d4f0556c9d027b5980846b350d9a9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idae309d6ad7d4f0556c9d027b5980846b350d9a9","subject":"update constraint for oslo.rootwrap to new release 7.0.1","status":"MERGED","created":"2023-02-21 17:31:09.000000000","updated":"2023-02-22 22:46:31.000000000","submitted":"2023-02-22 22:45:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874673-new-release","meta_rev_id":"02d57c25e12eb3e4e01b50c7521589cf5d68ccc2","_number":874673,"virtual_id_number":874673,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 22:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-21 19:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-21 17:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-21 19:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 17:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 17:47:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-21 19:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 19:38:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cf53656977d14dd99759100c1f4ce3e573a8919e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-21 17:31:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86dd4bb9fbd40228cef9645f809acc6feb89ea80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:44:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50c8d252e2440e9abccb26b70c70d9e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/660ae491bfb44147b3b0a0e997df73a5 : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"41e33be129c8528e6d9c8ca8fe48e7254fe9a29d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 17:47:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b513461b7d234ebcbb88d92043da964afcc9ce1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 18:02:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f15dda84dd689c470bd2cd46b1b0f171c93b420e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7282b454fe4867b668435c3441d1e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbb73cdf7ee4f8ead8c5b431f930c16 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f149f6922f26402aa2d90ffc2267730d : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29247ed23c504d42be86ddaad4de9019 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8bdea420e7a4fe9a8ad53a3b4561c85 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0fb7485b8a634fa2b39ba921fa1f46fd : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a93c485b417f4282ab6f7f826e086c64 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c65317b58a074bce9010b1645ad81209 : SUCCESS in 9m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b757173705b4bb8839ec5f4ea9dcc0f : SUCCESS in 12m 35s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/36533feae86f4832b2895677732eff16 : SUCCESS in 7m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e6267d6fe07b4bb8816a1d72fc8218ae : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c23de58bf3994a4aa6cf5cd910a4c334 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4d767dc2b0744fd5a9ff88e822a2f768 : SUCCESS in 10m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e5cd68d3e6eb4279b2f11f2fce3050de : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/caf93940f27440ce8a86284fc65bccb2 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/91a9560ce4c446a9b8fc60b914b79baa : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a420ab8edaff4307b563f9553ea5d070 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eba9446a61d74fd481fbf554a68f3b37 : SUCCESS in 6m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/59783924913043949ef8b0ada0274257 : SUCCESS in 9m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b08020d63e49411d9b8bb1093285eccd : SUCCESS in 11m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9b8d21f67d794e1ea1065d02109a4db6 : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5333a993e2fb4060a45858a3fdae7c0d : SUCCESS in 10m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d95b91a604e24ef198ed319596f742f5 : SUCCESS in 3m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3a41ea50e9f74d6e83751acf6b025748 : SUCCESS in 10m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/874dc607a17945b29fa34ddabf817dee : SUCCESS in 23m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7abe63e766a04faeb0b1c9d51273ff8e : SUCCESS in 13m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/afea6d8a68114ce9b79cddd86d705f4c : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a86c84628a34d89b82953f60c3020ec : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/152f1e57eaa84895a0d3a7b7f5bd6145 : SUCCESS in 23m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fed1214ea51d45cb90477359ee1d275f : SUCCESS in 4m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/70dce2c7fb204e9fb2b7dbd362a74873 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3448d44468a54a748326ee394d6d88b2 : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8a39341ca99c480ead16cc39c5c152c1 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57a73fa5ba974acfabc74249ea08f5eb : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e3b8efc8ca94e5b84825fe299ac3703 : SUCCESS in 1h 24m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ed4dbf24456df4eb53fb170f25301da2213e95e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 19:38:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46d5c887079033c2648f4f3f62590106e8f51667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:39:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb5e554e07c834f2d9fb3bb31668f979436e7204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:42:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f96d968ce9f4053b944c5ff5c4d852c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f431da673a441a2bd1155c5e8984311 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/801c87c63bcb4a7c98fd2e9c24e44842 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d22d4780efff46ea8b675918784c3c80 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e87ead287e85436ca61a9b46bd06ac86 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55d18b3e30bc42b991ef722b4b96ba2d : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d29521c02a44f57b939e5362ee5ff86 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8942d34cc6f47e4af759a39fbaa1072 : SUCCESS in 7m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b033a3b4e5624cd89fbb03b80ba776ba : SUCCESS in 12m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8d5a8836297d40439ff2e62b6c716d83 : SUCCESS in 10m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bbb00e3a3cb94de4ba336ae6d8b21293 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c4f3d6acbc7a4c188ce7b93f269ff8b4 : SUCCESS in 6m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/56aef41410b64ffb818ac6cdba5e224d : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0993c1dcb3c14fdaa0c46a8ec0f0f8d3 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5a8fd1d6e334b8eb0c21db370065b85 : SUCCESS in 27m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9abcee3551d54d38903a4fff8428fc62 : SUCCESS in 7m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/28826461f8c1499fb9a75ab90b5b6069 : SUCCESS in 27m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9bd11529350c4458b94030b33c368592 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/19f5481b899f4aa3b851666f8cb7d002 : SUCCESS in 8m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/35346e5bbb3246f1a3865a356a05c684 : SUCCESS in 4m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/db5e1c0f89ec44779d0fc3b7c136a570 : SUCCESS in 23m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f890f3fb42bf489dbeb3162b019a3d1e : SUCCESS in 11m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f049cd6ad4f1458c91503d78e5705ded : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1b7ef41fc654f3fa8ba010aa0e211be : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cdae4d5f7bc4f208979d75206920eb1 : SUCCESS in 24m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/364d92890b994165aea60aa576976914 : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e1fa428605364b7392fdb88876612743 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/504fccb9578443f08c68c6aa3b043a0a : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/18dadc76acef49be90850c2914bf24a9 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdc231260d8e47b5a5710839107c130e : SUCCESS in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59c74e21d1a44bf5909cfb4a4cea3086 : FAILURE in 1h 08m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b526460c97569233290ac892338e9f7609253ae9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 21:55:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"87625152c5a1a0860e4b007d2cff30abed9b06e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:01:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e55ec3b7734147088b2d4a3b3a826569\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a305175b3e124b93949bc0123234a951 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5ac79c43e4a6916e02e9262275802ec3d3210024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 23:37:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/494b7c097ab0495392cd0e429abcc0a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed0fba7ce514713823ddec945381803 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c042c09cf0ab4ed99e663e795c1e4f77 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db2f631a67194593a067b7524b37f421 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf2e433a20614503820cb293763be875 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0a888ec3c774a36a01ee55429e5191d : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec3f35c1076647ab83aaf77de7d537e7 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/388defba5670450a83ede6edbe5d6bd1 : SUCCESS in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/854f5d779c8845cd9f55fa4be56b94c2 : SUCCESS in 14m 51s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1948378a710c48ba83c7a5d3bd217adb : SUCCESS in 8m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/32c9fcba15314786bcc62831cf054743 : SUCCESS in 5m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/91cba2aaa0ca40b38a354ea8268be8c5 : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8cc92ea635ea4f1bbf47cb67f548c80b : SUCCESS in 10m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fa491f1d54bc446ca9f5e35d356c83a6 : POST_FAILURE in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/869111cf770c4bb78cb43fa86da6fc10 : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/14b4ed49bbbb49e9b2c964eb0a8221f6 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/57944c9c7f7248e786f77f7c9960cb64 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/443df29cff4e407bb6c3e70b4481f562 : SUCCESS in 7m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/aa4b7be8637141f5afc8bc46ce13bc91 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/802c8e50747445268e6bafd599f0c245 : SUCCESS in 10m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7faf3c5432364c108c9533a23346d778 : SUCCESS in 4m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/416a7dcb8ba24af6a536fa442405e753 : SUCCESS in 8m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/12521b87ae134185b1187ae9a8443af8 : SUCCESS in 4m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/55ee9bc7080e42329c9a1b775456a84f : SUCCESS in 11m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/827d975b43d745deabad63e6895b76d8 : SUCCESS in 30m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ef7b866d18804366822a9a438a27e7d8 : SUCCESS in 11m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/db7e6bcf804644ab9cb090a5376f5603 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0441c360a890412d8621bafdd51eb7f2 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cc66d907a1a4bd5ada4d5221277cd8f : SUCCESS in 27m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6f1666a957ea41f5bb45bfb57944800c : SUCCESS in 4m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7ebf40e2bb9f434289be4b2222fedbe8 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90c85c3f31a44dc0899b3b6895f98b64 : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/43b75ecb6033495686ba264139c6c9a7 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94bb5fc6d6ab48ff9143f1b4e5d998f6 : SUCCESS in 53m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d31593f0d76543509ecfb73921b432c7 : SUCCESS in 1h 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cdd0994af0eaf87e05e824b43978d33da8173572","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 16:02:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"351f213dddf59bf3ea2d9c005212d8ad616e0b46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:09:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2d14668ce984ee7b3b3536eec5ec0bc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b421a4e8cf0941b1913f6a348f854345 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1ab02c8d2f273718a4c95315cce679a77c418f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:39:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26bc33cfd596466d8ff3a32c3571fe3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66625371b30f4421bcf21021f11f571a : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ec19b890f449d9b868847244d3b30f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c81c2df56ca4dc1878866cbba02f484 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4dcf30409204683a9dfe4406ecd2563 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08d3260ac5b343d7aefccc3ee1e1606d : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/825df42e34c9415a96c7bae85afc355b : SUCCESS in 6m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ec106c7c3f1c4e168aad828905d65f53 : SUCCESS in 7m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4eb31a99ac744512beb198505242debe : SUCCESS in 13m 53s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8d2dd750b2b24785b9184133a7ffae6d : SUCCESS in 6m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c08665482dae43b6b13334473fb99594 : SUCCESS in 4m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ace514a96a9f418499b1adee273bbabb : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6919dc0eae2b4ca4a78dabb1665926bb : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6c3b5cdba4634484b6cdbb0e9ebe8254 : SUCCESS in 5m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/84183f8c8f9847639ad108bb317ef76b : SUCCESS in 5m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b2ce1455526c42bdb02f72f474824996 : SUCCESS in 7m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2e2aa10f856640e5a655f6a9906e4897 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49292e79307f4577addb09ed02c9fd66 : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ecea830caae04f55937307844b1fad5f : SUCCESS in 9m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/17eb37f864b747e9a6fecffeffa6846c : SUCCESS in 10m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b4be4d625cf443e3b68004ea2f6c9473 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c7e67de836af4878ab2466031ab725b9 : SUCCESS in 9m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4a6a4bbe1e1e422091fc774c84e1ce78 : SUCCESS in 5m 03s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/575095fec05246eca5dad8a53f093500 : SUCCESS in 10m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5d86c7d23f6f4d1780aabe10ab9f80bc : SUCCESS in 32m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5adbdf57d2c8474a89250db26a06c89d : SUCCESS in 21m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ea35a8eba9564cf8b02a82bb2f341d01 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c39f2bbc8ca48b8b7d13773aca702ee : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22c099a86df047cb96927dfe1d2e063f : SUCCESS in 31m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/333244fe00564f479066b9d06077cbd9 : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5ba6ea97270841d99946ac185bed12f9 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf9732c7b4554b10bb4ce26e93a14333 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/21bcd9aa4a224653bd1ddd17970813f6 : SUCCESS in 17m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfb7ab8ff34c4f97911f3a7c0063c4ed : SUCCESS in 1h 05m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/337ff2a310ec492487a50cd65e7d6d27 : SUCCESS in 1h 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"17483323372ea954d4f79bc53d57ca8547c55d72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:40:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf7e3371f5851a49dba7b85c023f4c4010f42061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:45:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac5f8eab5c44bea8a51e0bf2f38f0d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc34e925f00e49558f0bde8e0f476e49 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8add2087768d4816bcc5311e17d2b113 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3991345aefcc44e4a0232aaee571bf5c : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d1083265d404e17a672c3c26d1027f7 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/190de29c32c5463abc611943bfdaebf7 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00390865d7794fc497741501e1dd7e55 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e1e850b46b2743c182e44d6b3a4d863a : SUCCESS in 10m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/39f7b1c5a61c4d769406bd541bfe138e : SUCCESS in 14m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/300e2468bfb14c8791104438ef33a86e : SUCCESS in 11m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6f1a2bca4d654a0090e6cad45435a3ad : SUCCESS in 6m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3fc686e20138441ab13b3fbf36b23b24 : SUCCESS in 7m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e51d3df0e76a4fb982b83d242ce7c5e5 : SUCCESS in 10m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/704d38f549234a74a5831a527dfaec19 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a8f66acc2624a8bb2d9b2e01f0fb628 : SUCCESS in 10m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d5f79372f3d14a99b57c19b38a9815a3 : SUCCESS in 7m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/93522426be8a4405addb1d7ec3cdbacf : SUCCESS in 13m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8153425473864f57974262d2342ced5d : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6c009bf75d664073a687eb5262180673 : SUCCESS in 12m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a1948ee5520c456e96bd4c65ff790039 : SUCCESS in 4m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c89f7805aa944a5bb972ec19a41e2e3a : SUCCESS in 26m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e66f9ed7b15d4575aab6fd8f2d5c399a : SUCCESS in 11m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7e88ee18a30e4eb4bbfde72539b0eb73 : SUCCESS in 7m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/075f99966c224b5ea9318576d3c8fd9e : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fa19e89ba2649d2b27d83a3e0b32d66 : SUCCESS in 24m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/df739641bb5e4f90a878ccfc875935e1 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9ced73794a144d5f9f3ec3586548b9b8 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b8dc743f05d425099f4708c6d6235e3 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/42998af2a22b423d964a5a0fae8cc0d6 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d259f21aea6421cb9bf315f6d1c6c2e : SUCCESS in 43m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31ee17f6c80b438f95f90222809294df : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9374d0a9859eb5096db8172c731f7b1bd3714538","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:45:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02d57c25e12eb3e4e01b50c7521589cf5d68ccc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c94c5cb5644ed092fa1f62c272358d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d05577c850f483c917db69ca89a883c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8e1e3d0b25c8f4213b3b465f9714397b2580314","revisions":{"e8e1e3d0b25c8f4213b3b465f9714397b2580314":{"kind":"REWORK","_number":1,"created":"2023-02-21 17:31:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/874673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/874673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/874673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/874673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/874673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/874673/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:31:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:31:08.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.0.1","message":"update constraint for oslo.rootwrap to new release 7.0.1\n\nmeta: version: 7.0.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4a971a18d490b465b7aafad9c8b8f10e0bc96ebe\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idae309d6ad7d4f0556c9d027b5980846b350d9a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8e1e3d0b25c8f4213b3b465f9714397b2580314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8e1e3d0b25c8f4213b3b465f9714397b2580314"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
