)]}'
{"id":"openstack%2Frequirements~874679","triplet_id":"openstack%2Frequirements~master~I19fa3e8e807806ca0ce7d53f7d986a70e7e54546","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I19fa3e8e807806ca0ce7d53f7d986a70e7e54546","subject":"update constraint for oslo.upgradecheck to new release 2.1.1","status":"MERGED","created":"2023-02-21 17:48:05.000000000","updated":"2023-02-22 22:47:15.000000000","submitted":"2023-02-22 22:45:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874679-new-release","meta_rev_id":"715dfd8448ce5add074e0548efed973e5ff42fff","_number":874679,"virtual_id_number":874679,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 22:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-22 16:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-21 17:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-22 16:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 17:50:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-21 18:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:17:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 21:56:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-22 16:02:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8da0cabaf4b20cb90cd8ed3b65d327ee29205b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-21 17:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b566504675ebcff4937dbe0b9e74b25a004eea9c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 17:50:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2902fbf7eb685166075043adef2ff6e811b91cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:00:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2b836bf6ae4523886ac6e7d895c94f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f49871a1d48840f6b8711c63456da73f : FAILURE in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2d754ca59179a2f74eea72279745d54a35ce616b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 18:04:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e880215a4f48c5e3e19e3fd04429daf17114d8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:17:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d8c13fc234f4dcea27fb813a296c30c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84faada834ee40afb1abaa1c79991663 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e157d30c4248cf9bb64ce08ed3d038 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/289e03462f394997a3228716700a40c4 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0af906a2b4d40ce8ffe75cbb0d7e9f0 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6eee0debaf2c4bb9be9abb1feb41f6b0 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cbba7833cc84facb90f233a591b418e : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bac0deaad3414a8ab279d99d71002888 : SUCCESS in 7m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a22b345afe344277ab3e217e8a76dba9 : SUCCESS in 13m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1d3e8d65eb424f7f924194d9fd19ab29 : SUCCESS in 8m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4ed7f4ff6e2b431e94338d103d230284 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4eb71d45ac6844b0aba23bcf94825ee7 : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5a525b3e6a7c490aa21182cdb4a987ab : SUCCESS in 10m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dec9256656f44e8e8c2122ce437ed8b4 : SUCCESS in 4m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f5824da06f8a45558dc44d0784b3b8a5 : SUCCESS in 6m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/665c7931c5984d8cb2cfe43881509953 : FAILURE in 3m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/10f84a81989d49a3bf9029ba5b6a1e23 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb83026120794c0196cb6ef63b0c2892 : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/518e92274fe44191aa9c1af1bd0fd4d1 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0ca4c99a09e54cbeb1c5999c10ffd03a : SUCCESS in 13m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d5d72c326afb41148e4f1a265b8b5e95 : SUCCESS in 3m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/99561575da0242a3959aab60f875bdb1 : SUCCESS in 8m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4fe6911e8d0e412183034d47804603f8 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fa2c61b67e7a402db5cc15e4230cc182 : SUCCESS in 16m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4340524c1d694e61b0713840da92b525 : SUCCESS in 29m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/36dcb4f8a5c549a2be537ba3bed2d36e : SUCCESS in 13m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/db79003b5ce04a64bb58eb23b5d11939 : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9fbe9a38c094999b6ac483a5230992c : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73515dd7fb884258a66aa206d668e043 : SUCCESS in 20m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ad359e68a7cc4c7bb762e524748919fd : SUCCESS in 7m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dd4f5e5efbed431184957fa78afea436 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00617c0f0bf04660b4747dcf6227f130 : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6ed1be6e11ab4869a8c2861d157f871f : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/871dd62db4104c38a0a32f5cae6103b4 : SUCCESS in 51m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22e53d7ced0c4ff4928dff9c9ebaf889 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c3411f12179116b6a48815e5395af228ff5cd2e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 21:56:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c6ecd9c29010e34721af9d4f8e0bcb5a12373d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:23:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f4dd95f85c6432dab38a9170e9e1101\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e45997b164494fa4a65a3a0aef671a5c : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"13e066e3a874942d446850cde5b30d9695e2d9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 23:13:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4db5f491a2244a5180dab30b20a11faa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04167989d6564b488a7d567c4e551cb0 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fccff9714c440d92c7f06a4b64a4e1 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a061babe212b4faf920cdede43ce6e02 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6a3072832fe4149b3c7058928042051 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0013bcd03174cbfba45f41c3ad00fcb : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a5f967300b74fb88e7156b98e6f4f5f : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/106a8c783cb94361be03dc5490624250 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2cd116fcd67f4f3592ac62f2d0dd28c7 : SUCCESS in 13m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/91cc424d6409466ba06eda87c70e2556 : SUCCESS in 8m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c28c4ef41d794456b9be22a730afdc32 : SUCCESS in 5m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5d326f756397413982e06569b5cefc43 : SUCCESS in 6m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7df8e01690d74e1fb74615078f272450 : SUCCESS in 10m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2597027bea364b3aa0167af49b2dc571 : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/171b283be79f41829b4c80408041b21e : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/66d6f54afa2f42ef9b153d0a9a3ee984 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5e6c66dedd3d47a984a98e14e46acc63 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e161afb560342218fa3812e74d013a4 : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b88a27cc795e4dfd9e5fc6a0296ac757 : SUCCESS in 9m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/049ea907b9164996b9f43de9c25cae75 : SUCCESS in 12m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5394c6853779467fbfd0b73cc8e56b9a : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e753a5a0b55c4b87a422ec977b910a9b : SUCCESS in 9m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d765e6fa9ae04dd58852b598b441ba41 : SUCCESS in 3m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8a3ff24ba2cd4eef8ce01d51274612e0 : SUCCESS in 11m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/03242d090fb54d579f0846cf2eebd52c : SUCCESS in 26m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4c7aef48e57e45948329de2297ba26bd : SUCCESS in 12m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/985f9baf193f448daf96839e2056caaf : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e6f6d3471114487a5b3772681dd44ef : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b861ca77c7424c40a214b1dd9ea76247 : SUCCESS in 21m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/da5f499d4a344a0aa5e3516f6cbc1f87 : SUCCESS in 5m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/152d8d0bfbaf4319bb16c7c8136344c1 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/793958fa220a4b4b8cb35974744d76a2 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/30fbda9285ed4fd08dc336154791e97a : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddeff2df42974c61936eff1f868071d0 : SUCCESS in 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe99822401794f03819425ac510ed00c : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"443d58846a98c5fc23c71ddb618f3615b864b6ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 16:02:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8acb23aa53b260085c91861e65e8b5c36e31273e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:09:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9382de737c17452093198a1388e74e9e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c01d3379d9614e61b2ed00f5a2bfee3d : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1422f7327a9db9049e5c504cf7b9ac2c722dc3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ffafc92e394b50b93208f294bfaa40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec788229f6649ad90652000624e686c : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c82643ecf94467d9cc45362f2f9731f : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add8d4df02ee4c5fbf59b7d18850f6c6 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c790c49d2842478397a4dd456e18e8b2 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89c83f16b46549eeae605543c9826805 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef0812da8c2f4745bda28c63bd44f40f : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a9b80a1a4511431ea492521169d71cae : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/002c5c6620764e488e8510ea1fcdd4c9 : SUCCESS in 13m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5304ee3d7f074f2cb7b63d7ea8ec314a : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f67dfb71e73c4e9d9350a4e9fa115194 : SUCCESS in 7m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f71ecbcfbb1e4fb3a085ea6fdf2971bd : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e99506b5076541389ef798b62c15bb3e : SUCCESS in 11m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/174fcf689ff948cf968d60e9a04c8b42 : SUCCESS in 4m 27s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1904fc4e40c94d1199d3583da47e8d4a : SUCCESS in 4m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/895127b9c96a48f9a4dfa690a0cf30dd : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6c14c359d4124b2fa3f03084c4c08dee : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87ee7669a03f444d85c2ed11bd59f63c : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e6d09745727d48e0872b9167ffbe11d8 : SUCCESS in 9m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aead8b79c9e046e48bb85bb8c9df6ac3 : SUCCESS in 20m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/71f68a631c114f30b375c58fc55cbd04 : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6d40ecac96844774a4021d7f6c989d02 : SUCCESS in 10m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0a778a67376b4130b914aa8f3540d56d : SUCCESS in 3m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/097bb4ac023440cfb3287cba5445e39d : SUCCESS in 11m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/03ff559f771d48c5bd6c427644d2803b : SUCCESS in 31m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a69bdd45a4484dc8a5a82ea3ead9057b : SUCCESS in 11m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0578b500a6fd4fdb8a34236ddeec959b : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5983bb8e803244169525afeb8814bb6f : SUCCESS in 6m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ace087b77c9840f2a1a3f75fb3449915 : SUCCESS in 21m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c34a1cd45ad14387a1425fb69c91009c : SUCCESS in 4m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9f9d32355fc343778e11de1195750755 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e54af6cff08a4bbfa0c3b431c62cb89d : SUCCESS in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e2177aec0ab4c93bc3754c9e514d023 : SUCCESS in 18m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd3a37dd99e842268ce3ca40f00586b6 : SUCCESS in 54m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1bf3c7e5d9b415fa15363cb8d8ecd8d : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"699d6e9e150bdd32af54b63b0d7e0d57b031d93b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:43:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e62207558bab7aa91fa8b8b77b124ed91075d41e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:45:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adcde871eef84eceb88c28348e815677\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c953053e0e247339c692bd5afff4569 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0870f89eb4e9411e9ae40f028db424c6 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d87d026d3d31425ab0e70918931b3a09 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5adf2bcae994af1bfbe8f25faa8ba54 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dcab645a70934acfb6b70bf6e5a9a3b3 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ef929b9419f4d72b77ff365bfdc2dc2 : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82dcc6c097ee4f76ae2ffd5515bd8fb0 : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b5abeaded6546f0b61a6137cc3022a3 : SUCCESS in 15m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/42d0e4d8a09d4e2a98a82597d75ee4d0 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ee319834e681477a85fa51cceb39d934 : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/82a13f8c542b43cc87af8d23469974b3 : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/532b7985f0c241ba93bec59171f9d17e : SUCCESS in 9m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a4fd7d5c99af42a7ae5399a4b95c8e19 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4591d9b9c2a244d58619696cb18ee2cb : SUCCESS in 8m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f02126cf090c48bfb4a4e90cf076e13d : SUCCESS in 9m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1fae00a95ec64e1fb64ec75337b33703 : SUCCESS in 13m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9d7cf0738add44a7b98f524e1871bdb5 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/250f0aff53ba43b28d73a7d522461859 : SUCCESS in 9m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2f180944c593459d951988b1445f549c : SUCCESS in 5m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/151e6336d3f745279ec62d7fe040ee4a : SUCCESS in 29m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dfe5274df86f40a59cf2238728211a4a : SUCCESS in 12m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/513b574526534179a1c50ccc66a3572e : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d867944de0a42a7aa1234eeb4cb21cd : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/449495472c8442c4889247c0059130d1 : SUCCESS in 21m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e534003df130496cb5a1f850e282ae62 : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/43742939b6ab4d30b2e59940cdfca6e0 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f193276655204a56a8319b99ac4a0ef9 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f38c723bd10343398e92e593ac7cefa2 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/171e37b79b254a54aac2263cef5b44b1 : SUCCESS in 50m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91bf3915fa5b4883a074f00a8012ec15 : SUCCESS in 1h 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e605f6660ad3069184ece60d6dd01a7c867ca974","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:45:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"715dfd8448ce5add074e0548efed973e5ff42fff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:47:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5634ff1bd784f689f3d48a1a7b148d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2d6c92d38642199f1dd9e85dac6ea4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3138f1ebe3c22b6495aaede6d0251774e4fb4d21","revisions":{"3138f1ebe3c22b6495aaede6d0251774e4fb4d21":{"kind":"REWORK","_number":1,"created":"2023-02-21 17:48:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/874679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/874679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/874679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/874679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/874679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/874679/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:48:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:48:03.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.1.1","message":"update constraint for oslo.upgradecheck to new release 2.1.1\n\nmeta: version: 2.1.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4a971a18d490b465b7aafad9c8b8f10e0bc96ebe\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I19fa3e8e807806ca0ce7d53f7d986a70e7e54546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3138f1ebe3c22b6495aaede6d0251774e4fb4d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3138f1ebe3c22b6495aaede6d0251774e4fb4d21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
