)]}'
{"id":"openstack%2Frequirements~874683","triplet_id":"openstack%2Frequirements~master~Ie8f8733b32c4c2c4f266e1025f75f23293843d29","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie8f8733b32c4c2c4f266e1025f75f23293843d29","subject":"update constraint for oslo.policy to new release 4.1.1","status":"MERGED","created":"2023-02-21 17:58:18.000000000","updated":"2023-02-22 00:19:23.000000000","submitted":"2023-02-22 00:18:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874683-new-release","meta_rev_id":"25a14e1e6b3be87ca22b504b5d83d9188c093d26","_number":874683,"virtual_id_number":874683,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 00:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-21 20:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-21 21:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-21 17:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-02-21 21:53:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 17:59:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-21 18:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:31:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 20:21:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-02-21 21:53:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1da6ac763ab06531bf7dae8b8f7af3c58de0772c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-21 17:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b96f1f48f34b07d02ee7c8d397d9839c0a9a73e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 17:59:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"096fc7de6c1026f621635a68acaec5a9af950fc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:03:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e7127bc627496fbd4dd2a934b95baf\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5b490825738d41a285388d0f01ef2c02 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c67781f0256b5fcd24c9d7d153ea59f33b1123f5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-21 18:04:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09315f4a98f1b25177a3bc406b86310226589c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:31:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fca87df0fd034952a314794e636aea5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735f394ce79b4a5e92a7d9c589516a8f : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5e5ef4adb340fd851a05add8fc9328 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbacd6175ab24415a496068e82fcc3bd : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44d6bc21de534b378856150046b6a754 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94e10afc9cbc4e69a8bc617149f02c52 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76fc4e13ec1a4df8ae4d5e41632cacb9 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6da0a9e3360f4901ade799f523f469b9 : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0a8865ab6327496fbe81e9f9d427442b : SUCCESS in 11m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4c185d1a63ca4e72a9ff12da47f90237 : SUCCESS in 7m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/767d2ffa9b8244dca003aba2b99ef2fd : SUCCESS in 5m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/91a97ef95b9a4389979cb585f4a4ca0b : SUCCESS in 6m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3ac6528db8e54216b4ccf3532016960e : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/36d77b9d93d746f7a877b75d3a09f0e9 : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8fd75422668a4730822a9ee8601a196c : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1ddde7544415460792d585e5f633a652 : SUCCESS in 8m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0df3f36d18f34c959cbdd763cb7017c4 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729a227c5014494e8ec680e1723fdb82 : SUCCESS in 9m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d36ff8d0949246b3a6a3a1e173b6074f : SUCCESS in 7m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dea6d236607a4cd1a744946c1d78c648 : SUCCESS in 10m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1276f59c6a364b818016f62d6dc00f20 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f258ef67d14c4662a43975e0fc9231c1 : SUCCESS in 8m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4bcdb180e45e41b19774968480fa76e1 : SUCCESS in 4m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7d4079b4cf90493aa8e936460b8543f2 : SUCCESS in 16m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f14ab0ac3b714fcabe7db1af6356f4bd : SUCCESS in 26m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0ebb16469ebc49828dcf27e46b53131e : SUCCESS in 12m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c73fd23e8f974cf4846e2b7b7d860911 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2ac57c22ae4411fa81ee82c0e1a05fa : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/290af8751ea94175bc683a9279f925c2 : SUCCESS in 25m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c3a8f6fc4d8c49a79bb04719f12cbb69 : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c3bc0f910d024902b91f3bde71714306 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b287b83e346f4ca48f252224e0bbea02 : SUCCESS in 5m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa701eadc5614db9a78520ede1af99d9 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/540ca6e4a5e840b186741ec90b5c9806 : SUCCESS in 48m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b05eeb16e4d4da796f4ac9fc1853ad5 : FAILURE in 1h 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d3f94b8d2ec403ae9454238d6b1cd53900887677","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-21 20:21:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62450301254b20e68d78c4ea9be0fb8edcabbdf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:36:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4741cd8f83b4011bfd43e6706dbe840\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ff2459b24aaf45428da7cf21eb266521 : FAILURE in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9ae83f8ef28ba9a2d30ba823ea0e7237cba6249d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9bcff7b337543cd974b6cf4e37b8aed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20d06cbcb6046e882afce1be0b2fe68 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38c6dff98b1b48779d8629d1e57f0923 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a060a13bfec49779acdd5a74beee1b4 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccaff00891b84c9e9dc0553d8c93657c : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a9841516c584b4faf303581d6321587 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8e8121d1e0c4e7e98c0ea274c0b88cd : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/92bfb20af4d24cd3a3d204dafb2fc0d9 : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b2b63306f89b4427873387ebc3b43344 : SUCCESS in 10m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/33dceb3b186c480ca3109174bd52bc56 : SUCCESS in 6m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/aab0d2fc7f404a0ea0fa8d59da680726 : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f35fc90d333144bcba21934ebe90211c : SUCCESS in 5m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/196cfd7b184843a499f073105a5a44d5 : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f662cd7e33764e05853ab2d936f5dc76 : SUCCESS in 4m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e5b52af59b0a40b28955e05e38642d4f : SUCCESS in 5m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/683c6da3719e41c09c919ffa4d891b5a : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a14bd81e936f498290aace0c18178fa3 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bd77c0587294e6b91ee9eb71637c0d9 : SUCCESS in 17m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e8c710cbf3c24a1bbf5ac0de336eca58 : SUCCESS in 12m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/71962322396c48a99dbc5941225b773c : SUCCESS in 9m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/acc5570801cc49a59465b9b34ffec433 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/73340f7d789c4342a1540b0133f415b7 : SUCCESS in 12m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/52885038bc5d409480e24123b10e09be : SUCCESS in 4m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c7516a47d89c4843bd974effcc826285 : SUCCESS in 9m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b856c7c54d714da7a4c965dd3b8d796a : SUCCESS in 21m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e5947eacb1f24859b09247e9be853571 : SUCCESS in 11m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2dcdf657034d4af3bfe1725b60ec087e : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9376ead1316e4e72afcefdbb8f760541 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98c2b92434784066acf92d553727c801 : SUCCESS in 25m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f2a1dd3723cb4d2a8f1ee3c375b2ad92 : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d29269e900df4a58997dc7a08ce08ab3 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b3255ce2f05461c9a5bd9f3f6e564f7 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09e81c5d458d471fbc03397273ec4c02 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a69fb5e37345430699e2856cd7ab4615 : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6fe87014827497fb2d4042484d7d216 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"862496ebcf7db8bf45af02d548f40e6b60b86443","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-21 21:53:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abd254178215a0f10cb635472a68a7d4ff820ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 21:54:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3c1e9d9bd8bdfac14cdecb0601eb27d5477d5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:18:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e687b23d304488182502210b1973a88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475eb027eabc4cfda52d8fbe3b404c76 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7209358cfcd74596bd02cbd9ff10fac2 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a5e9329384e42ac988ae4c698e1af36 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05c1cd526b5a48d382fa3e9d996fa24c : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/798e2fd9b6bf4c5a891da7ba0bb8fc48 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71f7587994824c9693ad7f610a091f0f : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08654a022ffb49e3a9eb1935d4892644 : SUCCESS in 7m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1fa50b14c1c84f099317fa2eb5724b5d : SUCCESS in 13m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3384f98757f542a59805e6bf09a15481 : SUCCESS in 12m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d05393be06a24db4b0ebe19369591a3b : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/98ea5a2a4cbc442992ceebaa16637d79 : SUCCESS in 4m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/61e20cf60aa94d28b334499aa1c88f11 : SUCCESS in 7m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/76bd4a99f61c49d092b5b32a4602b1a2 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b17c174b54df46f9afdb97168ea62636 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2f11b07b7eab4e2a9b1d97ef9b4a2440 : SUCCESS in 9m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5a852eed87334053982339de5ddea32f : SUCCESS in 9m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6d270a83143348309f49c7abd29a2c80 : SUCCESS in 2m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d073ae9475dc4903aecd01ccd3de47e9 : SUCCESS in 8m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d9658c7f716e4283b8e7680238f19167 : SUCCESS in 5m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/62faadb85dac4a4ab658ef2734fdebb0 : SUCCESS in 22m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1cd941de834049c59c4379fae48b0239 : SUCCESS in 11m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bdda49c1805a4954baf7009b0f0d34cc : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5a943ee3d1e4e858a63fff100d41b61 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eff0d3004aba4e9b8b098efbcb7d606b : SUCCESS in 20m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6a5479ea802f4002b353fca3d7c684f0 : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f1038a62b5104135b869b93643a749b6 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/edc626117ba14fb6b88dd26e5b592fef : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fda191a903574c5e828610533a47cd87 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e3ea5961b2a44afbaf4ed9dba170fad : SUCCESS in 1h 02m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ab301c9f2f54cb9bdf20c99156878b0 : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8b614dbcc6c8962b8e06d5bbdf06f488699fa3eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:18:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25a14e1e6b3be87ca22b504b5d83d9188c093d26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca4c7dd3e1b74e0eb9719dcb89a5f9a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd0cc77191f48b88efa50d6eaabcb86 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27129b66465e3b180a3f674fc5cabf1c01e48de5","revisions":{"27129b66465e3b180a3f674fc5cabf1c01e48de5":{"kind":"REWORK","_number":1,"created":"2023-02-21 17:58:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/874683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/874683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/874683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/874683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/874683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/874683/1"}}},"commit":{"parents":[{"commit":"e7b8174b920d4f0895ee967d2b145ed2eefc2af7","subject":"Merge \"update constraint for python-cloudkittyclient to new release 4.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b8174b920d4f0895ee967d2b145ed2eefc2af7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:58:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-21 17:58:17.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.1.1","message":"update constraint for oslo.policy to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4a971a18d490b465b7aafad9c8b8f10e0bc96ebe\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ie8f8733b32c4c2c4f266e1025f75f23293843d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27129b66465e3b180a3f674fc5cabf1c01e48de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27129b66465e3b180a3f674fc5cabf1c01e48de5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
