)]}'
{"id":"openstack%2Frequirements~874685","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ib80d28be0cb23c35821e9ba0217191d7d66a003c","project":"openstack/requirements","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-11-09 16:53:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib80d28be0cb23c35821e9ba0217191d7d66a003c","subject":"Fix setuptools upper-constraints for py3.9","status":"ABANDONED","created":"2023-02-21 18:03:41.000000000","updated":"2023-11-09 16:53:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5e79ec29e3b8aefaec2e1e4c6f91945ce570656","_number":874685,"virtual_id_number":874685,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-21 19:44:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 18:24:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42ced424fc8471944bfe8e010a8e83e8ed67f4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-02-21 18:03:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3353df1b1fcbffcbd3401da2ec11a943f2f3323","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:24:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ee599bd4d943a09fe0b156c48949bb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/333969530d15432baabaf7fb4130765d : FAILURE in 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9d10c0525985e3fd53a68220fd811e4e83aacdd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:44:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f42c0c0e1b804ca0a4a4a5d2b9df5d4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbee8393de247d781b826bacb6d49e6 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34032f6ff78a4cbbb15260bc74e06dc7 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/291d63dc76974c31baa789f0ccea43a6 : FAILURE in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31bf92844e7e4d4e94e7d9b379c8fecd : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1d35a03e5bf481bb47db305141cfdd5 : SUCCESS in 2m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/79e62ddd1e634aed91df6c1c8396d960 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6fcab36f9d4c4448bc92c2f4a95a874f : SUCCESS in 8m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/068abf53ac9c451c89d282d232deed46 : SUCCESS in 4m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/df615de446954b07af027a93a27a00cd : SUCCESS in 10m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/053e275e78cf4d9db00dc10ed7d049c5 : SUCCESS in 3m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/655f253a0f14481880f6beee4c30b99b : SUCCESS in 4m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6ead02682da246ce8b33c98ec8f88aec : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db3916ad5a334b5abaf07148f6dc0e7d : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31e6ea5c9c52429786adb1b41446d4d5 : SUCCESS in 7m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5586e4a60c9b41df87fd844c37435483 : SUCCESS in 10m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f4ad9a0091ec420f948bd9a4f9d72afa : SUCCESS in 3m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6ae473e8350b40feabd4e89fe088cf09 : SUCCESS in 17m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/41b85230cd554cecb70516e5f2cf1f81 : SUCCESS in 3m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6edb7d3bd5214653beb28bd3bb11a859 : SUCCESS in 24m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/368aefebad6f4ea498abcec6ed6a610f : SUCCESS in 10m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eb4b85a5228148b6957bd56d6d763b16 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ae7e5508cb741829da89944b46725c9 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0594c4c6630f4bb2b827459285a8f320 : SUCCESS in 18m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/71251a725744462bb0665fd550d00afe : SUCCESS in 5m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e5f5ced48ab74e31adb29fea768af938 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dccab467b15a4063ab607bcc980badfe : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f347253d9e6247f98812091802d6a9dd : SUCCESS in 21m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/902916236ee74353ba808f4066dd693c : SUCCESS in 46m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8663844cdf6f420fb99699ce258c075b : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b5e79ec29e3b8aefaec2e1e4c6f91945ce570656","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 16:53:44.000000000","message":"Abandoned\n\nI don\u0027t believ this is needed any longer as there was a later setuptools release that papers over the issue.\n\nPlease restore this change if it is needed, and help us understand why.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e50d60b1b6605465ffecd616bdca34c8aab2a1f2","revisions":{"e50d60b1b6605465ffecd616bdca34c8aab2a1f2":{"kind":"REWORK","_number":1,"created":"2023-02-21 18:03:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/85/874685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/874685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/874685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/874685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/874685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/874685/1"}}},"commit":{"parents":[{"commit":"707da594d13b7f1ed47800660aaa2a8a804689df","subject":"Use correct nodeset for requirement-checks job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707da594d13b7f1ed47800660aaa2a8a804689df"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 17:55:01.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 18:03:39.000000000","tz":60},"subject":"Fix setuptools upper-constraints for py3.9","message":"Fix setuptools upper-constraints for py3.9\n\nThis patch fix upper-constraints.txt as currently\nsome kolla builds are failing for yoga version [0].\n\nIn newer versions setuptools was pinned in kolla [1].\n\n[0] https://review.opendev.org/c/openstack/kolla/+/874684\n[1] https://review.opendev.org/c/openstack/kolla/+/873894\n\nChange-Id: Ib80d28be0cb23c35821e9ba0217191d7d66a003c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e50d60b1b6605465ffecd616bdca34c8aab2a1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e50d60b1b6605465ffecd616bdca34c8aab2a1f2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
