)]}'
{"id":"openstack%2Frequirements~874757","triplet_id":"openstack%2Frequirements~stable%2Fxena~I99c20cf35d6e832a8b341d431a58dff25118b4da","project":"openstack/requirements","branch":"stable/xena","hashtags":[],"change_id":"I99c20cf35d6e832a8b341d431a58dff25118b4da","subject":"[Xena-Only] Adding newly released tripleo-ansible into requirements.","status":"ABANDONED","created":"2023-02-22 10:18:25.000000000","updated":"2023-02-22 10:25:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"51174487737247b06318b1e1e3d8ce847c1a90cb","_number":874757,"virtual_id_number":874757,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 10:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"5cac46d84f9bf1a728df194965195a5b4a353426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2023-02-22 10:18:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc91d74e0f609472b6e5bd1b6879e4c76a390aa1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2023-02-22 10:25:24.000000000","message":"Abandoned\n\nNot reasonable on stable","accounts_in_message":[],"_revision_number":1},{"id":"51174487737247b06318b1e1e3d8ce847c1a90cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:25:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c9df27dd9344909b4d060d8d42cbbe\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a65704a043b942508b3557933549e40a : FAILURE in 5m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3c84d5063839bcd856de8fcb8b57576b788fdf5","revisions":{"e3c84d5063839bcd856de8fcb8b57576b788fdf5":{"kind":"REWORK","_number":1,"created":"2023-02-22 10:18:25.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/57/874757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/874757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/874757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/874757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/874757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/874757/1"}}},"commit":{"parents":[{"commit":"e1521660a99a56caa9623488b017b7afd7b47797","subject":"Merge \"update constraint for neutron to new release 19.5.0\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1521660a99a56caa9623488b017b7afd7b47797"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2023-01-03 09:54:29.000000000","tz":60},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2023-02-22 10:17:37.000000000","tz":60},"subject":"[Xena-Only] Adding newly released tripleo-ansible into requirements.","message":"[Xena-Only] Adding newly released tripleo-ansible into requirements.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I99c20cf35d6e832a8b341d431a58dff25118b4da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c84d5063839bcd856de8fcb8b57576b788fdf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c84d5063839bcd856de8fcb8b57576b788fdf5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
