)]}'
{"id":"openstack%2Frequirements~874965","triplet_id":"openstack%2Frequirements~master~If4a56c8393447ca92f44e3063feceafc97c8b5a1","project":"openstack/requirements","branch":"master","topic":"kuryr-voting","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-02-24 22:11:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4a56c8393447ca92f44e3063feceafc97c8b5a1","subject":"Revert \"Make cross-kuryr job temporarily non-voting\"","status":"MERGED","created":"2023-02-23 22:10:17.000000000","updated":"2023-02-24 22:12:46.000000000","submitted":"2023-02-24 22:11:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874965-kuryr-voting","meta_rev_id":"729fdef20deca2195f7646dd5a132e14922c5f52","_number":874965,"virtual_id_number":874965,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 22:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-23 22:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-24 15:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-02-24 15:54:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 23:21:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 15:54:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"e4352eca2a7f130edcf970e746ce405ca534298a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-23 22:10:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c106cc73a6aa6679519bdc07fd147c542c4929cc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-23 22:10:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b715ac3085d1e3b16abaf655ff0bb97b952d4f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 23:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ec1ca5ad11e4daab2df7045ae4cfb89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2372ab16a5a42fa9ad267735e2bc528 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1a12d69a95494ebe5bc1264cf3b22a : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef70bc91b5a64725b8f7c888b526faea : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8ad25976bdb410d8bb0801409d2999c : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa17492f23a6417bad1643716b2217d1 : SUCCESS in 3m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d586ae11d73b4e4bb59a5ca32d90f6b8 : SUCCESS in 3m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce131d348aa44fa682973a3cb33a18ce : SUCCESS in 47m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/054180f2afe04ed6a21e184a5183e217 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"40f8991be511309fb1ab4abe2905c1489f7ecbed","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-24 15:54:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23842fa666d56205eeb18018b85bc3310bd807a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:55:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ed366f205414c1e8fa88fffd22db433ebcf5bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:11:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a4f42956dd4257ae3c0724b6fc0f09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dec2c2cf82d4d998591594c9e7bc0a4 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3fd47ff22c4e28b91166474cb41200 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a301f6d087374c529debc705887ab984 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68a7727b2ae24596ba77d5eb6ccbe0d5 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfef561960554a30a35010d4e47e9a34 : SUCCESS in 3m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bf40f5dbc6c645eb8376c87f348be6b6 : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd2cd67ba03448e0a5e8b29f8d8f11bf : SUCCESS in 1h 00m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/235b6fbce41a499c9f6aab78ac905a5d : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b556572d559a28c384b7d455edde9afc15112ed7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:11:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"729fdef20deca2195f7646dd5a132e14922c5f52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:12:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118d4d0f3d43416190111b23bf7ccfb2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431fb0d56c29414e9c77389bb0bdccbe : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d97e436861920f1d6ca257b930c03cb712e2281","revisions":{"2d97e436861920f1d6ca257b930c03cb712e2281":{"kind":"REWORK","_number":1,"created":"2023-02-23 22:10:17.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/65/874965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/874965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/874965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/874965/1"}}},"commit":{"parents":[{"commit":"d25bfc90763a1790ff8d27749c0c6650bae86c63","subject":"Make cross-kuryr job temporarily non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25bfc90763a1790ff8d27749c0c6650bae86c63"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-23 19:07:16.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-23 22:10:08.000000000","tz":-360},"subject":"Revert \"Make cross-kuryr job temporarily non-voting\"","message":"Revert \"Make cross-kuryr job temporarily non-voting\"\n\nThis reverts commit d25bfc90763a1790ff8d27749c0c6650bae86c63.\n\nChange-Id: If4a56c8393447ca92f44e3063feceafc97c8b5a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d97e436861920f1d6ca257b930c03cb712e2281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d97e436861920f1d6ca257b930c03cb712e2281"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[]}
