)]}'
{"id":"openstack%2Frequirements~874993","triplet_id":"openstack%2Frequirements~master~I29ec510dab912e87cb348a521160ddbced0bda89","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I29ec510dab912e87cb348a521160ddbced0bda89","subject":"update constraint for python-solumclient to new release 3.7.0","status":"MERGED","created":"2023-02-24 09:40:50.000000000","updated":"2023-02-24 17:12:47.000000000","submitted":"2023-02-24 17:11:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874993-new-release","meta_rev_id":"a164f7076bdaa5851126bb448e2513432cba16fa","_number":874993,"virtual_id_number":874993,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 17:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-24 15:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-24 15:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 09:45:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-24 10:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 15:30:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"eb4f2df620f6641c149e82d1759079b876cf1321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-24 09:40:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fbbab60e7bb006aa88b854f3ce1555cdec14522","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:45:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3284ba0f0593471d824179488a6942bf\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1fddb055bd86401591fb9561b0886466 : FAILURE in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c516cc9005f7b01dd1f397f6005e94bc3e0900b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:57:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cfbfea4975a4a5383c3ddca0498ec73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14db37a81216428ca3b854d09f0a4419 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f702fc086894406958285b3404df639 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6df971975bbd4098abfb89eb6e74fe13 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/171c62a004c848e2822da4d2e801118e : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a918a3c5a0b94e5c94173fe1c4e0b23f : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05af4c64c42d4c99b197eff4efdc5ef4 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56df0a5c3a9b4525b9c829390bf2fdcc : SUCCESS in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60efa5e8436146f889e963476e64ba0b : SUCCESS in 11m 57s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9e1ffaaba19e457faace7a86367234c8 : SUCCESS in 9m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/83dcc7ec1c044a05a148f0f2de808778 : SUCCESS in 6m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/86b067de9c4f43fd9df017cfe9c1b2f8 : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/22d2a3b789924463bf5244e20acdafea : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/22ce5ac8e49f457d972a66bd30846dec : SUCCESS in 4m 36s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/de24d67d3f1b4cbf962ab29c8d4e118e : SUCCESS in 6m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/43b18ad95f8b493fad101e2008133dd7 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7dbd127505a440bb6a62d2f1681d45f : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9ea3265351f464c8aaf1a604bb2ff59 : SUCCESS in 8m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c5ab033a088144b8b7bc7e30b473ca8c : SUCCESS in 10m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3b27597d7ee740be9485f82a1867c41f : SUCCESS in 13m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/52ecfb853f2b44e08fc6ab2ecb6408ab : SUCCESS in 3m 36s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9a4b8bd453f248b8a12a1f855366b22d : SUCCESS in 8m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cee662df8e2c4de7b21550aee59dfcb4 : SUCCESS in 4m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/11b17a975f7242648ce88de189f3692b : SUCCESS in 9m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/346e82ba26f4422fb5a2ea7cfda71915 : SUCCESS in 27m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/55157e5b77cb48528ef29a9ff61c7c5c : SUCCESS in 11m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a96ffb2f4be140599af64c35f0287b29 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93b813529f6c4b8e947bc6cb73eb3c2c : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b1d075f63484c409422bc28920e04a9 : SUCCESS in 20m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2e83b5efd2e748c797c2077e81d04c0f : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3c95094bec7143af92aa6631c0f6abab : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24e990770cc5468f935a5b42a03b7608 : SUCCESS in 5m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fa405753dc9244c2b7904dceff30d55f : SUCCESS in 8m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f47a505a9b5a40289dab702fefc7bf83 : SUCCESS in 1h 10m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/762e1e979b824657befc892e916abb16 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":1},{"id":"72c3ef32d49f26931046ac389e880b1e6629ddd3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 15:30:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e010161cfa15454f96e5f41a3e5164cd33738175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:31:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a41b20b0b741eb8c96a5b20ceb3f658ecfffc0d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:11:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7e219a628e4d22a9829052e79b14aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cef5a2c5424146bcfb8ee004e860f1 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f8bc0269014ec6aa10910d415901b0 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/980865f8c93342f380580b8f29236fe4 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11aaf8b9a91b4bcda8764b737391a5d0 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2a4a7f6b5b4428da6e3cacaf95a5800 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db75f143c2e84774919023bb90a5dcf6 : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8a432b940ec48f7adc3297c78505a19 : SUCCESS in 8m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2fafc589c684d3ba6cbe38257209a21 : SUCCESS in 11m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f28e54ff20094dcd848c719d366d48ae : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e43f447fd5e74fcd8185fdd888b2178d : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3c3762948f0c41798e60bf4a56006f5a : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ac0625af73e84d5395baf5f246835ae5 : SUCCESS in 8m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b2e5db206b8843d6867ea23b8f8842a5 : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f5388ea6f8047e0a36c3ad33bcd2b11 : SUCCESS in 8m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a949153c8d5443c981551cddf48b7c46 : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/36f3b1584bc44da3869f7daf3154555f : SUCCESS in 11m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/940efb4d9ac24827a8bd7eafa4488f86 : SUCCESS in 18m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad89ccd844bc464e92a4b6444f6fc1e1 : SUCCESS in 5m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6efcd9cc753f4d03ba4ef4717d0a4d79 : SUCCESS in 26m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/219973da50a246cc842b3ab93622fda9 : SUCCESS in 11m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2ca9a17a42f04226b550ea24c93e9940 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/118f77c5f1ac485fb415931e56fd099b : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1352ab84313f477c8d4846e34f439791 : SUCCESS in 22m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1755329530f04b55bed2940f5fa38187 : SUCCESS in 7m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/69377dcd6c1b451cae8eede1798f6933 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cda2fd42ed734ede890a4dc579e0b2fd : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7a8e0f31726744eabbdb88c9dbfa8146 : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4a0362c35a141ffa52228a63e7fb4bd : SUCCESS in 50m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0446b3d41744d57b1323edd6e5161c3 : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b84adafd6da27fe09d078bcaf7dcefca3d85639b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:11:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a164f7076bdaa5851126bb448e2513432cba16fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:12:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f9410dc01444df48cd5e69ac78a399f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5efae378958f48148dc25ff8e6a4a114 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02477c6ea86bd795a9d53ee1d16e10eb256429aa","revisions":{"02477c6ea86bd795a9d53ee1d16e10eb256429aa":{"kind":"REWORK","_number":1,"created":"2023-02-24 09:40:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/874993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/874993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/874993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/874993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/874993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/874993/1"}}},"commit":{"parents":[{"commit":"217ae9e58bba430cdac310198da467db23b989f8","subject":"Merge \"update constraint for oslo.cache to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/217ae9e58bba430cdac310198da467db23b989f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 09:40:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 09:40:47.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.7.0","message":"update constraint for python-solumclient to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I6fcad5d8c1cf5adde0c640014665658a86ee3df0\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I29ec510dab912e87cb348a521160ddbced0bda89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02477c6ea86bd795a9d53ee1d16e10eb256429aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02477c6ea86bd795a9d53ee1d16e10eb256429aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
