)]}'
{"id":"openstack%2Frequirements~874995","triplet_id":"openstack%2Frequirements~master~I1265028c450fa4e77c87f9dd20d38dd9f9f4f92b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1265028c450fa4e77c87f9dd20d38dd9f9f4f92b","subject":"update constraint for python-heatclient to new release 3.2.0","status":"MERGED","created":"2023-02-24 09:45:24.000000000","updated":"2023-02-24 22:50:23.000000000","submitted":"2023-02-24 22:50:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874995-new-release","meta_rev_id":"4fc606b1035d1d0acccd18426de7744b4421735e","_number":874995,"virtual_id_number":874995,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 22:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-24 15:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-24 15:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 09:51:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-24 11:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 15:30:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fada34abee6e623e1644a397c5e6220810cb2eb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-24 09:45:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"300d23c73a3824648e02d4eb4660721103e76676","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:51:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dafff0ee0d24c76b3cf48ffb7c6ce70\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a421fd595eb47328ed236080cb52748 : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b2c99af3d79c667949641d179c40c9531b05f52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51b76d7e90442029b850b36e699c8ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c605a7d0574e5bb7e3a631d28cc231 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4448ac6f01b943229df2cbb2d5baf071 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb27c51a6d8c4dd286c98dd955e07882 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f36c939d0474cc3907401241a394674 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fc5ca63c42e46beb7bbdf750e93f69f : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9fdb94518704261bc9d78d4e894eeba : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e042a0f120df4bcfab4fa5369fafec64 : SUCCESS in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbf7b70ec54747438227fb8a0350e4f2 : SUCCESS in 11m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec6eceed3a2f42a4be644efa5559d300 : SUCCESS in 7m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fdcc6ce4110f44dba7d6af3fa2a1ce72 : SUCCESS in 5m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7db023ffdd844c84a0f00c47ef074c5c : SUCCESS in 5m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2ab16b32d59141b9aabed72589bd0f6f : SUCCESS in 10m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ac8e1521f95a45aca530d8e13c13cb24 : SUCCESS in 7m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/74ab32a15e4847d59226a970c22502f4 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6bb1d53e796c4e2a8f87e67b83c2c932 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/706a07faf8ac4b2f99de3c4cecc93a56 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c301a64d5164c4087837a4fca856273 : SUCCESS in 9m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9ac000c93ba24775abe85fa43693f8e5 : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/32117ffbd36d48008588867f7b4589b4 : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a17fae0f2cdd459a9325d3babe3a83ae : SUCCESS in 3m 54s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d63aab8a9fb74adfa14ac7a2c6323919 : SUCCESS in 10m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f1c5142aa80b4320a729a5847bf33a26 : SUCCESS in 4m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/dbd16be5cb9245459eb05dcd64360d70 : SUCCESS in 8m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3641d3ad269140c9a2606000c459735b : SUCCESS in 35m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/792e1cc4df884c64bdeb45d29d821f49 : SUCCESS in 15m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d9f5cc921b5b4debaf2c9c1723d16d8a : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b6a7471c40b40009990bb47bb2f83ad : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8287879261c84a74ac85dee77967a6cd : SUCCESS in 23m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/581da2afdf5f4f0497ca0ba9e026b958 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/50ea4876b2044dfea9cfda2e025a92d9 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57b562df073b450388b5c94b836048f6 : SUCCESS in 5m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0df1eeb9559d4b8b9f48cce7c39c4bad : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ecc46d7e2b24875bdf5882973770a3c : SUCCESS in 1h 02m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/444ab91acef44c918f046313768d5e4b : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ecabfe15fab7e3858d6a6c1b0f76297b08857d74","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 15:30:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"641ce9168411dc7e30d8e0954e3995bb7b3372e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:31:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65b345537eb31d9a4fde7b009e65ff772946b1bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:59:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c506227c9c7642ccb143d220bfad4f3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d883c8078cd4eb2a6254f65caa9e16a : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecec21c9c734191a447f7466728e709 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bec8a76928684b5b8e38f285d4a44782 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c167fcc01b7f44eba8f5ac206bf7dc58 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad8ee1c7b9694666ae609735d3df9cc9 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f157241258644394b6f0aea9a00c2bc9 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/342940d97ef0472894f2b25f6ed2995b : SUCCESS in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb3a365c735743e48961d9d32fafb5f8 : SUCCESS in 12m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d53e9fb67147467e92663683816cfe1f : SUCCESS in 10m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f8e0fc0f0a5a45ab931176b235103539 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/358579ecf5a74098932ac71d331a591d : SUCCESS in 5m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6c1be5cc58b54116b788670bfffdcb8d : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4421c368ff4c413e9c90cf23a5b96e75 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c8633901b3c41d08be7c8b271c83ec4 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c54a2c2f0e5f4062b14c1b020648338c : SUCCESS in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/17a0dc428c654260b1c52cf50ed5ecff : SUCCESS in 12m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/433ea52a8aa54fdaa1c691e8c7fea567 : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fb941d6e16424bc98451e0230750a50c : SUCCESS in 4m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d62929fddc0c4a39a32507c72ceea999 : SUCCESS in 27m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c65fdc03a9184cf0a6bed6a8d9f1952d : SUCCESS in 19m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/403c8266761b443fabaa4f776e3e7231 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b311fa0589b440b0ae75821621d6b5d3 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1210776c992943df84fecad0c6f0517c : SUCCESS in 22m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bc187648c84b4e32b09021a078545d21 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3a7740d0af844f42913569c45f8fdc0e : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cde32de924dd434f8ef2d0e4911c7107 : SUCCESS in 6m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c9906f6bab3348dbb06e754fd60bf474 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bdc72135eb54fa88a6fee84dae2bd66 : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9881be579f464ecfabe7df6bfe5ce323 : TIMED_OUT in 2h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d2ddd907cc4346ce53c5966036370a5e33f569a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 18:02:38.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeout","accounts_in_message":[],"_revision_number":1},{"id":"b3e202c68b50fb2124dcbef70abce3e3d73cc2c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 18:08:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a18c23498e49d7bc9ed377bf5454a3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/628138bd99b9477fab80efb2a7cb4794 : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9e883b015e9ec67b4ebbc1e55ad11bc241a0b293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 19:40:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de87dfa81ef46f1ba3fb193f46f7d3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7106d711bfdc4ee7a93c75b9538595d7 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2924aad7984409a89c8911a2436fec : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fbac15223b847bdb8e1687124676f5f : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02dbd29d7f9440f097a09b7285e5dc0f : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e2708cc20594b3d973616ee99e2afb0 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59e07b8e355944cdacb4225bcbcaaa40 : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/279e7a86330b472e842072c2f5e8ca1e : SUCCESS in 7m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4423bf8563dd439a9cb9f95acea5b6bd : SUCCESS in 13m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2df402bca69f4602a31675dc072d1e87 : SUCCESS in 7m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8bd3762ee13e4bc1bcca6d9a7ccc0658 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/055596d748184de794bc78ecfb91c340 : SUCCESS in 6m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a7d40ab0ad5748a6b69caecf3467608d : SUCCESS in 10m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e9e29daa180f41ffa31f297bdf9b5011 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/deabff6a24d540bd9cd5fd7c552356d2 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bb1dec9718e2498da3a6d16ce968d6b4 : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b01df071e2274ddfb95702f253158cc3 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04c5ed0ce30c46e29bb96b78775b48d6 : SUCCESS in 8m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dbb5f48c6c5046ee875ae97478a3d5c4 : SUCCESS in 8m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/81412f55acd94b07ac6e48bc322f11c7 : SUCCESS in 11m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/24b42eef77244d90808f376e6727f90b : SUCCESS in 3m 25s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3a7e4f5282f24daa997e15a66b357b0d : SUCCESS in 17m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/76b13f913ccf4a2481d0169065f6d80b : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0cd8bd2fe2e14258b4fd98820aa76eac : SUCCESS in 9m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/11566be214084cc9a3de362d77dbe7e1 : SUCCESS in 20m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5f77c4568fa6401eabca61ec7e23163a : SUCCESS in 12m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a14554e9b5a14950ab265e830029819c : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c21ebf044faa448997bddda3bcc0a6f5 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9499a41a0c114ac3a63a09380f75a9b5 : SUCCESS in 23m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8337fce42c174586b769c1161c06ec88 : SUCCESS in 4m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/282e327ed7ce4496b65961bbc29f6d51 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b3588d5058c44a98e684765e8f165fc : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/215cb39d323f4799bd0c706cf02f7f94 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3aaa6d8dab95431fbf717e51e6f26b65 : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9b17814312c4d2eb810bc0122f8a84b : SUCCESS in 1h 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ed0ec46fd5dffd0b2401719d4d772fa7ba6d89dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 19:41:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d8ea46aea8bebddf17a6fd53cd0f7e87689124b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:50:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf4278866a74fcc8c8e752c030667de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1ba426d34f41d69713dcf14310d921 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4303b219cf9743b6ace4d12feb28dc12 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30a8fb89aee64ba69dafa31ebe114d9d : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7a8607d9675497cae08a8d8e08c9b35 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35c1f4ee025b4df8a5ba818bd491a968 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c78a271bcfca4924b139df698e71bbf6 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/47eb0bdcdfd744f8a7aa241e4f178e50 : SUCCESS in 8m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5eeaac5966c8423f80aad7185fc28959 : SUCCESS in 10m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d8cc2ad5a4b248ceafd3095ddf250660 : SUCCESS in 8m 40s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/53b1451efb7a49a7bcb15df2c490a2d4 : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/47bdd601513a4e5a898adeaff6d446b1 : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5034fe85ada946d59f1dc34867c97f42 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/068b3b8c7b9541b88b23da46f91742ec : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5885633fa7174d8a915e2cd04e9f20e7 : SUCCESS in 6m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2964aaf03e4a4809bf31f2fc7c648c74 : SUCCESS in 8m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ca476939145f4de69678a80b14debcb4 : SUCCESS in 13m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e288f59712f4a9eb29b9ebaffd47ebd : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cd663aa407f14b138998ea4ec6623b83 : SUCCESS in 11m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/66c2b3eb68e24669a6c3d794920f1530 : SUCCESS in 5m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0dc7da0660d74da6b09df56e211f9fdc : SUCCESS in 21m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dff4e2dc1086431687b5dbde157221b5 : SUCCESS in 11m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c3d2fa5f5f05441aa9d02e7dfbd1639f : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e60ac5527af24c229886152e2c39ddfb : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7352e785852438888b4427120e56dce : SUCCESS in 19m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a549b1112be44416aaa4cc5a082ee618 : SUCCESS in 6m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3f001e99e3b347178eb6a93b02bd8697 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30c61139e10545f9adbe1909359a75e6 : SUCCESS in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/29a4eaa26ada40d0b9ba853039e3bf35 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fe85f42928041959218a1e19103a7c9 : SUCCESS in 51m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e79b9e1880542f79e75cb8ca3f4a9c7 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4fc606b1035d1d0acccd18426de7744b4421735e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:50:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed289149f52abf8122647ba9f0b550019f6ce567","revisions":{"ed289149f52abf8122647ba9f0b550019f6ce567":{"kind":"REWORK","_number":1,"created":"2023-02-24 09:45:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/874995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/874995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/874995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/874995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/874995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/874995/1"}}},"commit":{"parents":[{"commit":"217ae9e58bba430cdac310198da467db23b989f8","subject":"Merge \"update constraint for oslo.cache to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/217ae9e58bba430cdac310198da467db23b989f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 09:45:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 09:45:24.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 3.2.0","message":"update constraint for python-heatclient to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I4be991bc2d15b3cf9fa5c537db5ae7c6359a2695\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1265028c450fa4e77c87f9dd20d38dd9f9f4f92b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed289149f52abf8122647ba9f0b550019f6ce567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed289149f52abf8122647ba9f0b550019f6ce567"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
