)]}'
{"id":"openstack%2Frequirements~875047","triplet_id":"openstack%2Frequirements~master~Ic8b5ca5cbbaa5c1ae68eafcddbd047f6ca995a55","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic8b5ca5cbbaa5c1ae68eafcddbd047f6ca995a55","subject":"update constraint for python-octaviaclient to new release 3.4.0","status":"MERGED","created":"2023-02-24 14:32:53.000000000","updated":"2023-02-25 05:13:03.000000000","submitted":"2023-02-25 05:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875047-new-release","meta_rev_id":"79cc6d01a4e97ed4a3be714fe4df81dcbc3850a3","_number":875047,"virtual_id_number":875047,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-25 05:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-24 15:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-24 16:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-02-24 16:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 14:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-24 15:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 15:59:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-02-24 16:12:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d0736cd20bd4cce8a37247d1e2e640bc42ac869f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-24 14:32:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6ad3a2d6a52daee2a44b2e44f00598410917c88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 14:39:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfd7a7f9ef384c8da089e8998dc751e4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6385835508fe49a5bde4d6b5cd7192e2 : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"07a8250164545cae6ba16d9f980a96b1237c6c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4daf41e441054f24873de272a1f291d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884dd976cbf24d359495f1204b9be8e2 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933c7dad275f4fd0b7665568c071162b : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1285f039594f1f9244199d741ae3d9 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac4d838964b24b209ae025166931b404 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43f47c27e4d94594b30bbd95ecf513ff : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d049298127b4a8d90d20c9020f614d2 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c373c2818fe4a95909a77cfd83ae9d4 : SUCCESS in 11m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/515c40c4fd02413e8bf109028d302c42 : SUCCESS in 14m 05s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ae6561dafd4a49429d64ab7935556418 : SUCCESS in 7m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d04702b722f949c489f58a893795fbaa : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/320b6289a19341bab74898a69f9c2ac7 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/01f0cade43634893813e763ced5b528c : SUCCESS in 11m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8a5c18fcd7c345ab8d59baa570ebaed9 : SUCCESS in 5m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9eeced35d1c24933b7dee4df063cc843 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b7a7e75318104eb5bc99d11de6e9cc62 : SUCCESS in 9m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7d9b532b83204595b189d85efcc31782 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8731abb8616642cdb030c9feec4690c9 : SUCCESS in 6m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e0b3c3c118a74b08b301cfb792dbfa4f : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/89fa9e2240d845f9a9cdb5bbb6b1cb4a : SUCCESS in 21m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/994f7b30093d4d799b4b95df0eafc343 : SUCCESS in 3m 14s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4acc41a70ab94b46bcadb4b40d9bb39c : SUCCESS in 10m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9eb5065c84764052b533125ff36b0b47 : SUCCESS in 4m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c4ac602a626d42faa7d4b4a331921ea2 : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/57873b27e734452495bbc8c66f52cdfb : SUCCESS in 23m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/737231d171bb48ed95ff12b9ba8cb071 : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/efcfaca1f46848a0b439c945c632b699 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae392eabd60f4c69a3dd97541f17cfd9 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96b19d4263934506a6034dbb88740e46 : SUCCESS in 27m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/13d220e279cc483f96d450da9b2ea075 : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4e25f58fa84441bf999e977bac0f6f33 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c73da2c11c443308026f42784cf86b7 : SUCCESS in 6m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ae11a8b2c134eadb79f8c7ecffbd7a4 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e0a7cd4d7c74d7eb19338175398f0f7 : SUCCESS in 1h 10m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5964e0cf18344379c85880c8b972c4f : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a28217530cef759d3b4b03515d7fde1c585ac52d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-24 15:59:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6198504f087f87ea78b32291b97551d3d600f964","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 16:12:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d03f7730e6a8dd3873ad080f6a9dff9a68868f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 16:13:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9315ae41eb905640d0cca24349cff702d5282894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:11:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ad4697308ac43c5a0efe4dd96130ab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f495631382af4f5a920824970d740abb : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232a98ab0fa242a2a09355d7eec5b5dd : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30cc4c5eb714eb5b868ead50daa470f : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9d2f193046d41ce99f6b59e32cb728a : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77361d03adb1459aa8170fef208a90d9 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cad8a15022bd499ab2a4353e3032c487 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1249be0eda1b4176b0e9dc62d40841fd : SUCCESS in 8m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cd4d72e858c245018941ebe062b5ae42 : SUCCESS in 13m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/869eadfe4f524378a9fa2cb8acac8aaf : FAILURE in 12m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/332e19465564428889a05dd881b7ab29 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ec8bd6d1c6114696b5eceef18ea9b04a : SUCCESS in 5m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8423ef5d17d54d27a2d4196ed5f10ec3 : SUCCESS in 8m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/119122b1be514fb7b84d40aa87f6414d : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/824aebab99ba4d0c94ea98c0b6822af7 : SUCCESS in 8m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2ce5e2616fe4435092cbade9db06791f : SUCCESS in 8m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/38fb577d8cc14853859809b4eaeb4e04 : SUCCESS in 10m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d65b0f405c8f48b8b60b73cd1e67d31f : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cd3bb00503fa4103a6318cea86127d5b : SUCCESS in 9m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4b39fffdcf4840c2a35a72926439f12c : SUCCESS in 5m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6e77d8ec5aa044ee9df5c0b08082bab6 : SUCCESS in 22m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/001caf37dc8e40b5a5769529be8deb6a : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8072f3b2d65e4557b55fe65c41616450 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8023e7e14f3c4f2abdcba08a907fa601 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa41be064b074247a0bde2635deaca48 : SUCCESS in 21m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ad4e599c317f4d318ed2d5d323233393 : SUCCESS in 4m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2e7c9a65a35c4cb3b1adc0d116b65012 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79b2224a91a34e1296f1857754299191 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/58ac9295514d410c85112b6abb112317 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09a38b63927b4559802ab9aedc6eb9ac : SUCCESS in 1h 00m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a61578f12744269ba2e7d7efa91e742 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":1},{"id":"683d5a2eba9dabcc7a25a8fd6a7b65d4bbb5048f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-25 01:53:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"126229151b2e26d5b33dd55e20b74d524cb2e49b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 01:58:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ffcff19129846d6aad9c0c59f521e23\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d5b77b669f474e3cbf4dbd90be4b8fe5 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"148436051d9cf6c7056d5361e0a3a02a14f806c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 03:29:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31a46ea287141d3b7dd2a5d6f71a3ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0313752dd143ecbbf3fea118afb1cb : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddafcae8856545f8b8ae52510f9a4a97 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31e01550f8db483c988fe2ec439a2add : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e94a2d0249124b82a9091dd3d95c4cbd : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9dac89e0147414590a4598cee0a8cbf : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3eaf7487f7f748b0b04a0f42dd839279 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/814971f2da6a424d8230d158c05e2090 : SUCCESS in 7m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33a94606c3764d1a9634cb5a2da06524 : SUCCESS in 11m 18s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8d54dc44ab7842318cb9817c970c48cc : SUCCESS in 8m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8f721ecc63fe488990971999a2a7801f : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5232eb1c34ba45adb98eb7b86664ac3d : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c2ce58735cd24a1faec2864c5dae8752 : SUCCESS in 9m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c516aede4b4f479d8b662aadf571cc07 : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/69b01bce44a74626bf24b8c78304be60 : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b4095cb9993b46d38a90bc0400e1fd70 : SUCCESS in 7m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bc0a9c25253542eebe9efee5c11d0bfd : SUCCESS in 12m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b0953b440d54d7e91de71b3472c59a0 : SUCCESS in 6m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/96724829b1ba42e69c05863cf74fcf3a : SUCCESS in 10m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4e4d34c99c4946ba8013e32cbf2818ce : SUCCESS in 16m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2e730210f223483f9fb5440fde80d34e : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8e950fe7b3d7428d9981d3da33448211 : SUCCESS in 16m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ccf5e70568404d1ca8c500e8d0078461 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b511298476b24c36a9863151ba14fb2f : SUCCESS in 11m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/61d53d1ea0494094aeadf3dd02bd2ce0 : SUCCESS in 21m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c9c4166b98134a25bc0050e7bc1a1cfc : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6a53d4fb137f4319919019a278c6c741 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/422528a29d46422db78104deca465faf : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbb780e83fdb4ca486b1fac5686f6290 : SUCCESS in 21m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f9a2191943594e2185926e139c349558 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fe9b8ef20ae94a939cbf547c40ec892c : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d1752aede994bde8fb90ff67d6ed6dd : SUCCESS in 5m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bb6a75d682e4417aafefd46cd2daf9a8 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e368450fc81442686073229323d6720 : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43995d1fae8944e8ae0333299761c7f1 : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4cd87f3813b7ddf183f252d9868e3bbd192f6f26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 03:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b16d985bf25c6293f5068db3dd1beb7e1077eb73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:12:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a156d9419fa4913a4d6ab458cef3d3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9991f071416a4210b5153604e7689d19 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f72ee30ed114c799aebae43d917c960 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bd7cccb0fdf4c0297937b4678987698 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30a31ae3799e484bb52010fdde2ca2a8 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/438fb67d39ef4872af23cac78d04c23c : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84eed7929d534abcaaa6a2f04ca268c3 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d318d9b0195465fab16f67d57b2fcbf : SUCCESS in 8m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eea0b28fc49f410ba67c789de91ec2b8 : SUCCESS in 12m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/49d33f3aa90748f9ad61015df00150be : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f56f5d60a07745e5a07347e8f83e829e : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ebdd93efb1f747dcb70a1d248a7d1058 : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9281fdec2c454b838523edbfb4850876 : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7645ab2a173a40afb42c843b7c0a746a : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30906c2b0f26491c9c73718998671516 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3258aecd95004fd0859738672b86f917 : SUCCESS in 9m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/907a62d651de4155ac27fe67ddf0b106 : SUCCESS in 10m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0e264cc07f734c5e82cea1fdb5f6135a : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/46ef2ee13e384301a959665a185084e2 : SUCCESS in 8m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9db233246144478086d8239b1d46b1bb : SUCCESS in 4m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6cef913fa13542a0b56f7f27011849a5 : SUCCESS in 23m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fdbf92e337ec4caba11fae0ce0ea39fe : SUCCESS in 13m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e5a216593a5b4e4dbe5c81ab86e7ba5d : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa70d4cd4c9c45158db5c5c6199e5992 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5706add250e4a1ebd539150e9ce7950 : SUCCESS in 24m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fef43ec762f24c0898196b1466a25bd7 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2092ba168a2744eba2bd19d3c3f70813 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2beeb8cb3dff442ba17bc340b78bc7b5 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dd5f059d209e44f180176f3db7a525b4 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3136c46f167d44ac92bee2c05c56ad1d : SUCCESS in 59m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bdcdc7e57eb4c8eae77ca63531228e3 : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"47b1e4c4d301b2d89ad95a30af24f71db39e100f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:12:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79cc6d01a4e97ed4a3be714fe4df81dcbc3850a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f0eb0a1cbd44a586f3f6691d4bddfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb8c12cf53146dfb144290e696020ad : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90f0ddaf00ad88f99e3ecf72383770831ec835ef","revisions":{"90f0ddaf00ad88f99e3ecf72383770831ec835ef":{"kind":"REWORK","_number":1,"created":"2023-02-24 14:32:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/875047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/875047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/875047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/875047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/875047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/875047/1"}}},"commit":{"parents":[{"commit":"217ae9e58bba430cdac310198da467db23b989f8","subject":"Merge \"update constraint for oslo.cache to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/217ae9e58bba430cdac310198da467db23b989f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 14:32:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 14:32:52.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 3.4.0","message":"update constraint for python-octaviaclient to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Commit: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Change-Id: I50104b3e1fad1346c3273515fc177dea5acce385\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ic8b5ca5cbbaa5c1ae68eafcddbd047f6ca995a55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90f0ddaf00ad88f99e3ecf72383770831ec835ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90f0ddaf00ad88f99e3ecf72383770831ec835ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
