)]}'
{"id":"openstack%2Frequirements~875204","triplet_id":"openstack%2Frequirements~master~I425cab648dc744dbbddc014b140d824107528935","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I425cab648dc744dbbddc014b140d824107528935","subject":"update constraint for python-saharaclient to new release 4.1.0","status":"MERGED","created":"2023-02-24 15:34:59.000000000","updated":"2023-02-24 22:15:25.000000000","submitted":"2023-02-24 22:14:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875204-new-release","meta_rev_id":"a743b0a7330c0d8b2202aed50c5c654583943cd9","_number":875204,"virtual_id_number":875204,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 22:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-24 17:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-24 17:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 15:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-24 16:54:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 17:01:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ccce1d559ac1150b0d74d1683376c76da59d57b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-24 15:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7e6f6182ace827d5128395dca9f783b32860640","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:40:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f00286d7e242239594233ecebed5f7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bff8dbd688a241f3ae06612191b4719f : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a873aa668be4339ca97e1bde3cc2f29dbae3e8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 16:54:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7295a1a9f7cd43858cb9200de4eead29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb81df08dc240bd9f6690aab4a8830c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e0011c8e6646ef94050b7f3f7a8bc2 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246aac81a88942ad90a3e47bb8591101 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9763ec61787946d88547542396e0374c : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7f5b86b2a7a46969113868479b01a4a : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0aa4d32e82243b885655385ca4bde58 : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3fde3ecf38264742b5255e93e575d703 : SUCCESS in 6m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/66d621de7dbe4d70a606a83535af5506 : SUCCESS in 14m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a4fd8389eda246578d96578a0e7529f0 : SUCCESS in 6m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c4d279d74ffe423b9c2efeee0611f3f8 : SUCCESS in 6m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bce9fd996fb149388c8eb9e95353e76a : SUCCESS in 5m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/696043becbe24a96b790ad82d4687340 : SUCCESS in 10m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7e95ee9a48b4076b39fb5fd9c673c24 : SUCCESS in 6m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1ac36d8985694928b399098abe6fa1d5 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7b618e5702714a1588b429494900cbbd : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/827cff572d6f44cd849c0bc3fb3e0d3e : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9219eab5fdc445ba96dffcd55c06573 : SUCCESS in 8m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/daba8acd82394c89a2346d3483b8449a : SUCCESS in 9m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e21660c77b224a16a00900459cc29fe9 : SUCCESS in 10m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9ef347b196914c97adf886a34a62acad : SUCCESS in 4m 47s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3fa7f5aed0214c2a8b670e934751d66d : SUCCESS in 10m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/67de9830c258452381fe5d4fe51d08ea : SUCCESS in 6m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c26656b0931a454ea3e2285a800d8a50 : SUCCESS in 10m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/daae30900ddd483e9b23ee073978d749 : SUCCESS in 30m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c41fc5f822a9418f9fbb780c3335aa92 : SUCCESS in 13m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/954b381ff57e4325bb2f5e150e76acde : SUCCESS in 6m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1ff1dede2c9549dea6de26bbbcd518ad : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c284a4abef041acbf3a4f6f3427c29e : SUCCESS in 22m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5b00892639a24f9f9a1423e04f90ee98 : SUCCESS in 7m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/99c9b3bab82043c7ac968a008b99e115 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7203df5aa5e4cd89dcbd33e679926d8 : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ce5ffe8384a44314b8dc20a23b4e9ff4 : SUCCESS in 21m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9be410400504864ae956b02a7cee1d0 : SUCCESS in 38m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57c4bed46e4d47c0a1d93b551cc40791 : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0d07d7f03caf41db835777a00a9123a7421f3597","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 17:01:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ca6cde35c2740d04b34e91a727db5cb135e26f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:01:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8929b61ca11b932b990d19760ec8aab705c1e3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:14:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551a5790d4744148a13c40e9f29702cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df57ecee7f149c99428405d704911c0 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781dc125bf4e4fa7a118b5bbc5f06061 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2fe174af621480c853f44196a87df38 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2afad205c3004d169b61a1de7cdc4dec : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa01e293b9034ed2b605c6c3ed240807 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74ba7d3d2dbb4742bd59ce63de8bdc8d : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cba8fd7a5ffe48289cce56b0779bf97b : SUCCESS in 7m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c92601f3ae9140db8fef99b46cea3e6f : SUCCESS in 9m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ebba4047388a42e69771347ce2ea6c88 : SUCCESS in 9m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8c06e8e1128141eb86d57ac84b629305 : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/17073158bbc04a6bb375c0c93848a0c9 : SUCCESS in 8m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/067d9ecab8944dc4a1c0a7b141ba3f90 : SUCCESS in 6m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ea29fb49b40243d9ad50308d7a14be8c : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f08d11281e814870b894ea7d5c8aee22 : SUCCESS in 8m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fa1095c1583f476da8b2a4c28f664639 : SUCCESS in 9m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9eb90bb8f55247f7af28b715a651e828 : SUCCESS in 12m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/094c42c2807543fabec7e39eb5383fb0 : SUCCESS in 4m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3aaca36add9c4993bfe0345631c52b56 : SUCCESS in 8m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/79bb86e1c1bd4154a82f80bbf161c404 : SUCCESS in 4m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6af3347853eb4716af033cb328bc8cc7 : SUCCESS in 30m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3f67db69468b41098bdc06ba42371f2b : SUCCESS in 11m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/88b4e727f73a421687b71f7fcfc36a73 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7cafed640c784e8d90a08e5b307401ed : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad0ec7cb6835478089ae4d62c6a36575 : SUCCESS in 23m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/43ff9bf47b5f4689af91e21c5d1d55d7 : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6d62cb368e4d48c69160d22ac9aa54eb : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d36a31687a204874ab902ad3f840dc78 : SUCCESS in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d76f6717fb2246118e98bb4679aef6ec : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4def97a725c54881aae0aead5db50384 : SUCCESS in 43m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16199ef865bf467597dac49d4470c808 : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d8fa1b53139fe6605a97ae7d10cf9390f5d19c22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:14:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a743b0a7330c0d8b2202aed50c5c654583943cd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 22:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07a046c4ab7481884f05a17d04e2317\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d6a17fd79448cd98a85a01d394f247 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5675791b4b6ccc8080f021b80526c5cb20970057","revisions":{"5675791b4b6ccc8080f021b80526c5cb20970057":{"kind":"REWORK","_number":1,"created":"2023-02-24 15:34:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/875204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/875204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/875204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/875204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/875204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/875204/1"}}},"commit":{"parents":[{"commit":"217ae9e58bba430cdac310198da467db23b989f8","subject":"Merge \"update constraint for oslo.cache to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/217ae9e58bba430cdac310198da467db23b989f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 15:34:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 15:34:57.000000000","tz":0},"subject":"update constraint for python-saharaclient to new release 4.1.0","message":"update constraint for python-saharaclient to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie821b12599abf40fddfcab5d9d1ce2b9ffc44cbd\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I425cab648dc744dbbddc014b140d824107528935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5675791b4b6ccc8080f021b80526c5cb20970057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5675791b4b6ccc8080f021b80526c5cb20970057"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
