)]}'
{"id":"openstack%2Frequirements~875206","triplet_id":"openstack%2Frequirements~master~I5bf69096ad20b65cf2569266a8415727362aaa4a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5bf69096ad20b65cf2569266a8415727362aaa4a","subject":"update constraint for python-keystoneclient to new release 5.1.0","status":"MERGED","created":"2023-02-24 15:37:29.000000000","updated":"2023-02-25 05:13:41.000000000","submitted":"2023-02-25 05:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875206-new-release","meta_rev_id":"a4cf5f2a4ae95b21a5c9ea6bb3d44951a54b0136","_number":875206,"virtual_id_number":875206,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-25 05:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-24 18:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-24 18:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 15:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-24 17:04:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 18:02:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ca313077da3cd53d460f07f7c6133d8fc69983b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-24 15:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d7f7944be7bd1115ec052cd03cbe618ac80e497","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:42:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84b6f77884f14e56a86f5322712b7515\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/29bed32f0af04956bac48b49e77bb9ad : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5563c6d5b67b96b62f3dfa10357b2c3d71b46cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:04:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa40e6253f754c60847ffefd48bac26c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b75cbf907214e7b832d9e3d2a4e10f9 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6b2337f506420296329f3cde271b60 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feb5d82eba73477d9dab6392c4b76fa3 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abe43341b80a4077b4ebf32711379a8a : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13ff1907b4994c3981bea10c48c45a28 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18fe7ee868b34f2daf83b486acbe4f3d : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5485d4101ff44f98a1db11f74e543b7f : SUCCESS in 7m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e046c3ac4bd548d6857d69265aed3748 : SUCCESS in 13m 01s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/14e11ab2da13477fb89a5fb2c62a8610 : SUCCESS in 7m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/de339d5419114964b42fbfa5bff87f54 : SUCCESS in 6m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/74b4a3b6bfd04b0681b545601f5a54b1 : SUCCESS in 6m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ff682485ce2f4ce0a643ca88d6bb163e : SUCCESS in 12m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0a8bd43a944f4a18a0c82fc70686ebac : SUCCESS in 4m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/558b2b9244fc40eba90ac90b82afc808 : SUCCESS in 6m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9c55614f06ee4f6dad2b1f7a497e1dce : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/561a50ca9b3449a79c4cfd949f4eee45 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76b56c35f47d405388421b4debcb549c : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/baa66e1b1c694b0298ea56f231b61c1d : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b9aa6d912be24a7bafaa765da02db482 : SUCCESS in 13m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/adcbfbc1e5ce42eba960eb20366cf67c : SUCCESS in 3m 35s (non-voting)\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/19ab20b0c7a3413cb1bb5e8469bf74e9 : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7fa6d509fa1d4829916e91bac509aa85 : SUCCESS in 4m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e39d089fcfec402f9924db74c2be643a : SUCCESS in 9m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9ade569c42f44de68d9aa479eca8b4e8 : SUCCESS in 26m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7f8184ba04fb4c3fb949c59e01fad78f : SUCCESS in 10m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/affdb70d77884396a7813afe6e71a11d : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b9c20b948c2408abf83cafb14702567 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcb743cdabf14244ac3bcfa42195a328 : SUCCESS in 26m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d87538afe6e947c8b809aab83266468e : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/02ffcad9483a4a09885a8d07503fd603 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/451bc8ae8d5344dca43f8e3e5084fe86 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/abc0c803b8474ee8aad5d9bba811bdd4 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/329bb393455e41bfa75b04c62f661a67 : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed71928cab424e7d879d938587e468ef : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cf167ea729e7dce5119848d86c4544440e71b21c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-24 18:02:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ee58678e447120feb5788c9db34998b3f1fbfe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 18:03:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"041694d61c4209d5963514187342c7204d38f1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 23:13:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88bced21f0be4ec1a92e1688c3cfad2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76725eeef7a74a88a2f515046253bf93 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd1237955cd4692b5ff066d62334116 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/204a05f8b2e44a39838dd27552c8a664 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e9fe8e5a883444684202ff7fd7d2800 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/274065e307794e6c95b1ffd0392db999 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1d643364fd0486392d974d611792ff8 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/010db0d0437a4eda99c36b3b8bf94a33 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a34aea4f969c41b2a5a34803c0927f27 : POST_FAILURE in 14m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ec4f479349a464cb7a882ce6b22633d : SUCCESS in 8m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5165e44ccbb14a60a71234919e50469f : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/beaac584ff4940628fb073f10a75ce3e : SUCCESS in 5m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/689256f5241846aab968f0ebb4093514 : SUCCESS in 7m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f59aa03f0d5b456091c4b0481b0a3d0b : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d0e9a345dea42c68a3cb5f09c4e03bf : SUCCESS in 8m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/118849dab448495bb86f5a23fc54be69 : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c37a2ec872894c5982d35307a0e01c7d : SUCCESS in 11m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9d9de34543d94d0fa13ae788e9a94bc5 : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a166b042e0c145e88bb5396131bb8e9c : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/22e7241268d94e79a077f8c271a45e3f : SUCCESS in 4m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f585329eaa2e4f0b91ea737b6dfdca95 : SUCCESS in 21m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1c883ade956144c69a58ee14013f8aa1 : SUCCESS in 12m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/28d72764c5164bf3b2e3b56c37a7e7de : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42211b896fbe4b4a92417233e723ca1d : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c53bb848d9446949a6229df06a16607 : SUCCESS in 19m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ee3a090f3d994a84afc5b97ead290282 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cea6445f52984cf1a5243b41a33293fe : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19d811f8f8fa42fd8c0fafc4c04b56e7 : SUCCESS in 8m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/897e656f9e6b4042a20c60273836ccc6 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/220e39f61dfe477480af31a880e6bc59 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6fc7dd61b264437b5769d4e08abc25f : TIMED_OUT in 2h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"900ba4bf9238143d6508e140651bc9c1e344b1b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-25 01:53:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9cb88d90259a3b3bcb03f9c754185980c73bc33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 01:58:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c2266bb6b148448bb899c620aad21c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/afbdea9386dc4d3b8df52977bc93e7c0 : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b0476357538fe739c2051f87f191c61a4d430595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 03:40:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/488c600d4a864c2c929623552f2fc306\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28a860e45a44bc1a730174fbfe4eddb : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f607c5292674a7a9c601c4009b308e9 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7526fc75fc8641b5ae1efbcb8de5ccc8 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0a8882545d64b6b97198f7319c4dd48 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1ce4f6f475a408cbcabf2eb250a2029 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0893a5c8a9841ac902edc0e7f86de69 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dbcfacfcdb0c4af08d6130387f9567cd : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/146ad77302094c22b8ba8584961a8c1e : SUCCESS in 13m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9ec48bb5b7b24b9ba246920d3a191e08 : SUCCESS in 11m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b3987e40231d4f4f913b33b11f91aa3d : SUCCESS in 4m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ac866a75b6b040c087cb06154d50967a : SUCCESS in 7m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5212211a0282422a85d7b6b359800257 : SUCCESS in 13m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/10b7cb7f1f57427ebb25412312b52f62 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bf73fb96aac64cf5a0796f249b70c160 : SUCCESS in 7m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0fdd3cea6b424f1f8ad86bc735115ce4 : SUCCESS in 7m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7b0dc1662b674d1798adf3f048d418b8 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff4a857185de469994d65c375cac07b6 : SUCCESS in 8m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2d08ce7c51af4780b7f5c6da12eda134 : SUCCESS in 7m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3f349bc276cb42519090d01791a851eb : SUCCESS in 11m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/10d381c25f3f4395a40db9b69241779f : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/91cc621182f743c5accb29824c302695 : SUCCESS in 8m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c58475096dff414ab13ef978b48b5165 : SUCCESS in 4m 46s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cd6ce74c9c904be389faf1b36ccc5594 : SUCCESS in 9m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f15c5968e6c04b5ca620dd4d2734d9cd : SUCCESS in 20m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ab41e20d664f4bf1a5f8ec6a1f12defa : SUCCESS in 13m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4a10ae588f7d431dbc23fb869521bcb5 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e39280e3f004dbeb1e16675d29419f1 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d2cff88defa4949b9be4583a32452de : SUCCESS in 20m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/256e0a6e08f449b78f53ffa1afe0fdf2 : SUCCESS in 8m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/359a7f381d3e43c5a256c77c0359e37b : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79c2d8417e6e4918a7f007b16a8a2e7f : SUCCESS in 5m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eee1481740c04fa48ac367f88c90a4f2 : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85f3bea9177e4915abe586f35c44a85d : SUCCESS in 1h 00m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/482c7c9865ce496a99485117333c4593 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ba2cc56becab9f97a40e9fc13cc5d757e7810d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 03:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7d145b606178302253b83723b2a4ce8ca3f387c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:12:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a35bbab89740b9ac8c18bb13238463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af4996abd21471a972341e89290709b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f4577392f54a74a43686055b79a403 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8373bee2ccb346019e7d63fc0d039b2a : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e0cb0b2a2cf41a8900f3219f59aab9e : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74e201df337d4982a3651ad7313011f6 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b67f7cdcd97e47328c4708d44bc83dad : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e495cd2696d4c12beed2ad70e69b5fd : SUCCESS in 7m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7902472f40b04ad59bb9e473e4743837 : SUCCESS in 11m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a5e44b60bf474139b44baea3388fb9ab : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0ebef00557514e03b16ad5e7759bf163 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/26aa955bfe3e4e4893cbcb442e3ab8ae : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/74039f867f6d4907a486a8f66a2a52cf : SUCCESS in 9m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f6e15e404d604e609ef21b9bad6630e7 : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a4f0b7f3225471f8e73b2af883c6f4b : SUCCESS in 8m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4ad364b9d2e64ed08e33763577a0d1db : SUCCESS in 9m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b9ca43011f65416fb2938568fb4431b4 : SUCCESS in 14m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/befa2c91bc834ae8bcc24f39dac3c2b2 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1ba2156c5bd54b6b9e235284b726a3fe : SUCCESS in 11m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c6c38124400141d48d754d4e3ee4af53 : SUCCESS in 4m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2b406926f74f4083a16ad7069b91af21 : SUCCESS in 26m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6bcc36fd2ca04b0eb5a0519b182b80a5 : SUCCESS in 11m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/de45fe57283e47919f189715f6e399e3 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a283eb718d4943ed862c4017e4c0b22a : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b0dcfa32c9c439290f9ae020350054c : SUCCESS in 22m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9555c6e244b54a1491bab6bccdea5c9b : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5ca4ceafe1d04deba3fcf9e162df9354 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c8195b0411645e5a76f763612f45734 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b83bd0f94a1749a7846b81feed6ab477 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4607e6139baa400d8630f8caf6c47148 : SUCCESS in 44m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42149f7d17c34377b46766140d0f28ac : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d5c32bb4fd12234ca0a31bddc56d507a609f8105","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:12:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4cf5f2a4ae95b21a5c9ea6bb3d44951a54b0136","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:13:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19f83bd091041ec9c510fdf52cdee21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b722a2925ee43399f692a3bf8b9a9f1 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42e017ace05fa8a71b586e7ba9d5ef97b401c9c3","revisions":{"42e017ace05fa8a71b586e7ba9d5ef97b401c9c3":{"kind":"REWORK","_number":1,"created":"2023-02-24 15:37:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/875206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/875206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/875206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/875206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/875206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/875206/1"}}},"commit":{"parents":[{"commit":"217ae9e58bba430cdac310198da467db23b989f8","subject":"Merge \"update constraint for oslo.cache to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/217ae9e58bba430cdac310198da467db23b989f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 15:37:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-24 15:37:29.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 5.1.0","message":"update constraint for python-keystoneclient to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I8b9ba5eb350c502565e3022fc5f96054f9f75f3a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5bf69096ad20b65cf2569266a8415727362aaa4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42e017ace05fa8a71b586e7ba9d5ef97b401c9c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42e017ace05fa8a71b586e7ba9d5ef97b401c9c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
