)]}'
{"id":"openstack%2Frequirements~876158","triplet_id":"openstack%2Frequirements~master~I0501b73af7079d8774be83bfe927a710a04a9647","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0501b73af7079d8774be83bfe927a710a04a9647","subject":"update constraint for metalsmith to new release 1.10.0","status":"MERGED","created":"2023-03-02 14:30:21.000000000","updated":"2023-03-02 22:56:07.000000000","submitted":"2023-03-02 22:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876158-new-release","meta_rev_id":"2d89f3e6f6d9ad508bcdd903f5ed324b971808bd","_number":876158,"virtual_id_number":876158,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 22:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-03-02 14:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-03-02 17:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-03-02 17:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-03-02 17:53:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 14:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-02 14:42:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-03-02 17:06:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-02 17:53:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-03-02 19:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"893a55518304cdfde15d90259ee6ecabda865fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-02 14:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d648ce32a5e80fb7c5d65d436931c5fbaa6383b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:37:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070e6b8308244b5bb3ba34b71254f359\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7f80bc4fcb6b41ed8eaf4d05262917d8 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2cbb2fea93d337ab7b94ed0a7bd91b2253e7de6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-02 14:42:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b766ca728134157c677a1352d1c60c38edff0272","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-02 17:06:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a0cc8308ed8b6d4b440a95d8258c9c9d40e9075","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-02 17:53:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a65f16829a7053b0178cc864afb138a412531a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 19:14:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869ea2cd913a4d41bf1e5ac5e18b2fdb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55c9e869659498fa267eb45cc5d61e9 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba8e77fddd949dcb15f6f2f76e0f22f : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70bebe1afb934b1e84c9d2bf9c908bda : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7511096050174cf9820e76a4b93fb30c : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9447b6f54e4a41228488f0f9874b741a : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2a9158afc6d461e91097ed7e3029a49 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c15f96b133c45db9d581ff1d14ea091 : SUCCESS in 8m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ff959af5f90478a97f83a752bbb81fd : SUCCESS in 12m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a3813b075c914a3daca41c54c3ecd24f : SUCCESS in 6m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c7c3f716b5ad48f8befa8f9807664e03 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/12f4f6b997164736896435585af8c34c : SUCCESS in 5m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cda6c136637c4e62abec2b6896b72174 : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/357daf19c8974b6b959cd9e54523fb5e : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/befd6640659b4c95a418468cc0f5a08d : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e6587dcd9c714f2aae3f8d9f694c0b13 : SUCCESS in 8m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d0efbb735dbd438a890705f6fd7308e6 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c89d12e820b54a8885fb93ae1f0648e1 : SUCCESS in 9m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ba8a1898f8354f9ab5a15a6b7f83f235 : SUCCESS in 9m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/967f8b03a85046cd9965327d9793b12e : SUCCESS in 17m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ce978cb5657241cd88f185a051b9c3a1 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/13e0cacdd577433285681957ee9697ea : SUCCESS in 18m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/019f26260dec4057922b7e155720d54e : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/89071cabcdb842e6b6c856fa2f804a9f : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7c5ee067a6dc4411ab12cd6ce06fc4d3 : SUCCESS in 32m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/46c906c9bc0e466c85b9acfcab9fdc99 : SUCCESS in 11m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/31270d443ec74119b46b7bf308f64289 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db781e9c432b4878a37a16965f2a3825 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c79cdc9012f3420285d7e5602ccfec10 : SUCCESS in 23m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/453f2bb660514f3caa01f830b2e4b747 : SUCCESS in 4m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/703830f1409e4369b2db26cf941ca6eb : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82b415f0712d4060be738e36a1afc5f3 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bf907fba7ab041859b1a44ee3e39a549 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b684f51a33047c0bc20b2218f94ba42 : SUCCESS in 59m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ab013ce26e74535b87c70f188fc7d1a : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8023c5f448ffc0409ffec05900b008d48a61ce58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 19:15:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25630f8bc3f885f1f55f5e8ca7c39bf3bc056b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 22:54:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25c31d4442f429d9c7d910911f81a34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52083801b336423283cd7dd48367a196 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5185e01718451aaca63ed7970a22a9 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d163b51e8f8644a08476af65b6c78be0 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd05d89e4c4f4337b865d8ff4db5f663 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00ac6b9cc809466faa22b1c0b224fe01 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eddb24a8373848a7ad363ac0c0ed6b3b : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bef1fc0aae584a30b76708f63348f5bd : SUCCESS in 7m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/34e8ed149dbf4d57985467fab1b599c1 : SUCCESS in 11m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ec3531c1216848fda151921debe8e7f2 : SUCCESS in 11m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2097094541e145f6bcee4600ca2fe30c : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/428b4264c2714d8f8d10b7ed6b654470 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/969a0349b55e46cb8baa62d559f186e8 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fd6c752bf0b5416888f4220fbdc3b76f : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/663d5947896b4719b69b8a854a7738b6 : SUCCESS in 8m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2ac2a64e3ef242df80d801c79ad2fa1c : SUCCESS in 8m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/932124b8780d4a6888efe512f5f190a9 : SUCCESS in 10m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eadf58fb838f42958cb53186bebb3be6 : SUCCESS in 5m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a10ceaa2306046e485b94ce0e59dfdec : SUCCESS in 8m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c5ee75a757844b38a4e0547e2e7ec961 : SUCCESS in 6m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cf6891f1304b43e5b1256e1fbd872fe2 : SUCCESS in 20m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/38ed6c0f820d4d03953566b4dd288cfd : SUCCESS in 11m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/24d09410955e4b2294f11a950736a963 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/871bd8bb5480440abbcbae7818a41104 : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71ee7639f7104e1ab1522a9b0c1e35d1 : SUCCESS in 18m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8ebfabd172cc4541bc0dfcf82d9224b4 : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/87ad1f64e28546e4a0576697dc6baf3a : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee2285fad4a5458090ca002451ff86d5 : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4a9b1e3b6f02471ea3743e41c6fb7a0e : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ebfa941f8524db588498099db5fda16 : SUCCESS in 45m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4449a41172f44820a572eaa3ae8cfb4d : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e5cb5b96e9aa8f4106848e2d52385f3e40ef4f38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 22:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d89f3e6f6d9ad508bcdd903f5ed324b971808bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 22:56:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f319faf15dd94d95b9b798818c02a277\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a99890e2fa24c999705e7086a181506 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1feb1ea4bfb2818166aa64da5d1784fb577a5cf3","revisions":{"1feb1ea4bfb2818166aa64da5d1784fb577a5cf3":{"kind":"REWORK","_number":1,"created":"2023-03-02 14:30:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/876158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/876158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/876158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/876158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/876158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/876158/1"}}},"commit":{"parents":[{"commit":"4eb31299dcd912d55ec659319237bb4389ba12f1","subject":"Adding tzdata to global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eb31299dcd912d55ec659319237bb4389ba12f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-02 14:30:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-02 14:30:17.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.10.0","message":"update constraint for metalsmith to new release 1.10.0\n\nmeta: version: 1.10.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I07a53b656708ad14b426fada1bc23c46492e1a1d\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I0501b73af7079d8774be83bfe927a710a04a9647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1feb1ea4bfb2818166aa64da5d1784fb577a5cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1feb1ea4bfb2818166aa64da5d1784fb577a5cf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
