)]}'
{"id":"openstack%2Frequirements~876320","triplet_id":"openstack%2Frequirements~master~I1e7be29db9a2b52a4d1bbb5e835ba9ae34e2c3a2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1e7be29db9a2b52a4d1bbb5e835ba9ae34e2c3a2","subject":"update constraint for ceilometer to new release 20.0.0.0rc1","status":"MERGED","created":"2023-03-03 11:36:32.000000000","updated":"2023-03-03 18:49:30.000000000","submitted":"2023-03-03 18:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876320-new-release","meta_rev_id":"1015b32791501996e5438f10a1edd7d158cf8bb0","_number":876320,"virtual_id_number":876320,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 18:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-03-03 15:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-03-03 13:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-03-03 15:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 11:42:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-03 13:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-03 13:38:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-03 15:18:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"35c8bb0cb7a056376cccf77e2d6695ba3e46dd8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-03 11:36:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a166046e53028f923f0445a273dad253fb50fa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:42:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a72a9f22b5436abb21e5e29194e8b8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d303a55186fb43349c2902342c58fe5d : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"748fadcc3db56bfd230f5163779f96457b86a912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:27:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0933910b17d24467896dbbaf27d33d67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421a7e7f639146fa8f757de9bdd01627 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d1763c6648413ab97aa9bae6cbd036 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd6bd999bb394f49b6e2f9add154c10f : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f9e66d43e7a45549d115bcbace75db4 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c993b065b11e429b9c2f04668067dce6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6153ba0c16742c89b71562857578bc3 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ede26eda342b418d85dec9872bf9216c : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b5e70cf349f94f0abcd5d8567ce810fd : SUCCESS in 14m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/226fd54794c1477b97179546a7cb6687 : SUCCESS in 8m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/38326ff46db34d9d8865e9582041cefe : SUCCESS in 4m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c8c71ec99dff4a6f950ff21b73f4634f : SUCCESS in 6m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5b9abc6d693d4e7baa7ad88f4257bf51 : SUCCESS in 12m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5d7dcd1c9095417bbf2531ac1babd827 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b16c777015b04d7185980356824b063e : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e74064a5f76441b79448a5c8151653fb : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/53a7536844834c078775ecdbda6c8ddd : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a34c807ded7460faf0d614934188109 : SUCCESS in 8m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ef5642bb7d4c45fb83c06c6fdab06bf3 : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/280621834bf34ecabfd1338d700d71ae : SUCCESS in 10m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e8160efd32b540179e76049dbe028de1 : SUCCESS in 4m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9e796e79c06740f28debe7817afaabf9 : SUCCESS in 9m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ed48024134be4c14b61257146eb7d297 : SUCCESS in 4m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5cf74225afaa4f929b0ce2e4203bcc07 : SUCCESS in 10m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9e2b2a55d20d4adead91a622d00ef6b3 : SUCCESS in 28m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e7427b67c16c4955a771066a7bdba8bf : SUCCESS in 14m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/679e919d842b489dbc0cd1fd1390747d : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8252e263953b416fb9dc660da1319bf9 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0968896d46af412eaf0faa2e0aea1c6c : SUCCESS in 26m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95848c7bb0c449f2ab080d430dd370e2 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c1cc04666cba4dbbb03de43d29edb2dc : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e005407eee645eb85d6aad728585d3f : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1db07dea4e6944b89a0fc8663348e89f : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19ccdef55ca94765ac17edc6d1c65273 : SUCCESS in 58m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dc217fa6dad4b36b8700ad8fb630ef0 : FAILURE in 1h 43m 41s","accounts_in_message":[],"_revision_number":1},{"id":"af066c4c8bd21feaad29deccaddca001bdf397a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 13:38:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f6081afe9c1e3fe778ce1943d4a6b5c589d0b99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:58:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79b1e4288924ca08c9ef05aa53aa1ee\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9bc6286fd6f4418ea5edb7aff7548f20 : FAILURE in 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"835bc08e860c266c418d31dc04b3fa48eb6c1a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 14:51:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ffe1e3ee5642f9a9d86bde4dd6c774\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f36dcb93959414f845c51738b1c8ec9 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2223c47efaea4e02b9ab68c8e188cc1a : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65df20bc4bd1484faebfa1009cb3a19b : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73f56ac3dd0d4689bb5cc9b6ac4f39c7 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1efe7abe8fb643c8b847aef95b795805 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d16bc781b134ee09d223eee66dbcd96 : RETRY_LIMIT in 1m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95e192ac0ed748bdb72f36d6d32d9d68 : SUCCESS in 8m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08ebb9de591a49d48d81ebbc6528ab5d : SUCCESS in 13m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dee95dd692714c469187682093a026cd : SUCCESS in 11m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7bd4f7dff5bd45a4a53e383b81638410 : SUCCESS in 5m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9c2bb4bb9bb24ac7906ef2aa95321e14 : SUCCESS in 6m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9d4bbd063a92426da84c6a65a704d1b0 : SUCCESS in 9m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/64a1af1cae7241f7a118168178aefebd : SUCCESS in 5m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/40dd917653be414ea3e75f5b2a3a38fa : SUCCESS in 5m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d7705ea226f64497bdc9d589449456a2 : SUCCESS in 9m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4d78d12c4cc8457da43bb6219911c714 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc0c891ed3a34aaa883129b25016235b : SUCCESS in 7m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7564f0e0f4164f159f6f887516fa8288 : SUCCESS in 9m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/74f1fa2296e44279a40d1e968078b4d3 : SUCCESS in 8m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8ff60aaf49084c68b0834e8512a3040c : SUCCESS in 4m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/301cd829e8b14960b17940c033fe180c : SUCCESS in 12m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/57193329b24342fd80d6cce929bcef01 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d259b2eac00f4534807137cb9a712f96 : SUCCESS in 10m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4d72a01f4ce14cf48581aee1e687f971 : SUCCESS in 25m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aa073093b54b4d519760541a32860263 : SUCCESS in 13m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d5ef63aaed7f48398db2b68a0ec4cbf3 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5f04f3abae14b83ae421c584599687a : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcfc2ed1dc6f4b6aa5d15515621c0b75 : SUCCESS in 22m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/26fa4752418e4733a3e3dd1b3bc774dd : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e78b4e32613d40c48dc895a7cdc11902 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea474872643243e9928c6987b3affccd : SUCCESS in 5m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/75c197e460364500aabb893f4b79ee57 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb102caf0adc43618cb64e6152cb2039 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2283bddd70d84921bee2650763d8a0d0 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"961222d9749d994eed9097928c5bfc6c57eb2779","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-03 15:18:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4c328bb9ff6438e0e8103a270ac33c49f84702c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 15:24:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ab6098026b64c8899a3a130c893ce86\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e93b35b756464c138d23ead52dd983a9 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"def5220b03a263bde0af2db98c0557dccd646b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 16:57:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ddef4ae6634aea8bfbf0853b179eaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c72af89f0e0045dbb81ea41000909e57 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6707b6065e45f5ac9e02b81ded7458 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90cb62ccfab44931a2ba77e78e6c2b11 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/629bbf4d421f43ebb5a8e538ba656fb4 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8caebd42b8b94e9bb0e25a751343d229 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85f6707bf70e499b8fd30e5351d1fe07 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ffd2c0416fc4452b4ceebb70bed7fb5 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/53931aa49a8e492987b517049e4967df : SUCCESS in 14m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7253dce4019549748b9952300a515c4b : SUCCESS in 9m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/99dbbea9a601460ab838badfd0407b92 : SUCCESS in 5m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/00054d3d33294c529f6c79ad8777dc1a : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/52bb17284f5f4a99aab420a6bf9e3d97 : SUCCESS in 10m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e98d984b69f54a00958d79c6ccd6cbaa : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5276e314c18440099512e90c2de4c12b : SUCCESS in 4m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bbad45cd5c1f4aa585c838d0c2bf667e : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bd285c2c2d0d423ca268637a4d7d73da : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a07805a843f4f6e9b11e1ab836c879d : SUCCESS in 10m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/989c5c9fe22c4a82b6b9ae6063dac0ac : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0e8d95e065f3429ea31975460972cf7b : SUCCESS in 13m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c7f1d6cac1934f4f89214afcf0b13a3f : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2fa52ef07f974f77abda0430199486bb : SUCCESS in 13m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b25ff5c433c84adfb161f22f8ad1be0c : SUCCESS in 4m 05s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/37e20f901fe2451bbffcfdebe00c7d74 : SUCCESS in 9m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5336798281b94e808857c215f18e7ee6 : SUCCESS in 24m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a60c4bfe695b48bc9da7433c53ba8b26 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/55360c36671e4485bea93aafccc00b75 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1fc4367de9ee4982a9261fcfaf02507b : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8442e89073ee46b385e3431b0550f1da : SUCCESS in 22m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/537bf9a3038d452591e1f689c195d8d7 : SUCCESS in 4m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fe06586982ad4ccc8cef034d678bfea2 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1957d57963ec4549aa2e7b1dc82c5d48 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5dae08ba325947c0ba59b5fa92b65812 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39ee8fdeef284ba294041bb626246fa6 : SUCCESS in 54m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f5977621bf34cbba775262e42b14992 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5596a5e0a1f59003088db7723082a25bf05d37c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 16:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3665ffbe8a4a0f19998a3bc25a3f271d3677a584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 18:47:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16c6c24f7b944629f983556f3fb3533\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15f7e44d713452caccde338a3a18d29 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7cfa5fd4b204b239a16cecacfda34da : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd96e1a7ab5f4b4a8cb2b19d886f27b3 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7d983b725414368bbfa99efde853761 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96e426753c5246a58e0c0b03b87b693b : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/111c3ce809b74189a181704c694561de : SUCCESS in 4m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80f0a39d5faa40068d701447d3478083 : SUCCESS in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/656f274f2123452cb8acf178ef865814 : SUCCESS in 13m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/003b1a73d2784b2d95c53d9654580fc4 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4e021bb460d842978870e668f07d851c : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8a753ded01224979bd0ad260a380e2cd : SUCCESS in 6m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8034ecbb37764741a8146a0a9cbfe315 : SUCCESS in 8m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f2b83a2553044ca9ab4de220f9376c67 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0052a6506b7441a9977c448e9af70a1 : SUCCESS in 8m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/06117fc7a3f34eb498207c11b5bcf45b : SUCCESS in 7m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/def4011d597645449d01e1584556bdba : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6e10ff0d39114f50812e15f618dcaa61 : SUCCESS in 4m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e13802d17829453281234053532d4ec4 : SUCCESS in 10m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0849ba161d0540e384ca2d3353478e7f : SUCCESS in 4m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/74f93e1605f94655ab3dad6f8144f47d : SUCCESS in 23m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/100a9998c9404cea8490bd4831553670 : SUCCESS in 12m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0325d35525a5481b8fe43e759707446b : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e489d6bc94de4044a46fdc6708e4fce7 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba81a821ff14449dab24cb206f17b4ad : SUCCESS in 24m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3ed434fc8fd64d81b0c8d2742edccd97 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4ae19d1bd63f47ec9c8aa4fba360e2c2 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4db972b7aa08485bb102d724afa4ec9d : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ac6ae56913c7406ebdc7e8b8d031faf6 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3391fd5fe354e95b8aabad2a8f583b9 : SUCCESS in 1h 03m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a6a3e82b95f4847aa1410b36e41a6da : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2de3707f234e703dd38b44df838bd8809857185f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 18:47:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1015b32791501996e5438f10a1edd7d158cf8bb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 18:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9427b6843034593a41a2e4669b144ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0597fc854ae6489f86cd515d3fabd4c2 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b3cf95bd51b427d101a6b6477be0e63f9280009","revisions":{"5b3cf95bd51b427d101a6b6477be0e63f9280009":{"kind":"REWORK","_number":1,"created":"2023-03-03 11:36:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/876320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/876320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/876320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/876320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/876320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/876320/1"}}},"commit":{"parents":[{"commit":"5159bbeb19011c9d27cf163b80225c5aec66433b","subject":"Merge \"update constraint for metalsmith to new release 1.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5159bbeb19011c9d27cf163b80225c5aec66433b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-03 11:36:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-03 11:36:31.000000000","tz":0},"subject":"update constraint for ceilometer to new release 20.0.0.0rc1","message":"update constraint for ceilometer to new release 20.0.0.0rc1\n\nmeta: version: 20.0.0.0rc1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release candidate\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I2ee34c05776f83a2223858d2c342e5042c384a15\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1e7be29db9a2b52a4d1bbb5e835ba9ae34e2c3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b3cf95bd51b427d101a6b6477be0e63f9280009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b3cf95bd51b427d101a6b6477be0e63f9280009"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
