)]}'
{"id":"openstack%2Frequirements~876405","triplet_id":"openstack%2Frequirements~master~I84d9698845b092bbed050fc5636908c13ff08a5a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I84d9698845b092bbed050fc5636908c13ff08a5a","subject":"update constraint for monasca-statsd to new release 2.6.0","status":"MERGED","created":"2023-03-03 16:22:42.000000000","updated":"2023-03-04 19:48:47.000000000","submitted":"2023-03-04 19:47:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876405-new-release","meta_rev_id":"f5d6dfa269fd25c3a04168b32e0e864f7218a86d","_number":876405,"virtual_id_number":876405,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-04 19:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-03-04 16:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-03-03 17:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-03-04 16:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 16:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-03 17:25:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-03 17:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-04 16:17:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b760090645428b0ae1e0585d88c6c21187c4b635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-03 16:22:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48985246463ef8b371758806148a0ae4464602c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 16:28:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ac76e5ad204296908da4633e953c4c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb44b52b274e40d3967e87fb243dc04a : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"615393502e19e0cd3a88cf21c8d4f8b86afae4c3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-03 17:25:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1c3831042ca032c382b4b59a3978bf72eafe2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 17:25:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb23b0701aad488d9bdbe51580e34b02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd57bc3324240b19dd41013d4eae21f : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5491a1116114452acbbd69d997e98db : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53d9bc25c61a458db47f4f2be75f47bd : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cf2e9a122cf4ab0a3bb7ea8024713ff : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4876f5a50324b08aeba70b63ef8100d : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3780445aad6b4ffcb4591f8c69b923a0 : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0b9b6b7eaa24078b1f6302e65adac85 : SUCCESS in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c45b0939df4843cf80f5fb141b3e7476 : SUCCESS in 12m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/805e0aa5c17943379099f99edb0f81ad : SUCCESS in 10m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f392ce4be12d47e2898364424940181a : SUCCESS in 6m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9a049d80c5e44e03afe4af5322b8f46d : SUCCESS in 5m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/84752d21ca11494d9def0ae30fe4eadc : SUCCESS in 10m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4e550736b83547878ee31842f0362a05 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/75278b0725fb4431a1a8ac4d7b9b09c2 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d02a7971f9d041a5904625d9da6967f2 : SUCCESS in 9m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3af1fe0139004c64af44089a212d481a : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fec04e40ad634416af9eb681883b4106 : SUCCESS in 9m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2533189ca4c8444e85d7ebc8da7a488c : SUCCESS in 8m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e9df5fc8defa4c52a7cb056c8ee96ee0 : SUCCESS in 10m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/00e38760facb49dcb29038b98d36e397 : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8241819730874f09b2a703e474d976ce : SUCCESS in 8m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fcea492927de42fab9358f926e85a2dc : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9af5d34cbc824536bee8e08d58e5be81 : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/85f6a3bc72df4f80869bf4e68dcbb4af : SUCCESS in 20m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7e9f77f6f79b424eb099f67cbf7a367f : SUCCESS in 14m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f7cbac384f6b4f96abb5f69d441d0960 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37bd52b9cc6e47658d0cf486cccc01d3 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0735f2ff407b48b1b8855385c1c36792 : SUCCESS in 22m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/210c8e4e4cea4657a5d63b7124cbf87c : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0b4fa9895cee4c5989b393ec1dd01293 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6516f6eadb24589a66a7fd469b018ba : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad3401d6ad5e4992b978e98a9ba748df : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8cbdf32275b42c6b07952d08c7a773b : SUCCESS in 56m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92c29e02e0ba41b88bcf60ea18b491f8 : FAILURE in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0036baca8d3a6a67fa06838b01c795785c58e68b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-04 16:17:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d103a8524b91b6d7d8eeed0fd8bee853dd39be7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 16:38:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66aff269621c461999195ab0309f9de2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/68b39aacd3e84d9ea755b7841924784a : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c79f5897de9afbd80e3d583f7f4561d55a7dc4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 18:20:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e0238750474d12b9d43aaa57dc9e8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df98405d61eb4114834cea27cfd9948d : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fd227ea6ad41c7b0a39f5c5dcbeaa9 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c56be9f935e4b5998c3b06a35771148 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3fa6291344c4dca8db459471da60da4 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4db33367d1e4e51bf34aac035e15ed9 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c00049163984f788978fbc01aa4076f : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78c6dea08a1d443ba1ddf16d898d5188 : SUCCESS in 5m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/826e4bd82c0c4009bff15614d023860c : SUCCESS in 11m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/02106ba2a1264833a932f3464e02d0ef : SUCCESS in 6m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/73d81e761a9f4e3783ffa26f493254b6 : SUCCESS in 5m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bc6e6715796445119569641dfaf033bf : SUCCESS in 6m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8c1fca9486014cc09c45fb2593f37469 : SUCCESS in 12m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/99db98115c904bc7ba71e7be57176847 : SUCCESS in 5m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a00eb8453d46496c9d7258be9d61389f : SUCCESS in 4m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c8da5024859a4567ae915719da65c40f : SUCCESS in 7m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f4587f77ee6d4cb2b4646d770f04ac17 : SUCCESS in 7m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90a015f9751748ff804de913dd87f880 : SUCCESS in 5m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/577f4cd56d7d49ae9d9826a6da3f1c7e : SUCCESS in 7m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/92b9df82dd9a4d7caffb6cddd9748239 : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ef6a1e3048ab492b8f21125d52b51d45 : SUCCESS in 3m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/46afb5982ba14371bb02bd0bf61e713c : SUCCESS in 9m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/37e4648ff2f54ceeab4a2dfbb4d8ca29 : SUCCESS in 3m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/54ecfeaa6c554f8696495b6b42fed613 : SUCCESS in 8m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8076135978684d55a0560fc94749d406 : SUCCESS in 19m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/82aa4fdfc05743e4af0a9e0b30c66e52 : SUCCESS in 11m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4ca87ca1b8394aec929b3ff9c3842aa0 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a84883dc6fa74154a1b2021d5dd1514d : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12ce370db254401d8f1ecdfe31348c59 : SUCCESS in 26m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/38eb786a34994afa9316f98336d934f7 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2e91aed895d246cab368b96a6c62f88b : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39edb088bacb492c8568dfd22fdc6955 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bfb82ff981e44b38955da41ecd62ba2f : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a215ef070f1141eaa394669c77b8ce0c : SUCCESS in 42m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/791f9eee0fee4c7cb26fb37cfe46d796 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5eb5396dca41c1a10fab46221a610430c9e31efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 18:20:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3337745c9668774ef10790ad8d91af067c7b4a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 19:47:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93754b2bcbca4e9f8829f791048fbc3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0639fb71eac4e84b0ab63eebfeeaebd : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4723c338475e46e997564d78a0b4cfb4 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/435804055cbd49beaa9a339e50dfc3e9 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cc96c6582904809a053fc274221e0ce : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ce129e728144bada73966e0bbb448a3 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9af94106b5b422f943c5e6f75497559 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea154a5214434f80a68cd05a92a02ee5 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ff674b5a4be446ea5a86272693a6fe6 : SUCCESS in 14m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e036ab25e3164ac18f680cd200c3fc48 : SUCCESS in 11m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7073c29311884254b08da3f681036cdb : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/60efde844dbe4040886a2e7158c13709 : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4a10b8ee37f644149579004401f63422 : SUCCESS in 8m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/35969c7956604346912c9dd73336e3a8 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45bd7a291b50448dbddde6cd51aab3d6 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3def9a5ee7c94849b8961f397454d4a3 : SUCCESS in 8m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/cd9a4a33e9a8433ca8647e5e1a3499d8 : SUCCESS in 11m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a5f1cab6abe54c4dbad66b3d81dd9a8c : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/435eac9214c349b8b11e48e68d9642cc : SUCCESS in 9m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e32c7212976447abb91a6c888da37919 : SUCCESS in 4m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9793386e666440edafea7fe7dc2a8680 : SUCCESS in 24m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/da59fbdb36f240d3bb61c57fe5426e47 : SUCCESS in 14m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d94c6acb2f314c8e891c70a0aec16539 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/390277e1b14f4d9eb8e47d91bd60bc02 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/095d921552cb4736b04736bbc92dba62 : SUCCESS in 18m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f2c0720b85014fc38f9077c32537f7cd : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f233f95cdf914024a2c926978014aca3 : SUCCESS in 3m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c02a681197824109bffaf56671640509 : SUCCESS in 5m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f77ad8efc83b45459f99fc2a6502f643 : SUCCESS in 8m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47b39121e7c544d4bbba7e1bfc2303be : SUCCESS in 37m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d5dcb705dd74415ab3528760b335846 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"142fff4223fa5a451fe97d7e9f7517cfb6bdb784","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 19:47:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5d6dfa269fd25c3a04168b32e0e864f7218a86d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 19:48:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0106bc902f146c68c07a8769bf62a6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f0c008a82047a5bc73f0d166b5a42d : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a920c5d9a4de723a317699cd081588afafd1f962","revisions":{"a920c5d9a4de723a317699cd081588afafd1f962":{"kind":"REWORK","_number":1,"created":"2023-03-03 16:22:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/876405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/876405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/876405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/876405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/876405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/876405/1"}}},"commit":{"parents":[{"commit":"5159bbeb19011c9d27cf163b80225c5aec66433b","subject":"Merge \"update constraint for metalsmith to new release 1.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5159bbeb19011c9d27cf163b80225c5aec66433b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-03 16:22:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-03 16:22:41.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 2.6.0","message":"update constraint for monasca-statsd to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib5299bdcf5ed36680dad3e28e0affc9cbacaf2d3\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I84d9698845b092bbed050fc5636908c13ff08a5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a920c5d9a4de723a317699cd081588afafd1f962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a920c5d9a4de723a317699cd081588afafd1f962"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
