)]}'
{"id":"openstack%2Frequirements~877348","triplet_id":"openstack%2Frequirements~stable%2Fxena~Icf81c6a590fb43348f3c08d32bdda12d5fa066ba","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Icf81c6a590fb43348f3c08d32bdda12d5fa066ba","subject":"update constraint for neutron to new release 19.6.0","status":"MERGED","created":"2023-03-14 09:29:04.000000000","updated":"2023-03-14 17:57:47.000000000","submitted":"2023-03-14 17:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877348-new-release","meta_rev_id":"a9627ed65a46bdffde2881839f82c736bbce71e1","_number":877348,"virtual_id_number":877348,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-14 17:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-03-14 14:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-03-14 14:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-03-14 09:58:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-03-14 14:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-14 09:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-14 09:58:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-14 10:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-14 14:01:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-03-14 14:59:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c97005f25c1bff0443f8e80eaf6a4db52d1e095b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-14 09:29:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37719ab829f08c74beaf80ec8c10b487d0aa909d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 09:51:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1d08fc9c634496a0b682be718c606b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/92a3307354014a3d809216d935e3f7c9 : FAILURE in 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"75f7e817b3edcbd95887ff82c22be2d72f942135","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-14 09:58:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9919e240c084bc5f314a05dae305db568198ff98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 10:48:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2da6921548f4d11901f222131093ca5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ad92f255bd4d88b08de73fca1b1206 : SUCCESS in 2m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161e00a59fba4bcd8a5762839ccc098f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/945ebde9d545404989e60a977efea6c9 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0088ad36ab804961af71844d0cd2e360 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0a48043a25e4a0eb3fd9a74bd514a8a : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3d57171786a44b1ac51068b80c0e1f5 : SUCCESS in 2m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/746695f9de8c4aa0bf23fae62c7ab958 : SUCCESS in 9m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c67ba2bb39e949beb36cf97f87bd94c0 : SUCCESS in 9m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c8d0f1d29c3d46328adc502757f39763 : SUCCESS in 4m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a2d66c88253440b29eba16e0ed1c22b2 : SUCCESS in 9m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/55a0cb41cedb4836bec9fd697da9b02d : SUCCESS in 5m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e9237a8a60a24f38b53372325db3fe4f : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e3556bccdde40b492e9d92f61ffe08b : SUCCESS in 6m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/38748d4578d44a2ba74456f3cd9e4a69 : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9e45e9ae26714282b261385d57283352 : SUCCESS in 15m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5fc72c4e198a4665a189dd44ab5550ee : SUCCESS in 8m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1f6b9316735a4c9b89aecd3fea195d5c : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b721f1d54aa1499f93c95d9db6ef293a : SUCCESS in 25m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9cd0d3f37481452fb94ba2130dad9d1e : SUCCESS in 10m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5ff4399b00764f59ad41b3d35b61cfd3 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7bf487cc090140e3b6b4b32a471667a8 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05e9cb2f180d4d9e901d1ff3010c2316 : SUCCESS in 18m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0cdeca68019c461184ee3c6244df18e5 : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1ace9450aa3f480da879e10e9eb1e386 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/171ec2dbfdea4ddeae9703550bc7b6e9 : SUCCESS in 5m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a4e5626292524ea28ec036d4b7a201b8 : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0862ea947cb84dfbaa66f60735e3ab12 : SUCCESS in 46m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84a015ab3a4e48a387fefae5681cdded : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"44df485efc3f73ac0901e6b5c7732f0fac9cfa51","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-14 14:01:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08573c36b8cfcbe205c2ad618c50f3530e9a25dd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-14 14:59:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d981f8332924d6cf7dab7fcb7ad8eb621d9ad170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 15:00:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15ca086cc40e3a38e1e4187a66c0cccdfd5447f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 17:57:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338f0bb3806244f288d07e9a427e4af5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9762f2d9177b4034b96c626d5183753b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83b4da3193e74ebb971ff57b71153e8e : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71aa00605a9c47229ebf452c3b02d6c0 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21a45d16c43c4fa78ac01b5990f0ea51 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4eae934c67d14b6e897da3ed62474101 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/911e9dcf54cc4c4a9adeff065185f2c9 : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0de13c8280704b63a8c302b6889bda2f : SUCCESS in 7m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/960e6b194d8c4302a00ea986e02dc986 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/38d85ad0f9c448b2a8c1493c0cecbb9d : SUCCESS in 5m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9fca523d1994ed3ab31bda641923d5b : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/baf65297bae244639d53ddeb260c0391 : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7cbab7b4a34f4bf48b614b5ce32bf599 : SUCCESS in 7m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c50f814ac9054732b591e36a85fa2291 : SUCCESS in 13m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/32d669043be14e52ae502c59f38ef48c : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cd916bb3debb40ec95dffbaf2e470a8d : SUCCESS in 4m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3d4ce583e49f4e6a83382cd0b70913ad : SUCCESS in 22m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f422ecdf1443484ba433403199d3fdc1 : SUCCESS in 11m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/869cff1796d7475baf027ab7ff743b0c : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59b678ff4f614be39233d8a349edb7a5 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/057b43af9bf14fca970e379f83fd8c78 : SUCCESS in 19m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/891d715d8c154a1b869d015d1e61b349 : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9dcada5652604a5c8d297e86ffd28b34 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad9e7cae7b7d48c399742c39b94401c7 : SUCCESS in 8m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be9c2f3cd9b748c78feacd8f5c6baaf4 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72da15d2f7a749849a4fb9d66be98ec1 : SUCCESS in 53m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89e7e12778234945a94994d299d51c32 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a9627ed65a46bdffde2881839f82c736bbce71e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 17:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45e3947967109cc945fbde7bb149ff701f7bb871","revisions":{"45e3947967109cc945fbde7bb149ff701f7bb871":{"kind":"REWORK","_number":1,"created":"2023-03-14 09:29:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/877348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/877348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/877348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/877348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/877348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/877348/1"}}},"commit":{"parents":[{"commit":"e1521660a99a56caa9623488b017b7afd7b47797","subject":"Merge \"update constraint for neutron to new release 19.5.0\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1521660a99a56caa9623488b017b7afd7b47797"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:29:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:29:03.000000000","tz":0},"subject":"update constraint for neutron to new release 19.6.0","message":"update constraint for neutron to new release 19.6.0\n\nmeta: version: 19.6.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I2fdf7d36e0438ba400ddf07f51002b1dcc329517\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Luis Tomas Bolivar \u003cltomasbo@redhat.com\u003e\nmeta: release:Code-Review+1: Fernando Royo \u003cfroyo@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Icf81c6a590fb43348f3c08d32bdda12d5fa066ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e3947967109cc945fbde7bb149ff701f7bb871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e3947967109cc945fbde7bb149ff701f7bb871"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
