)]}'
{"id":"openstack%2Frequirements~877351","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I4165174b21cc84cc95022cc3a26b2c7886684a49","project":"openstack/requirements","branch":"stable/yoga","topic":"yoga_neutron_20.3.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-15 19:36:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4165174b21cc84cc95022cc3a26b2c7886684a49","subject":"update constraint for neutron to new release 20.3.0","status":"MERGED","created":"2023-03-14 09:40:35.000000000","updated":"2023-03-15 19:36:10.000000000","submitted":"2023-03-15 19:36:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877351-yoga_neutron_20.3.0","meta_rev_id":"a00128dce06b4deb1484ffc24f3b5b6ee5cef61c","_number":877351,"virtual_id_number":877351,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-15 19:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-03-15 18:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-03-15 18:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-14 09:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-14 09:57:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-14 11:17:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-14 18:00:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-03-15 10:53:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 18:12:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8497576af61f999e95022dc8b2666e8136ab78cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-14 09:40:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddf62bf6076a660c5556faa42d7f4ef381209db7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 09:56:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6585c7717a4417bc75aa0583796a9d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f5cae1df9d0a4e10ac468f860bec2016 : FAILURE in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dbdc8e415a635ac686ada9667e5a041184953172","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-14 09:57:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"595854182ae7a0c61fc189a83b858adfbf492adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 11:17:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d1d2ed48156447f8697f39998cd51eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c409751e5c2439999e58aaeb6752311 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e703789c1aa4caab8420eaa7dbeda4c : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a795a8e2b5a64bb0ae92e85804a11a4e : FAILURE in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e692ea9e59cd4ab1a7ca8abede3adcd7 : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3053eff9c67493488fa8291f68ef522 : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e9a828d38ef48f78e1cff9ad8e86e8f : SUCCESS in 9m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f9a30c0a76340668ead8870ddb854c7 : SUCCESS in 10m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d4024cf0ff31479daa397acd4b02f665 : SUCCESS in 6m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a9834792baec4d95b0c6f8a3e63bef2a : SUCCESS in 12m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a5a39e64ddd04ff9b404bc7c1cbe018d : SUCCESS in 6m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cd961206d56c420497e3492ec70b2891 : SUCCESS in 4m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/35d996b4208142c5be8246ffddcc2886 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12c1f433ca8e4e4187c9915bfd4075f2 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3895e0a4e45c4a5a95faa6c1c14bfbee : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7823d1dec9c44ed1aad1794a2d09310c : SUCCESS in 12m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/df11c92fa8a44e4fbd1cca08c950ac74 : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e1cb413c43444dd0915a4feedc07ea7e : SUCCESS in 9m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1081d636a8c04fafb3d4e2517d81c88d : SUCCESS in 4m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5f74e53037384b65add90aa27e9e73d3 : SUCCESS in 23m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ce4e4d511f7e4c598b4962e699e77e72 : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca07475031a048dbb2c8ef04f6f4d920 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/165af84099324aff9ebe0aac7296ebcd : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b62799e713c489a9b4797de2c725218 : SUCCESS in 21m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/22704ed3cf3b4f078d6c6e577e647878 : SUCCESS in 5m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b2f6ced59ba347769e92c5a1d3dc8686 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df20719252e8433bb60ad591499ecbe0 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a6a61a6ccef040e0bb6c7db5aa5e7524 : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d97e64afeec4d0caf82b006ef095741 : SUCCESS in 1h 06m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f1ccaf44fd5429c9b5fc5dc1e620cfe : SUCCESS in 1h 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"45ced05a433b5b5336866af6e4781041d15a8b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-14 14:55:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4530446ac6f4ff251c75be47d5b20fd3a2408c75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 15:10:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe22dcf2d7341b8928241745e376ae0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57f07d2dcb3246cb9bc3dda101a5aad1 : FAILURE in 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"bb7cfdda2fa06d5e2b1f778aacb177b8e505f3c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 16:28:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e62d038d21641cd91ec930c561c3332\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec378171852497fb27796fa1d416bfe : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c971664c4b7349279b1d1951b83fa678 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843b9f2d6c3d42adb94bc57304798638 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b84004506e67485ea31611d7b8dd5dc8 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8986f102c97145a182736cddc6e21e98 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7c083513c4354897b3cbd5ab89b3a8ae : SUCCESS in 17m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c20dd25c55e4191ad177acdd8bab020 : SUCCESS in 8m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dc3d1493dffe4bb9a4ff7bbe14db6ce4 : SUCCESS in 4m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/55af2c82f3e3436e8a7cecb2a26244b5 : SUCCESS in 10m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bdcb415a0e2141b68447ec5263fcb093 : SUCCESS in 3m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/511bc08ad62b430b8c645322a4aa99a3 : SUCCESS in 4m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/167f6fca9922446f990363d6039913e1 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e2f71a9182c4fc09a74813949873138 : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91f95877ac8b4b438453028d5042e950 : SUCCESS in 8m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f27ddc4756904062a819b008726f61fb : SUCCESS in 12m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0d308446dfec45eeaef4cf4753543f63 : SUCCESS in 3m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dd3b4c2c179340078aa40ebd3860ec80 : SUCCESS in 12m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fe4670e2e242451a9bf0f2b1909a2f89 : SUCCESS in 4m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/289512536faf47c09fda2abd6243f0c9 : SUCCESS in 25m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f75db46198864dcabf28ed14504c5619 : SUCCESS in 12m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1f2575e4b28445a7b110768638555e42 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9d21e686881428da816116f27698921 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19680091720e45188653962d062c3bb6 : SUCCESS in 22m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7718be03e1d84e2a9756927b6390c7de : SUCCESS in 4m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ba9ddcebff654ff1bb6f9c8179235374 : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d095463cfb7143cba8a8ade670afe9c0 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/55ac0253df4b4dd7914b2669785078e8 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9522ccda3849431c8b747c068a5e83ca : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b65326613cc54b71a6efac59404edce9 : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f1c50e24016c74f4890f552c06ecf75a05fd4d20","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-14 18:00:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbf31f9df0517013bea7b04b714867b2a14bc131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 19:54:23.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 877351,2","accounts_in_message":[],"_revision_number":2},{"id":"3ecbc29d81a6cb1c2a7e20e42823fd03e2bae4e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 10:52:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9db9bccdd18711c29f4b36bd107ecfa1952f9e96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 10:52:54.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 877351,3","accounts_in_message":[],"_revision_number":3},{"id":"bcc19d1e3b980b845ff3f55abcb30097e544d24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 10:52:55.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 877351,3","accounts_in_message":[],"_revision_number":3},{"id":"3973fa3f592ae1de5a362865ef1de28a2eb4b5c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-15 10:53:06.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 877351,3","accounts_in_message":[],"_revision_number":3},{"id":"0edd1c923e0083253e442dde3aae273c58fb031e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 14:56:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1eacd52c377c92edc2e4f56a03b8b327b51c3bfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 15:06:00.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff872056f474a6c9e14665baa7b3c99\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13aaaf50e91b4c61a3c2306d547f9c8e : FAILURE in 8m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f290ace14e83c7773408430783f1af228b4777aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 16:43:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63439edde634e6d930b243e2ffe8553\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27df5bd1908b428bae90ec12e2ab1203 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac62929bd2d42f8a51725f633ac66f0 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c52f275b7e4454be61d80dab8a048d : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30a4a3326140471e985edf6a9429ccd1 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c3e37c25bf24ba992ed461bb408438d : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efc290fb4fd04775bd1be04e5c312e93 : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a57d02702b064417a54d77812cc774f2 : SUCCESS in 10m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/67d0790afefd4a4ab336051f55f963c4 : SUCCESS in 5m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9d0082fe9f684b578957a280bcbf33f2 : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1b6c0e42363a49e4b709087fed18dd55 : SUCCESS in 4m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0e428110f46a4e2d821ee05db384b7ee : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/27d5e4c74da94ee3a8bed0b61f465a38 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f142418d78c4710a56aa357d315fd19 : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a54a8f6eed144f51aa1a995d7ea2872b : SUCCESS in 8m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/39242ae095c84588b8b7ec7b8c524dbc : SUCCESS in 10m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8d3725de1de94b1a96d24199e13ac63f : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f3e67abb83c046e4bbdb4402eedc2d32 : SUCCESS in 10m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fc6dbc041fe740948cbfed95e5d3405b : SUCCESS in 5m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6d19bf26ca38409fa7bfe8ff63713041 : SUCCESS in 24m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b839eace1de04eadbc16a2a3d8938d82 : SUCCESS in 13m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf01577fd7c945c3aa5cf78849ba16d9 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6991945d4bff42e7b2c48c1f49556e85 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f6f1715dc584366b1b6828eefb4d82d : SUCCESS in 19m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62f5008383944f8997abf34563db0b3b : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dc2a66b7d6be480ea6ca3603796dd7fb : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6cd0a23749b9403ab0aba6697d385d5b : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/332a025d1b014fc5ba988d6f98c86d6e : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dde7a3a8d0bc464abd168bd060734aa5 : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72b27cc4533645c89c51cb3417ceb9ba : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":4},{"id":"af989fcb7b7a7fe0dfb72c35cca074b71e83200e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-15 18:12:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5562f5cffa97374e9ffdad2ab1405074ef5ed56f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 18:12:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e257e089a40cf6b78457710b70c93ffd2d082955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 19:36:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9ffe505461486198775a9e710e7ca8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfe3cd660f54ce38105edd4f0811adf : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f1e5eee3d384497b2118ce9feebe6fd : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18dc3dc877b64462a1bdee62f366e48d : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61610883b0ec4415ab1f805bb465eae3 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b666a2041e1b4b5f80798b2653e0aafa : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/507d18cf49a347f888bafb116c0f4256 : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a7eae9dcf3b46648521c4430197bee5 : SUCCESS in 7m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8a7acecc77114580b1a40831c435c148 : SUCCESS in 9m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3123f48ac66d41168a874343e54ddcc8 : SUCCESS in 5m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a5c350a1f804c74ad623dda4b0833a1 : SUCCESS in 5m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d16c5d2eb6c04e26bc6bff8464ab173f : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b260303b78d84e0487a220ccb8de41a9 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eeed7bef573d456fa68e18135885864d : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b6aea16c415a481fbe1fb913894a3ac9 : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7d43d654bf584c3ab3e82f0525cc9408 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d3742ec235324c08b68db33955c6da87 : SUCCESS in 8m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cb4e55e51a254edaab260150231e9cfd : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/41cd36e216bf49aca0003ec8b7ab7698 : SUCCESS in 25m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9a8c5ecb5c454b76a0c0389beed7b7dc : SUCCESS in 13m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a2fa12fde6e54320a33599667e36ec6a : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/293e2895e2e84da1817f684e8b2b9fd5 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d740c9b846cf43f48a3b1c1f42f7055f : SUCCESS in 28m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b92512d806c246158c5ed8671f4e715d : SUCCESS in 5m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bab294fb89db4124920a3fdca715d33e : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d20b3d09a524525a0aeb1d016611b4e : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/89cdb8efe40a4e2a9b874651cab314e4 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b19cc4bd01ac4af2ac683cfa71d1459d : SUCCESS in 59m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/acb29a6206864ce09cf6513526841c70 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":4},{"id":"a00128dce06b4deb1484ffc24f3b5b6ee5cef61c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 19:36:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78fd28d70df2ce6d3a91081ae1fbc35cc74651de","revisions":{"198b24d276daaa60db0a4f27868376bb5708716b":{"kind":"REWORK","_number":1,"created":"2023-03-14 09:40:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/877351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/877351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/877351/1"}}},"commit":{"parents":[{"commit":"707da594d13b7f1ed47800660aaa2a8a804689df","subject":"Use correct nodeset for requirement-checks job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707da594d13b7f1ed47800660aaa2a8a804689df"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:40:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:40:33.000000000","tz":0},"subject":"update constraint for neutron to new release 20.3.0","message":"update constraint for neutron to new release 20.3.0\n\nmeta: version: 20.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I02f6fbba555e4e844e70bf42cfea29e06b84c2bf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4165174b21cc84cc95022cc3a26b2c7886684a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/198b24d276daaa60db0a4f27868376bb5708716b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/198b24d276daaa60db0a4f27868376bb5708716b"}]},"branch":"refs/heads/stable/yoga"},"54ff373d49bc0e0a5939b33adf55aee42e2a28ed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-14 14:55:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/877351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/877351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/877351/2"}}},"commit":{"parents":[{"commit":"474ac63f12bd21d731a7c1c7453580971c3e667e","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/474ac63f12bd21d731a7c1c7453580971c3e667e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:40:33.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-14 14:55:46.000000000","tz":0},"subject":"update constraint for neutron to new release 20.3.0","message":"update constraint for neutron to new release 20.3.0\n\nmeta: version: 20.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I02f6fbba555e4e844e70bf42cfea29e06b84c2bf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4165174b21cc84cc95022cc3a26b2c7886684a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54ff373d49bc0e0a5939b33adf55aee42e2a28ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54ff373d49bc0e0a5939b33adf55aee42e2a28ed"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"3f8ddee114814a1a5a2b5f744626bc70f1095f0f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-15 10:52:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/877351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/877351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/877351/3"}}},"commit":{"parents":[{"commit":"571b86756fc09cae10c0571210ceacb1d6a6d8d1","subject":"update constraint for ovsdbapp to new release 1.15.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/571b86756fc09cae10c0571210ceacb1d6a6d8d1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:40:33.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-15 10:52:37.000000000","tz":0},"subject":"update constraint for neutron to new release 20.3.0","message":"update constraint for neutron to new release 20.3.0\n\nmeta: version: 20.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I02f6fbba555e4e844e70bf42cfea29e06b84c2bf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4165174b21cc84cc95022cc3a26b2c7886684a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f8ddee114814a1a5a2b5f744626bc70f1095f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f8ddee114814a1a5a2b5f744626bc70f1095f0f"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"78fd28d70df2ce6d3a91081ae1fbc35cc74651de":{"kind":"REWORK","_number":4,"created":"2023-03-15 14:56:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/877351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/877351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/877351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/877351/4"}}},"commit":{"parents":[{"commit":"9b1e004c7379bbac3a27c24ca89e9990c37c8828","subject":"Merge \"update constraint for ovsdbapp to new release 1.15.3\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b1e004c7379bbac3a27c24ca89e9990c37c8828"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 09:40:33.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 03:40:04.000000000","tz":60},"subject":"update constraint for neutron to new release 20.3.0","message":"update constraint for neutron to new release 20.3.0\n\nmeta: version: 20.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I02f6fbba555e4e844e70bf42cfea29e06b84c2bf\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4165174b21cc84cc95022cc3a26b2c7886684a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78fd28d70df2ce6d3a91081ae1fbc35cc74651de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78fd28d70df2ce6d3a91081ae1fbc35cc74651de"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
