)]}'
{"id":"openstack%2Frequirements~877369","triplet_id":"openstack%2Frequirements~stable%2Fzed~Ie762d145d5ad6949119fc03a87e528e0d8928719","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-15 14:31:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie762d145d5ad6949119fc03a87e528e0d8928719","subject":"update constraint for ovsdbapp to new release 2.1.1","status":"MERGED","created":"2023-03-14 12:28:51.000000000","updated":"2023-03-15 14:31:44.000000000","submitted":"2023-03-15 14:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877369-new-release","meta_rev_id":"36def02702e02fcc411f6a0314efa4cc0f204d61","_number":877369,"virtual_id_number":877369,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-15 14:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-03-15 10:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-03-15 10:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-14 12:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-14 13:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 10:15:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-03-15 10:38:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7197590c0c1859d1c527f48b250d9d9ca80b1bc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-14 12:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16faaff39a3d85bab7af3bcb06396cd737b36664","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 12:41:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad264e0dee3e4fd394528b65da2ae982\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7f41781abb30402fa7cf0570ab26d770 : FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4cb0a4808b29518951a08c23ec0e5d9aae9ca1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 13:58:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/071fb8ce06584762a301206062e7f49c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4736d415364d421d8ae847d775d89ab1 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2a03eabbcdc4a51ade1ee4c7ac3da66 : FAILURE in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea4cbff571b64548b03a47ce2ba66254 : FAILURE in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4c44b95ce484d2c907616c9239ccc3b : FAILURE in 3m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53479cbd14684cf6b5aea6bf1030d910 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c49a50682934a7cbdbfe3fe478f03df : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/849468d3f5384ff9bb5adf24818ce2d2 : FAILURE in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/93c4143487534abba26abf47439a8986 : FAILURE in 12m 39s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d8364350a0db4d128c018773873f6cb5 : SUCCESS in 11m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/93dcdc3cb79e441f9ae0ac7c663a28bb : SUCCESS in 4m 27s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a7b43099a3b04cf79d1ffac13f6a2044 : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/62cb3d3d67d448268849b07e3f344f71 : SUCCESS in 9m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/229a0ce613cf4ade9a0a1e2b851eb9fd : SUCCESS in 4m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/188df28ec4fb4eb7bc975fb4c17b5224 : SUCCESS in 7m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8705db6966094087aebcd9bf3f0154b2 : SUCCESS in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/83a21980ed7f4c47aa76f107ae27aecd : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbd5b2977bdb40aeac9ea1caedddd49b : SUCCESS in 8m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e0ac284eae84433bd7efc6d89882e9d : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f8ba08e3706b433991cdb4174173218e : SUCCESS in 10m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f29126e60e754fb294e4ec2f69c4a1c9 : SUCCESS in 3m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/99e085b5cb5c47d0ba8b75abf2537767 : SUCCESS in 8m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee1c198b1aa449369b3207d3f27bd8be : SUCCESS in 4m 29s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/cdcf7012eaea4aa395aeb2257ae4b208 : SUCCESS in 8m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/db181f92b1ff4609a4b9ae31234b0e95 : SUCCESS in 25m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b23a5e580ee041dd8f76c5c90ba52c53 : SUCCESS in 12m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/91358c63f8fe49f3bca0669d4b4e779c : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1283f53b655a461f9f3a3218921f0bfe : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23ce29e586744f39921bb695691a66f8 : SUCCESS in 22m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a313702a29742cea113d1ae6189d4bb : SUCCESS in 5m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f4371fff78cf4d2084de8a9127988fba : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1266b56e1b24877bbe41b6e114c7071 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/79aaf4619cb94ec4a8463fa18b32eb4a : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bc95222003b453283bd04016d795bb2 : SUCCESS in 44m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71e1cb423f23435a8a2725b59d1ccb7e : FAILURE in 1h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"70dc9d41227b6a9b0b5bbc10767bc7aef76bf55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-14 14:55:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"44f0bf468626b15aa7489d095feead614076c582","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 15:07:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85bcdab7395a45cf809875c925acf714\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0d5ccbdb7bb4925be60a9ed7cb7ec8e : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4acaa71d38863f7bbd8195240eee68e750bcd59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 16:23:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/277502d03d3c4db0b2107fda476d5b08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c084bddf12eb425c9b55cb319442ab3e : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c6684d4ca9416c80aec7cbdc7c4aa3 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940d713034d040b89e0c6d011708bff5 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fa3320703484f3baea8321c3a8f6d70 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d6682ab08264fa2aa7f1cd86ef120c7 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef98430bfcb14b20bd306c17f4235970 : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d05890b284140d099ece104fc6022c3 : FAILURE in 9m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e7dff233f004fb1ae4614855b189a42 : FAILURE in 13m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8602b9982caa4f7f837c81ed078e3da1 : SUCCESS in 6m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3c1f0707cfbc46aea857947dca8a9257 : SUCCESS in 4m 54s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/56050d50bcf24e36a37be41764b4d5bd : SUCCESS in 5m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8f9d7bca7eee4660b4e46ba96635e4fd : SUCCESS in 10m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9ec3b25f1dec4a5898aaadb6274cdbbc : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/afe6608e9b7441a19e83e67f00667828 : SUCCESS in 4m 49s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/392ddf6f96864cc6868b8f9b3726cf5b : SUCCESS in 8m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d264eae75f5d4e5db07bffa81d59a624 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c04d8f0487f4aa6889847fe2ba6b899 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3368226adef4bc0ab8033bc7a5dc0b9 : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d4fad58f19be41ffa963fdbbdf3c2364 : SUCCESS in 11m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ca6d4b07b80840c4bc8e9e44e74f2c97 : SUCCESS in 3m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6f6446e3205e43bbb72cf8d50053b66c : SUCCESS in 8m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f4c304653874653b3abff16d8535b3f : SUCCESS in 4m 17s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/c5f67cd5e82a49ca84eaf064c12fa26e : SUCCESS in 10m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f2af6423c652424d989a8cd03c08010c : SUCCESS in 22m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cecf5e5afc97425db00593cb5d27e86a : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b76557143db74536b6f9b43c555be5fc : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6bbcf68434084b3597529d10690f9a2a : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13edc791a55745b9bed7a7067fa86be6 : SUCCESS in 28m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a9a230bcf85f48a88388b1cdf0eddc75 : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c7be7023dc2d4c668da87d383f1df293 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28dc65ddc0a44d37bdbfc3492402b609 : SUCCESS in 6m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8d9810be948d4e39b56bbb64f7001a85 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9317293b2a134040956eafa18ae52709 : SUCCESS in 45m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9235fa0d52b44a5cbbcf6c5ba30ca88e : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f2bb4313d03874ef5cab8e9be0d7d320b515823b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 10:15:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4791e824128fdcef4cf864f517c64ccffa33bcff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 10:16:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a6368581988a7f318b949d7d7457b8489ca1e421","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 10:32:13.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e200ac083c6e42f4b5c221e0ba1ad352\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f44e903c61234ef4b95e116dddedaf26 : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4b8080bb7857aec9e922395df20a59f4ee65ee2d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-15 10:38:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dde2149b34eb252d0f9b5c098866cbe597e0d5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 11:59:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b38c3fa79f449ebaf82732cf45b227\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c931907107e4418d9e1de91f921b43d6 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/263f802140eb4d9192d0aadf15b334e8 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219d0bdbf71c4fc0910d77fe0540b164 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6344b9fd46464061af1aeddff525040c : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e64b183931b432a921092856a65ccf4 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/693fa436ff244739aff1fde1a3a1afd8 : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3af46dd8a72147328e125beb1cb7e9a2 : SUCCESS in 7m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/449182b93ba74996bfe00a98291b0f4f : SUCCESS in 12m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9b4d9b97773247f0b9b2abed0a7bea20 : SUCCESS in 7m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d1a093c0b49f4616a2e5950bf7bd7b37 : SUCCESS in 5m 13s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2105dac375ea4366a24cd73eb4f8f5b5 : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e146094e1a91424a9148b91681bf6de8 : SUCCESS in 10m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/96d13add9bc34ad091709ed7cb5cadef : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/43943b00c7fb4e8d9bf41435b97f2432 : SUCCESS in 4m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3f1d2ee58e8343f8a26ac0a262cd96a1 : SUCCESS in 8m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1631080267e345f1a21c3acdeba71ceb : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d80b32fa4f6847f0a6037b5545b3c71b : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/57fd722211d64d55b758dc269b6153bd : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4c37e744177c47ac8a4ee9e304522311 : SUCCESS in 10m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/53cf1de02f4f4c3ea9976ae66cf49946 : SUCCESS in 3m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fa69937cd33b4a04a816bf30c18453b9 : SUCCESS in 8m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d87bbbc27da048ddac4c7ba03edbdb11 : SUCCESS in 4m 38s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/7737551be8e54e609fd1c2ba1590fd9c : SUCCESS in 9m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e6e53dffeb6e4b0bb9f2a543be28e688 : SUCCESS in 28m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fbd06ab9e0dc49f1912b634438aea002 : SUCCESS in 11m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b43250bb16d24ee5b39d58e8260b5b10 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da2d8d9b911549f4b5571727a61c2bdf : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64890d86e85c42a080c7948bbed85353 : SUCCESS in 21m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2fabcaf90cdb4291a57b8cd2fcb60c9f : SUCCESS in 5m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5751afe00734449e930f71d908343cd1 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62dcde54854845938fa5a3a949067c0b : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/afaa3217f2944b52b67709ddc31f1048 : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6688a0922d214bfe9726b4efb0b00107 : SUCCESS in 48m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d35053aa450643f9ad766ed363b80d0c : SUCCESS in 1h 24m 23s","accounts_in_message":[],"_revision_number":3},{"id":"eb623f8815dc418004960b169d3cd200137fc669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 12:00:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6221e659a937c229876417cfacd408d302671940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:31:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dea275474544d769a8c7fd7acbe3d47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb1fa00d6e841bc96bea7c4a4aaa9bf : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a0152eacb743b49f71882590b164ba : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ad9a9c794c945098c7ecffb6b557fcf : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d84b01ba2584de0bcc5a91b8dca6172 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5743c663f82349869c1cf6f83a8ad478 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/962628c5b66f4fcc9af3a30387c03106 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e22ab62853f64c009639da61858cb977 : SUCCESS in 13m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/99469434233a4b09acbb46c0e055e433 : SUCCESS in 9m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e4597112f9434f4a98375e2d29b822c5 : SUCCESS in 4m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/532c12eb8cdf4df183034f14a602af5a : SUCCESS in 5m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f09f58eae8ae40de8d0c39f4d98a3010 : SUCCESS in 12m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b4888f6476c24f0691c55933c520329d : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af65052309e9469d9aa9da202a781a03 : SUCCESS in 10m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0b0dabe3a584b9d88fa93334dcb88f9 : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/016f7929abf64b9a97cee56276220ddd : SUCCESS in 13m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3f4829b3bfb84a84a52d8fe1115c7990 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aac43a71e3834b3d854d57ab4c3e6491 : SUCCESS in 10m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/76abb32ddaaa45aeb915d03963194940 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/47cf2bb643d845a19d08f4c293e2aa18 : SUCCESS in 25m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0d712d8160a34f6eaf0c2def79414061 : SUCCESS in 12m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2560365044ec4bbe9ccfe9de32917e6f : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edab64e14f1f4fb8b15a4e46e4847652 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f8d24df0e644c448e297afb14ec9bc8 : SUCCESS in 22m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b4f5b040e7614c9186c9c36014d1d91e : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/81cb9240472f4adea3741d49142d43c8 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/256ee9090d50400ba421a03de0428cb4 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6d776ed158794d94aed7b228b06404e5 : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/053fc771e862463a8b04420a922f6db4 : SUCCESS in 51m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0971877c18f146cbba03fe197f9cb122 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"36def02702e02fcc411f6a0314efa4cc0f204d61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:31:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"65935c6c8bd5cd916f95641291b3ed2efbd9b941","revisions":{"d2c0993372e8d034b3051c2dfb2cb5a904cb03be":{"kind":"REWORK","_number":1,"created":"2023-03-14 12:28:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/877369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/877369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/877369/1"}}},"commit":{"parents":[{"commit":"fc7e2105e81c352602085bd2928a706d0ab8a80d","subject":"Use correct nodeset for requirement-checks job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc7e2105e81c352602085bd2928a706d0ab8a80d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 12:28:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 12:28:49.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.1.1","message":"update constraint for ovsdbapp to new release 2.1.1\n\nmeta: version: 2.1.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I6f4228ba5a3e1200533ec7e7cf8ba46a6c2e120c\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie762d145d5ad6949119fc03a87e528e0d8928719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c0993372e8d034b3051c2dfb2cb5a904cb03be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c0993372e8d034b3051c2dfb2cb5a904cb03be"}]},"branch":"refs/heads/stable/zed"},"a8723f1b29cd2df2525734e2937183ad8906fde4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-14 14:55:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/877369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/877369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/877369/2"}}},"commit":{"parents":[{"commit":"4ebc05e62d7af57481bab40c83c5320acee35e59","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ebc05e62d7af57481bab40c83c5320acee35e59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 12:28:49.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-14 14:55:05.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.1.1","message":"update constraint for ovsdbapp to new release 2.1.1\n\nmeta: version: 2.1.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I6f4228ba5a3e1200533ec7e7cf8ba46a6c2e120c\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie762d145d5ad6949119fc03a87e528e0d8928719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8723f1b29cd2df2525734e2937183ad8906fde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8723f1b29cd2df2525734e2937183ad8906fde4"}]},"branch":"refs/heads/stable/zed","description":"Rebase"},"65935c6c8bd5cd916f95641291b3ed2efbd9b941":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-15 10:16:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/877369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/877369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/877369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/877369/3"}}},"commit":{"parents":[{"commit":"37daeb6295a34c994ffb4b10d1cfc635e30f98c9","subject":"Cap virtualenv (\u003c20.18.0) to unblock gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37daeb6295a34c994ffb4b10d1cfc635e30f98c9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-14 12:28:49.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-15 10:16:25.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.1.1","message":"update constraint for ovsdbapp to new release 2.1.1\n\nmeta: version: 2.1.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I6f4228ba5a3e1200533ec7e7cf8ba46a6c2e120c\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie762d145d5ad6949119fc03a87e528e0d8928719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65935c6c8bd5cd916f95641291b3ed2efbd9b941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65935c6c8bd5cd916f95641291b3ed2efbd9b941"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
