)]}'
{"id":"openstack%2Frequirements~877474","triplet_id":"openstack%2Frequirements~master~Idcff5d79b4884dd1cc7ed1ab42c9e4ce89d4a6d2","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-03-22 14:24:10.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-03-24 20:02:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcff5d79b4884dd1cc7ed1ab42c9e4ce89d4a6d2","subject":"add opentelemetry to global requirements","status":"MERGED","created":"2023-03-15 12:47:11.000000000","updated":"2023-03-24 20:02:54.000000000","submitted":"2023-03-24 20:02:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877474","meta_rev_id":"952f3412f5e465e5e24fb650702a4ae9a755666e","_number":877474,"virtual_id_number":877474,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-24 20:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-03-21 16:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-03-22 14:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-03-24 18:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-03-24 18:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 12:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 14:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 16:55:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-03-21 16:46:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-03-22 14:24:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-03-24 18:21:32.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"d49222f5b076717ac2711a3949dca4780f8d9040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-15 12:47:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d34f76e4eb492bc1235793945cb0637a90ecb80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 12:51:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16461394c654b0c9f00bfee38aa2dbc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4cd9d16b711b44b0937f19d2f45c22d9 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8ce33625c18ddf7364032cdcd8de863551d78680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:12:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfaa12d36fdd49a094c86f0cd0950fe4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78086034fd8547d7b2fd436461b188a3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627f2117de2d4b289126fce60e78a3c2 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b369f9f824d842d89a5fee4b8bb3a446 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e74af50538084b3ba3e0d97beed94047 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f4c35c02ce64fa1a694c0d585e0c91a : FAILURE in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0edac60d13974b1cb8310e0c08581237 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/afa10fc2da73447a966755614f14917e : FAILURE in 4m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1231c8c7f8be4751a8303cccfa357fdd : FAILURE in 5m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9e06bd1e019a473ebf71f0b6f18d9ef4 : SUCCESS in 6m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dbd1509cf2914e8d90b46a2f409bdd8c : SUCCESS in 4m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9f0a9ac6aac840a5adf8ef8694fd34cc : SUCCESS in 5m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1dc9b9e71b5648a1af5dc67d3a6b4286 : SUCCESS in 10m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a8923e6b5a5e44e2809bf3683a919a3d : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ec51d40f6c56471eab7cd180d3606d81 : SUCCESS in 5m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9c6da478e4be41088ef1f893948fcebd : SUCCESS in 8m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1ac46f029e674ef6949ec39bfc52f610 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52df9043c07648a99c271a8e56d08cfa : SUCCESS in 9m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b96eb2788e714b4b878f901d95f101e5 : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/755cf15b3e6a47b5a3749d04c5c929a5 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f11beaa4b5954b02bb3a673c5141e123 : SUCCESS in 4m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/56a2e56a7efe4336bc5eabe4eadf7d7b : SUCCESS in 11m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b0b0e3502a714a9d8fbb4e1e74255ec9 : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a3bb6a62b75a46f9a928ab854d79b166 : SUCCESS in 9m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/35f6338883884820900d94ee97c2bdf1 : SUCCESS in 24m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/92c7dcf7f4c642ca800a32ce4865aa30 : SUCCESS in 14m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dd84cd388791423eab0d80108653c32f : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6474e99d278e4abeb84fa68a15dbc984 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b79b32d5a9cd47968e54d3526ab2fed7 : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e4f8f1e0b69f4da7a659712296abe1aa : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1dbafa7f018a4477bb127376d46c94e1 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2338dfb24004c92a97807649f4151ad : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/26d204b884e6456a93d991b48f2940c6 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a180f69a2f4441281c1a5d956fed7b2 : SUCCESS in 51m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/602e05bf4b344909b12af308347d8d6a : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"84ebc4b4a2d96cc1f3eff8a2e15e485592b708a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-15 14:23:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40180fb6e71025152f6de6d884805dffc95502c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:29:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8956c0d12f4901bba2ad6733b9a924\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ea41d17f6dc46bb8b3c168bf0b86829 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d6879d71f1f1efc24cf1eefec0c8930e649fe5d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 15:43:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9719b26840c64dd9b102c74520d892a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4255fc0edfde49cd9abd41f8e1e5af36 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018fc65bb69e46cf81139efcb3997724 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2015127736ce4334be416294c93cb3f7 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48249541dc084e3fa24c8a761db14eda : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31e5c08dfa7b4bbc9df5b59839406e2d : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0acf418b4b3d4b08ad72f2aca91558e9 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b72fae3985fb4a5ebb8838ecbcdfe421 : FAILURE in 5m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d39c50d0c99d4f47a8a61149c6b327bb : FAILURE in 6m 08s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a1a71a0cc7b47839fac6dc678d07db3 : SUCCESS in 6m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4c6e9f4e4624494189baa4d999675525 : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b752b609cea24447afde9e8b6452f5e7 : SUCCESS in 6m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/21f523a7fa844ff5852faa67773b7f25 : SUCCESS in 10m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7a1815570a8546df9801fcc1496a263a : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/98af5d72564c4677b8d52c94a74de3f5 : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/edd6ba1b47754c19b3807b604b45c98b : SUCCESS in 8m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/600336c9347f48598f17113650b3166d : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5bfb9093e054c78afcf668419f6d526 : SUCCESS in 8m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2444392c45a74f4bbf8ed61b97dcffc4 : SUCCESS in 9m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0106963444494efa80f16db228e37c32 : SUCCESS in 11m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/679d7fdecdd34f63958ab36d1d6eee70 : SUCCESS in 4m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a366362aa6da4c7c9e58a205b3d04935 : SUCCESS in 8m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2c171dcfa972421885a42e93eff92601 : SUCCESS in 4m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f87c13dbae9d43ab954be78d9382a08c : SUCCESS in 9m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e59f942e36784582a4fadbd16b1a437a : SUCCESS in 26m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/55a2e9aeb8f24df0b9ebeffe6268835a : SUCCESS in 12m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0648333b13b74bc2a99a8285dce3ddea : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c10c3c8c82449d5be2cf861dbc802f4 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc54a8c0292d423ab3210ebb8e76b3b4 : SUCCESS in 26m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7e173f8cc07d46b7a250afc34a8346c7 : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/065b41f7d7734b6399de8990c01901ed : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1d322d067d44c87ab91e2b5dd63e9da : SUCCESS in 5m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/987187d5a990484da56e776acb4afd3f : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc843879d08647ed84f7ad0b9aa3a6a0 : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e15996d52c5450084f4fd6fca65fda9 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a8991e424c382acd4536b75421f2d35c2c2cf9f7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-16 16:55:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f216b37d3abb859d861113b84c5a508478d5fa2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-17 07:24:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12a197ce0a4f42e22f1d501e26d4048580e0b910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-21 14:39:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1e024101ee3aaa318d1745779b93e01aded4ae2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 14:48:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa84831f497d413e9609c1e70dfaacdf\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e829773c176b4d838b06c4de83d42c3c : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"e79590d8aa949bdac3daa8498ec7ca1ac5377cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:54:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b902eff4add94c1ea3365ded5d069d2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c15d2567b1a4eb7922b431494ced033 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ba3278003947ff875f0572dea6a6ba : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9979a01e642d4448b0b4547fcd8443be : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cc04ead3a84422b85d0e9ac1fbf4787 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b5f73aed1f44632a30238700dd93756 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9223e2f5990d44609a64035f243e5069 : SUCCESS in 4m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e783948fa71442ea1b5eca72d176a75 : SUCCESS in 8m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2c6b62aaa024d9ea4cf5df42488f8d5 : SUCCESS in 12m 54s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/14dde3a975a24764873485f3a9118aac : SUCCESS in 7m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/92cd1573a707444a9cb5ebcb3faa02c4 : SUCCESS in 6m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/27ff3d7499564e6d98a169a44b2a8aa9 : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5085fbcad46746dc97c73137c3067076 : SUCCESS in 10m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff81274374ca4397836395818b1f88a3 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5641c5beaa7a4dba8e07ec59d250dc43 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9d99cbd92319433f81609572a2e19a00 : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6d7cd2c523b44ba7895ecf37fe0ae255 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c8f2407abc7461db46006e831ec2dc3 : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/43b7cd75b96c4c07896b5f3cf2819cad : SUCCESS in 8m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/19acf95b475344e9857ce3fe70c82c3c : SUCCESS in 9m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4a8a89aec8a6459ab04f2014e30d3a5b : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/713f25911e1342c88e6aa51f48172a49 : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ed2657afc3504ffea228465fddba6107 : SUCCESS in 4m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d2fa900ca4ad41659525f7e90ee97035 : SUCCESS in 10m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/55c88640ba4348ebac15850a4b686801 : SUCCESS in 24m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d0235adc38ff4e1ea3127846a7666cb3 : SUCCESS in 13m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/be4d58db32254da8b17f5640aec0d901 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e616fa5dcf8843809d850f813ac729b0 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6f3628f327f4373b14ccf3cfd4280d3 : SUCCESS in 23m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ed01fc57cdb64ebc85d292c059fc2d2e : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9032bbd0b6a544d8b1c7e21f2804d9be : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8f17e5ac53e4e788b392f57eec3deaa : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/441e8c40efd94a308ad0b4582ae015a5 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5457b78a936b4326b42783211371b395 : SUCCESS in 48m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed4badc539ce41198990992c4810d2e2 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":3},{"id":"86bc2036552aa0345ab53a2ecd34c7352a0bff66","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-21 16:46:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"de093162828a80c82b1648fff04f6632eeb99619","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-22 14:24:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3b4561efa14a4cc845e608390ad83b5246e23f69","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-24 18:21:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89387da69e907b34fb437536b2097d23108030bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 18:22:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fed181d72109bc633b480027f13facad9edab54e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 20:02:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4dfdc555704a72b166dfdd6cfde494\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41479609694b44849fcfff6591aaebb0 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ad4a2f46e54a1cae5766c5f488c221 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ee0663cc3044fa81a29efe0ff84b7c : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba1fd9bf870e41b7bf28d4d21390d675 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5e91bf89bbd43c7b1ec1da2ced76962 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59d37f1b609d42829753fdee1c8e235e : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8490409956114eda83ecd009ee2b9edd : SUCCESS in 11m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/de3a55275183468b8ee3cfd0d4755834 : SUCCESS in 13m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/af7b00db54d348cb96c1f72f795574e7 : SUCCESS in 11m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/880b65cb2f6b4b5684e27281c51148c0 : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bcbe0eae38bc495ebb977b0a77cdc5f0 : SUCCESS in 5m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9f9ccc2982884bc5b8eb314c7a9db265 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6c747f8131da412c97de404f213d0d0b : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b20746f542964efba03adeec8d16bb4e : SUCCESS in 9m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/052bb199fb1c45ab8fc3e2be7f1b222d : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/32513b33fe38421f8f22c7567ee4ef82 : SUCCESS in 13m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2dc823b4db6049708461a324a1fb718b : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b0416e90b5a44f99a8149ac147e67ee1 : SUCCESS in 12m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b18a571a70be42fcaa966553ff34ccb7 : SUCCESS in 4m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ba42577c6a394922b6d6b9251c7c19dd : SUCCESS in 22m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8478f7494d4c42cbbd1fe68003a59cc5 : SUCCESS in 15m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3282d550cc9d4e49bf5e4b3325609d00 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ff7c06d2bcd4787b903993172a05239 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65caee6ef4f546858c0ea78d89d2b7f3 : SUCCESS in 24m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f564dfb92fbb435c8fcaf792416a7447 : SUCCESS in 5m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/493bb98bf9814c4bb541d5c1ef7733dc : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f954387b4544699a9f206faed90a5e1 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/26958efe9d6a4e019296387cedb1642a : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4075f21620140adb3577ebef674f821 : SUCCESS in 45m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea1e7908b1c5463198ac6d981444a0e7 : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":3},{"id":"6669a10a8530f8c18e19ae155754a9ad0a921cd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 20:02:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"952f3412f5e465e5e24fb650702a4ae9a755666e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 20:02:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3236e49f0f47389dfab32aae681709\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf6268963b747d992d3616f4db523a1 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64f337e1c510afabe2e29e7d393e1aa9401100c6","revisions":{"eb42dd1ff99527d76c5feee77b42ab1d441597ff":{"kind":"REWORK","_number":1,"created":"2023-03-15 12:47:11.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/74/877474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/877474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/877474/1"}}},"commit":{"parents":[{"commit":"e7d3312a31340af93dc2f35d03f622272007f69d","subject":"Merge \"update constraint for python-novaclient to new release 18.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7d3312a31340af93dc2f35d03f622272007f69d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-15 12:44:44.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-15 12:47:09.000000000","tz":60},"subject":"add opentelemetry to global requirements","message":"add opentelemetry to global requirements\n\nUsed for osprofiler, see work in progress:\n  https://review.opendev.org/c/openstack/osprofiler/+/877358\n\nSome questions:\n\u003e\u003e\u003e Is the library actively maintained?\nYes, a link for it: https://github.com/open-telemetry/opentelemetry-python\n\n\u003e\u003e\u003e Is the library good code?\nYes\n\n\u003e\u003e\u003e Is the library python 3 compatible?\nYes\n\n\u003e\u003e\u003e Is the library license compatible?\nYes, it is under Apache 2.0 license\n\n\u003e\u003e\u003e Is the library required for OpenStack project or related dev\n\u003e\u003e\u003e or infrastructure setup? Which?\nYes, this change: I74cdcb2aa99b0162ba1c14059111f09d0bb534e3\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Idcff5d79b4884dd1cc7ed1ab42c9e4ce89d4a6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb42dd1ff99527d76c5feee77b42ab1d441597ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb42dd1ff99527d76c5feee77b42ab1d441597ff"}]},"branch":"refs/heads/master"},"b7dea1827a9d5ae32d83e6cc8d9aad9eaa0c93cb":{"kind":"REWORK","_number":2,"created":"2023-03-15 14:23:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/74/877474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/877474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/877474/2"}}},"commit":{"parents":[{"commit":"e7d3312a31340af93dc2f35d03f622272007f69d","subject":"Merge \"update constraint for python-novaclient to new release 18.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7d3312a31340af93dc2f35d03f622272007f69d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-15 12:44:44.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-15 14:23:34.000000000","tz":60},"subject":"add opentelemetry to global requirements","message":"add opentelemetry to global requirements\n\nUsed for osprofiler, see work in progress:\n  https://review.opendev.org/c/openstack/osprofiler/+/877358\n\nSome questions:\n\u003e\u003e\u003e Is the library actively maintained?\nYes, a link for it: https://github.com/open-telemetry/opentelemetry-python\n\n\u003e\u003e\u003e Is the library good code?\nYes\n\n\u003e\u003e\u003e Is the library python 3 compatible?\nYes\n\n\u003e\u003e\u003e Is the library license compatible?\nYes, it is under Apache 2.0 license\n\n\u003e\u003e\u003e Is the library required for OpenStack project or related dev\n\u003e\u003e\u003e or infrastructure setup? Which?\nYes, this change: I74cdcb2aa99b0162ba1c14059111f09d0bb534e3\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Idcff5d79b4884dd1cc7ed1ab42c9e4ce89d4a6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7dea1827a9d5ae32d83e6cc8d9aad9eaa0c93cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7dea1827a9d5ae32d83e6cc8d9aad9eaa0c93cb"}]},"branch":"refs/heads/master"},"64f337e1c510afabe2e29e7d393e1aa9401100c6":{"kind":"REWORK","_number":3,"created":"2023-03-21 14:39:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/74/877474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/877474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/877474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/877474/3"}}},"commit":{"parents":[{"commit":"e7d3312a31340af93dc2f35d03f622272007f69d","subject":"Merge \"update constraint for python-novaclient to new release 18.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7d3312a31340af93dc2f35d03f622272007f69d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-15 12:44:44.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-21 14:39:19.000000000","tz":60},"subject":"add opentelemetry to global requirements","message":"add opentelemetry to global requirements\n\nUsed for osprofiler, see work in progress:\n  https://review.opendev.org/c/openstack/osprofiler/+/877358\n\nSome questions:\n\u003e\u003e\u003e Is the library actively maintained?\nYes, a link for it: https://github.com/open-telemetry/opentelemetry-python\n\n\u003e\u003e\u003e Is the library good code?\nYes\n\n\u003e\u003e\u003e Is the library python 3 compatible?\nYes\n\n\u003e\u003e\u003e Is the library license compatible?\nYes, it is under Apache 2.0 license\n\n\u003e\u003e\u003e Is the library required for OpenStack project or related dev\n\u003e\u003e\u003e or infrastructure setup? Which?\nYes, this change: I74cdcb2aa99b0162ba1c14059111f09d0bb534e3\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Idcff5d79b4884dd1cc7ed1ab42c9e4ce89d4a6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64f337e1c510afabe2e29e7d393e1aa9401100c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64f337e1c510afabe2e29e7d393e1aa9401100c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
