)]}'
{"id":"openstack%2Frequirements~878422","triplet_id":"openstack%2Frequirements~master~Ic3fbd599309b85b5d15ce5d48eb523250fc2761c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-23 17:44:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ic3fbd599309b85b5d15ce5d48eb523250fc2761c","subject":"Bump diskimage-builder to 3.28.0 in UC","status":"MERGED","created":"2023-03-23 15:19:00.000000000","updated":"2023-03-23 20:18:55.000000000","submitted":"2023-03-23 20:17:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878422","meta_rev_id":"8d8dd530f6bc1d921f32f498f465eaa5b431dd59","_number":878422,"virtual_id_number":878422,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 20:17:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-03-23 16:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-03-23 16:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 15:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-23 16:57:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-03-23 17:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1275c5bddd05c3ae5672aa145018a93ccd002af8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 15:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4fb408bfba67518ff278394fc250b5e29e68f96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:24:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec57bf545061442280ca4fcd69247168\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aeae07500f5a402a99e773ca7fc4fb1f : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f1a99f5311d5085dc11e55ba69d568e1d9440520","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-23 16:57:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da2d00903da257cb32a561c43d70090446371e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 17:01:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/add60580da5747e797e045d232aa8b48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e213b870e6cd4b4d8345d0d9142a0501 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e5f932347d74fe2902d89d532d74644 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d5e7c5dfab546c787a65aef2490a693 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dceac06c89d74acabc46cbd672c69ae1 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b8ef33aac4d445494115fd95ecdc534 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f6ac0bfafc0541258b30ad782420ffe2 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c67dd1c984cc4f48bf9e5a4747eba1f1 : SUCCESS in 8m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b49e4a97227e4d50817c32f9115ada03 : SUCCESS in 13m 54s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/712c521e171e4158825653e98b772e73 : SUCCESS in 7m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/70949c314b5141ceaafedcb033656ca9 : SUCCESS in 4m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bb6c2fb22f54498fb210125fda9782ae : SUCCESS in 6m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/67324d606c7d46e7a913d065e010932e : SUCCESS in 11m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/51011b7acf124d8ea74d059b6b05c0e2 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f328df7186424bfdbc2ffac629411649 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bce93386dd544267a90ee6e8eddf92e5 : SUCCESS in 7m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b8a152a428c449d5867fefa94fb9a86a : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ef2ac9490a74002a659415e1f568c54 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/961a18417a744b7381232317b8a9251b : FAILURE in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b4a05f211ea54ace8a8ebd4aa60e55a2 : SUCCESS in 10m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e34259197e484b06b6cfc3bdb5fe6d31 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/051986b51bd54499af7110293282be68 : SUCCESS in 9m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0cb2f4e20666430180508d6860396219 : SUCCESS in 5m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e548883c69284746a0105508ea4ea756 : SUCCESS in 11m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/14f45c6789934cf3a363d4ee195bd7ed : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/61f13e2a9bc5458987572732d82957c6 : SUCCESS in 17m 49s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/22632d1c587d4b00bad8e79c706e313e : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/251dda80df314a4c9dec9a813a84e70d : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c28db18fc6d4963b0a990dadcff8417 : SUCCESS in 25m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3b7c92265eba4ba998819dbdc168723a : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3ce938d2389348f9bf9d77a5eaef1ff5 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea97eaaa717942779bbaa37aa1b8857b : SUCCESS in 6m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4e847f5b271f4fd1b135ecd052442ae8 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbd16ae0cb184b7bb9d2ebf4fedd907f : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bbe256bc3bc41fba2c91b19c693ada8 : SUCCESS in 1h 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d08cb4c5ab44f15e221a9767f5fa1d92ed603012","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 17:44:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac36467d19008b2dfb23bedbeea7032911aacb3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 17:50:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bfd0585ba7f4c56869fa8226b406627\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e2cbb1cd171e4a569fa4618bcd2ccfb1 : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"60b9b070f05492273beb1e0c2aa54c62ae8d9e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 19:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/495c1db50a0241e9a8ea02f72737c57f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bd9b7c902f45e589703e6b85ae1d46 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579d2eeea0d442d6a0f1db893ec99f87 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfb91fdc44e048e2a4c0c95d3bf13b33 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/333a904ae09041b2b0b2c59d05129610 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89c12513c4224691bb43ae027aae70a2 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de79845a402c4bc7b7cc96122c6060da : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d8b4da9d57b34144b4dcd9e537f2c87c : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75d71244c4e74c3e8418415281e41dca : SUCCESS in 11m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c7737441a27f4617a0967641a4091c2b : SUCCESS in 7m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bd7cc06dd46b4ed595e60a4f44694858 : SUCCESS in 5m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/55263f1b40334059aec4c322f2862d74 : SUCCESS in 6m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2a94f1d782cb484eaf0dd45b07e09683 : SUCCESS in 12m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f82fe9477b924156a64636ae24c08f42 : SUCCESS in 5m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0a59aa87e146468b8799f28be4bcf0ee : SUCCESS in 6m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3c780fd0b7614068b558f35d180f804c : SUCCESS in 8m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/549f9d8a6e424a52b7f1d7312bebdccb : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ed464ce41d64f1da0502e4c8f3111a1 : SUCCESS in 8m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a92a9d2b442d4487a16bf84fe461d6a6 : SUCCESS in 8m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c6d039573d7e4883919aebdd16567fac : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a383c2e3b0cf460397a400dafa152567 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a33eaf023f0041f8b8419f6a5170b967 : SUCCESS in 17m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8c8d1406cbe84bc1b3849707259c0368 : SUCCESS in 4m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/100c7738e6994619913d172018612592 : SUCCESS in 11m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4f0f62493c4b4591bd5536e9570e1b3c : SUCCESS in 24m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8169ee2af4d5487db9975243899bf4b2 : SUCCESS in 13m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6100348fc7a44c258e72d8cf9cfcceda : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad86bbac5c22475f97de103761f1c0b4 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73f8c6b8058e4366b761a836a60900c7 : SUCCESS in 20m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c188855ce8fc47e8a98fd8ea5baf69f5 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/644dab06fc5e4cb99c39477e380e28c0 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b9a81b6c15240119bf4f500c0152762 : SUCCESS in 6m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6459d7cd6e1b4a65a485d96b3b09eda8 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fef4cbfb87ea4f38a9ce19d0e4d6f72d : SUCCESS in 1h 06m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8bd7abecea4478583d6c383812af00e : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"df585db4d1acbfcc70c659607be15def52ad8261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 19:06:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd40cb3719891514a945be9260684d364963a4a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:17:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5caa5742a4284edaa64a5c0de24ed880\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fd6e038f1346ceab62b381fd47e229 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282f2f29d37c4b9cb7bf34fcfb351e80 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91ac446b23764b42b8b6b5d9bb554d63 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff5bc8de6c744a808f19fcace73a58f4 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbd8f00bd66c415fb7005727769c8eff : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acab479163204f83a7ae9d706e8cb524 : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/239a9576cd5d4362a9043409e03530e5 : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/79553521ff5a425991968d90c553ea1a : SUCCESS in 11m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a91be219ac014c3bbac9b1ac628a365e : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dde9c9fe7c82407589c16c3a634fc902 : SUCCESS in 5m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/44494763e2394c88b8d55b679c7b0d24 : SUCCESS in 6m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45367e7cc12f4067b1a001fc9e86cf09 : SUCCESS in 10m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/728452e4f74049de8fc8d64155c2d6bd : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/069c346f267342dcb38f40a2e966c3fc : SUCCESS in 8m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4606ef46ced14954b40b260d8b856c2d : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/248b79177ff44624847b073d2af989b3 : SUCCESS in 13m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b8a94a728a5c44378f7daa5877f0cee3 : SUCCESS in 5m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b066e8b8db5840ab9537e57ea5850b39 : SUCCESS in 13m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e95b8b1c446548708ec3938ab684e403 : SUCCESS in 5m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/037af30629dd478e890ff7d09c834e6f : SUCCESS in 19m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e1b6514f8a9e4d26841e7619a2c6a7ef : SUCCESS in 14m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8f61aa245c20460a9bbda9093665532b : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4b379a58d8749c0b025bd053810739a : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73902b964c56422a8e82bc40e4ec4103 : SUCCESS in 25m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2699aba04b1642a293789075e224993f : SUCCESS in 7m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/93029ab241704745b3cc3a3a1b53a065 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08172bb5c78641449d5a5025a39c105b : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41a22cd8c57f4c10876885f7a644a50c : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c4270672ed4eb79e5baefbb1a7419f : SUCCESS in 1h 06m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45216d164034430bab01f962e1519a8f : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":1},{"id":"09e5b36d2e14bc818bac6e69f44890f61ec5bc25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:17:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8d8dd530f6bc1d921f32f498f465eaa5b431dd59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:18:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4df3033a8164215aa09ff991c62c3e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f9de4e202749d1addc42bf1dde3ef0 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"983bdf7c475ba047081d9d428026b5ac79c99698","revisions":{"983bdf7c475ba047081d9d428026b5ac79c99698":{"kind":"REWORK","_number":1,"created":"2023-03-23 15:19:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/22/878422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/878422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/878422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/878422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/878422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/878422/1"}}},"commit":{"parents":[{"commit":"2e55a9b88093c34d6f2fc1f1c1f0f838b4cb7b77","subject":"Merge \"update constraint for tap-as-a-service to new release 11.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e55a9b88093c34d6f2fc1f1c1f0f838b4cb7b77"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-23 15:15:22.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-23 15:18:58.000000000","tz":0},"subject":"Bump diskimage-builder to 3.28.0 in UC","message":"Bump diskimage-builder to 3.28.0 in UC\n\nThere is an important fix for ubuntu image builds included in 3.28.0[1] that\nsaves up to 1GB in the Octavia images by cleaning up build time dependencies.\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/878089\n\nChange-Id: Ic3fbd599309b85b5d15ce5d48eb523250fc2761c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/983bdf7c475ba047081d9d428026b5ac79c99698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/983bdf7c475ba047081d9d428026b5ac79c99698"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
