)]}'
{"id":"openstack%2Frequirements~878690","triplet_id":"openstack%2Frequirements~master~I4d68d8abb48b4179dfd5e838837b2406fd90913b","project":"openstack/requirements","branch":"master","topic":"add-mypy-job","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-11-17 09:03:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d68d8abb48b4179dfd5e838837b2406fd90913b","subject":"Add cross-project job for mypy","status":"MERGED","created":"2023-03-27 14:58:05.000000000","updated":"2023-11-17 09:07:45.000000000","submitted":"2023-11-17 09:03:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878690-add-mypy-job","meta_rev_id":"04d28cca79eea58a3a35ebf8f81b1c6be9c2f2fd","_number":878690,"virtual_id_number":878690,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 09:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-11-08 17:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-11-17 07:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2023-11-17 07:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 15:12:46.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-03-27 16:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 17:07:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-03-28 04:39:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-08 17:14:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2023-11-17 07:13:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"969c55c1c0c65fdc50bb2d04af6333acd10c2eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-27 14:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f560960508e310f66cde80d08f1eb2f9625a0f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 16:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a760551edd342a5ab3f8672c1f8dec9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f43f3f325c48b09b06c8277cdfcf45 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af49e84bc138459b9f8a3a77e717b0a0 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/890e69f0f1d74984b9ee441bffd30936 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78abafa879514436951d25e63665f8bc : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c4760e00584462c897f897892080ae0 : SUCCESS in 3m 04s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/21652d5f73d04beaa6b04513ee3e63c4 : SUCCESS in 8m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e0ec7a7fde9441a78e4e035424345060 : SUCCESS in 5m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2381ff8a76524cf0b3841d8022427589 : SUCCESS in 5m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0cd442bdcb8b42a199a8b8ca918a49e4 : SUCCESS in 10m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ae9aa2b0458041c0963e5778edf15783 : SUCCESS in 4m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8eaaa3896fd146b5ac11f46a492c536e : SUCCESS in 5m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/711530b1f1114768956c5e3a2dbf11ca : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/db3638c59f7a43e49bcd4c7b5af81c65 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ac466b4e91f4c8fadc45f0b89798bb5 : SUCCESS in 8m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d9f2fa41be34400eacf7fd3581f8f1da : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/09ff79c853e04ea7b43e47a28525c3cb : SUCCESS in 9m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9298fced29be42c092cb4ba7c48f1f64 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/28db42fd19ed493486c2329be55b7701 : SUCCESS in 11m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ac99b77bf76741eeb8950fd47a125cdd : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a0ee71b7746f47f694b42c81417ffc11 : SUCCESS in 8m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a935af56dccb494a97cf5f31ed5a5ad0 : SUCCESS in 23m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4619d523fe554ca58a22b77feba8f0b7 : SUCCESS in 15m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7a866b348c4c4c249c31abb06732f253 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a71e84db38ad4f429b7ccf77b1ede1bc : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b10bbbb7d6c4dd7831ed4da50228eab : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5a35d5507567465ba1546aed82e44295 : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e9c0340b42e84439aa2431845323bfe3 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6022cff025074994a36a671ca29f0250 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/980b70e69c9e458296f9379380b332fd : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5078dc891ecc47b89d23ebe321bd8ef9 : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d21d5962d4a8472cb6ef9d6673ce85bb : SUCCESS in 1h 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0b725568122aa369118b57f9b5718aa1d4eed980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-27 17:03:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d515d2a8ac6a3e965c15039b1d5b2b8d14d6c16e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-27 17:07:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3475f689dbeee7fd9158d6aa9b381a74f5a2a65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 18:38:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a163b40093134bc9b2597015bf259697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1469fea184045a780def5c8703fdac0 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c6fb558702e44c4873e93e4f3e31158 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/741ffd7f39cc4072b1d9bd366e8ea67c : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a601834dc1d547c08717e578bed4d552 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcae47c582724cb1a29f974edcebe000 : SUCCESS in 3m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/213077d5b55147989169990eda0324a4 : SUCCESS in 7m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/309f08002e9f464eb8ec78d41d4cbf9a : SUCCESS in 4m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fba5701ca8f24b338272e7d835d29d04 : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c67e265d86a843b692649f165b9ad9d9 : SUCCESS in 12m 13s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/942241a98b284a1bb716bed05f6e9bea : FAILURE in 5m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b2a4946208644eb49d8010b5af97a803 : SUCCESS in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/69cc545d929a47d4b9ee5f7f96dea9a0 : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1c4c6668309b4d8d829dbf0ffc07518f : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/26b3020988b841848837546e289cde30 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72aa6f5010a4476dae4c2ccb25fd61ab : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/50896df7ad8b4537b3da91b3d6d181c9 : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/923255658e58439ab29a3d76deab6211 : SUCCESS in 19m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/20c1da4e2b6944529922c720bf6f576b : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6dc3bec655a64e349dda29011b39f524 : SUCCESS in 9m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cd409876637942d9a7b7f98c92e7dc86 : SUCCESS in 5m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5cd55e423c5c41e692fb7c612deb7df0 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fc46ee16bef24c218f5387e3f21add19 : SUCCESS in 20m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/de415385d6f44eabadfdc66b24b79b4a : SUCCESS in 11m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3891f48ddb3c449cabca0f1186273c21 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbeea9cc9259407081ec140b25781931 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe50aea447004d03b0552fd50a0533b4 : SUCCESS in 26m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/78c8e501e3c94ccbb43d0b0631b4fc7d : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cff85802d5d14dde9fede7c78e5bb893 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36437c9cbf354228930a113f278ece9a : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ab9104c0574741878566712ddb83f143 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6751e1885f442b9ac65d69f0b3ecbef : SUCCESS in 54m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e66888864c014b6297adb4eb77465bcf : SUCCESS in 1h 29m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8cdd4c4acd56de599ffc0730e9a528b2ec751127","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-28 04:39:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21e7889bd465efdeb4b71cd7458538ec38e3da2b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-02 06:38:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d78f818ccee802802b06c5648f9ca4afb39e9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 08:20:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f42da8249e34c3a913909fdb2682b3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92da0fa1234c4294922079288bbab5ce : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a76c00b5ae49d483f485c23ff8b2b5 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31fce75049ca43a09e6fdd1bfab02c6f : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05d7b4c2604c40a48f7830a4d4482125 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66522c2db1c54912a50a7d97b4ffcc11 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d1248e721604e54847e1172b6a1bd26 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a37f4d21b2f04d84a4f4b387cc2b53c7 : SUCCESS in 3m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f43820fea63944579b22953f46d263ad : SUCCESS in 6m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9da208759cd6443989f3d4519d902712 : SUCCESS in 4m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3fd093c28b324ed6af0d9b48abe31f2e : SUCCESS in 5m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/569757de0629424a91e5155b0d0df4b1 : SUCCESS in 9m 37s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/d194aa2426c14c03b858020aaa14b64a : SUCCESS in 4m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/57cf8b5c4d5a4bc08612b003d033c376 : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/63441dd0f0d04144a50f93d5a7f0f378 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8decda3e68b943f498e3506826aecc67 : SUCCESS in 6m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8c1d8f48314f48df8c2f198eac01681b : SUCCESS in 7m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cbacf280f804b02b32a02691d0e09b3 : SUCCESS in 8m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c3186b2447c648ee9412196b3467875c : SUCCESS in 6m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ca3f3adee3674c069c420c26b464e149 : SUCCESS in 11m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/300e91db2594473e8bc0e280c0ea67f1 : SUCCESS in 3m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/acd73a0f305b4d3881a3dc838e18d7e1 : SUCCESS in 8m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/23f2971f3a0d43b38ff343d56214cf8d : FAILURE in 4m 02s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f8d23be2cc694892bc939ce09bf43870 : SUCCESS in 8m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fc57393adb5f4f93b80cc1c60864d6a8 : SUCCESS in 17m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f7b5370feb543a08534d663b0922f0c : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5400b1254d2843be8f5ef865aeb77f43 : SUCCESS in 12m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/70470e9066e64e74823d5496c6110e31 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afbb4826dc194ee7965d540eff28f23f : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/229f35be87cc4232a1ecf1370eb83f0e : SUCCESS in 23m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0463db7c51c84f7bbc452e4a8654fd01 : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5241c9e31fd74ab6a4ea67ffb928e929 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce40e6215bb747cd88942fce7ed2565b : SUCCESS in 6m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8d15ce51d09c44efaf6f2f980e96eb0d : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b29b92a2a4242dc89a43e44793d7bbb : SUCCESS in 1h 03m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cb869bb875a445c835c65ca8bfe83fe : SUCCESS in 1h 34m 51s","accounts_in_message":[],"_revision_number":2},{"id":"33169a4a0c4d510dde7c961f0cb2bb0bec65ff35","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:14:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"983df134e42ea5c18303d3aecd1c1f2dfb6c95db","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-13 09:09:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"984b7287f84481360f911b6f68b71a040053bae3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-11-14 15:14:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8551e309ea6850185b0700abcf5028440fed429","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-17 07:13:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a3b32d7718300169f4dbf0f3d36cd80c601415f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:14:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e64cefe274b1d410f98a0f5802b72f4fcb392cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:03:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07dabe5ff5bf4da8835d496d1c37c95b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa15601767d74b84a52a5e9c6b1813e0 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd68fffa51bc46fb8e49aba452aa61f0 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d75da48ed0ae4f9396943afece3f5444 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4664d250d45548389dc65bd5ca1a4b36 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6501738c332745268c2abe48f81bda58 : SUCCESS in 3m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e13e26d7d74d4539b115440bef414ed1 : SUCCESS in 11m 06s\n- cross-cinder-mypy https://zuul.opendev.org/t/openstack/build/896cb5de06704a8698fbf11db72df150 : SUCCESS in 5m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/14bf424eec6a45b89075d740584cf5c9 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/496c2d70f8f44ec4a636663370eea0a6 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4d51e92aa4524d70b6b028253299cffa : SUCCESS in 7m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/63c0032a9f964256ac7458e60ab17b99 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/677b8f02efc24ba79bda6773e8d28dc8 : SUCCESS in 8m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e2ebe09927bb484a84cf07e3f5eb5cac : SUCCESS in 6m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/95bc8126b5bb47049cca1b846847f96f : SUCCESS in 11m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0ff3d5a62b254b9582fb4e176b34f892 : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d810b2884df54df8b84cd76264d4206c : SUCCESS in 8m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/20f4c427b9974b6eaf1cb0a1d0cb81c1 : SUCCESS in 21m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7cb48665f70c41e3bc5de2d823eefd42 : SUCCESS in 10m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60c0fd6bfdaf4bf3bc5389fab407abac : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d2e11d78edb4052b7e1ae3e023d0eac : SUCCESS in 21m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/25ed23c5a7db4c7faaaf77f09d510e8d : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1b7f55a17c7e491d9d1a6be5bcc450a2 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43a425e5a5e24b4fb33ee81ada7f5bae : SUCCESS in 6m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a55f5afcbe4945eab4102e30e738b958 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3c7108026af4b14ad9146eac1d4f518 : SUCCESS in 1h 04m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d2e26dd05c8440caffbb6c051a92ab3 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":2},{"id":"47f2385fec424c5000443f93f7e7d0370d9ada58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:03:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"04d28cca79eea58a3a35ebf8f81b1c6be9c2f2fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:07:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121f5cc7a6a14b2e9efa80d26bbff0c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99ad30941294895a9bff34ab5d3d3a0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2bbbdb9991b69052c80d83be7ec21985eb01bafd","revisions":{"dbbeb788d8f5c41db5514c11fa7eac893f5af4ea":{"kind":"REWORK","_number":1,"created":"2023-03-27 14:58:05.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/90/878690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/878690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/878690/1"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-03-27 14:51:15.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-03-27 14:51:15.000000000","tz":-240},"subject":"Add cross-project job for mypy","message":"Add cross-project job for mypy\n\nThe upper-constraint for mypy was raised recently via generate-\nconstraints by change Ifaf60a747200 without there being a test\nthat ensures such a raise doesn\u0027t break anything.  We can use\nthe cinder mypy job for this purpose, which has been a voting\njob in cinder since change Ib6c8a7a1a8eb (July 2022).\n\nChange-Id: I4d68d8abb48b4179dfd5e838837b2406fd90913b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbbeb788d8f5c41db5514c11fa7eac893f5af4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbbeb788d8f5c41db5514c11fa7eac893f5af4ea"}]},"branch":"refs/heads/master"},"2bbbdb9991b69052c80d83be7ec21985eb01bafd":{"kind":"REWORK","_number":2,"created":"2023-03-27 17:03:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/90/878690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/878690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/878690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/878690/2"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-03-27 14:51:15.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-03-27 17:03:37.000000000","tz":-240},"subject":"Add cross-project job for mypy","message":"Add cross-project job for mypy\n\nThe upper-constraint for mypy was raised recently via generate-\nconstraints by change Ifaf60a747200 without there being a test\nthat ensures such a raise doesn\u0027t break anything.  We can use\nthe cinder mypy job for this purpose, which has been a voting\njob in cinder since change Ib6c8a7a1a8eb (July 2022).\n\nChange-Id: I4d68d8abb48b4179dfd5e838837b2406fd90913b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bbbdb9991b69052c80d83be7ec21985eb01bafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bbbdb9991b69052c80d83be7ec21985eb01bafd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
