)]}'
{"id":"openstack%2Frequirements~879044","triplet_id":"openstack%2Frequirements~master~Ie2649e26a8fc8a635e0276a6690967f8e65f853d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie2649e26a8fc8a635e0276a6690967f8e65f853d","subject":"update constraint for validations-libs to new release 1.9.0","status":"ABANDONED","created":"2023-03-30 12:51:40.000000000","updated":"2023-09-07 12:50:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6efb06a0087783da0d0870f4bbdb262dd349199d","_number":879044,"virtual_id_number":879044,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-30 14:25:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2023-03-31 14:10:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-03-30 13:49:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 12:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-30 13:49:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-03-30 14:25:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 14:10:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"69ead77b06825fabd65ef42f4bea3a2b2c4b6a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-03-30 12:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a9f113b3833ea514fcd1e06e91c80e4397b402a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:57:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24b2a1b0ca94459ab95f6c6ae7be70c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f18ee54e8c4d490ba379ee81ba0ec236 : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d469305aa94e05fc912bd38629f031230c956183","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-30 13:49:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4a2e1b084cf6693ac87c208c643bd2a75d77727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:25:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9de8efb0d2eb4208bcabd64b4f6146a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e409f38e194039b52954a9f2ba4c69 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a82d360f11245ebbcb7cf02b02b0321 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/717cdacd46d64b5392d690cb272f1d11 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e79857bb7b94d7d9fadd58978f28c9c : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/577d3872b558459fae3e75350b48448e : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/687dff671f314522a0d10132b9792a21 : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03f6a2c51bb94ad394a00d7a43f313f3 : SUCCESS in 11m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4b09c11e911e4d50a6e130249e587ae0 : SUCCESS in 14m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d7aafacecbe34c94b8638dab22f1797d : SUCCESS in 11m 55s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bd063934bdfe44fb934f3c608a81af29 : SUCCESS in 9m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/72986eadcc0643fe9f469a33f477de8f : SUCCESS in 4m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a155d2bf58cd43da82946a9e4ccd2c6a : SUCCESS in 6m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f9c4855370124454824b7d6c2b1cac68 : SUCCESS in 11m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/000b7ab9bb864ad2a6f3bc41a80b5c44 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4c7c7f5d83ad4b7295362103fec5a4be : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/82bfcae6e2a440438e4e4472dec84312 : SUCCESS in 7m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/92663bafa3944b0b9e77fe2e20918487 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88159305e9a94647bb8e83f95a2bcab6 : SUCCESS in 8m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9cded73d75ff45dfa7d551545ffd2c3a : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/29971c96e5304237b31ed66f80ecb32b : SUCCESS in 10m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a313396338e14dd882f01ddbeb22ef9a : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d91c5096c10f4dedb8cf7015d0b63614 : SUCCESS in 9m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4c68a0ed120648b381262a376738ed91 : SUCCESS in 4m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/643240e7b8494d6e97db22d4c60402d0 : SUCCESS in 9m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c0627a0d3a1742dca6bda39915cb8e68 : SUCCESS in 27m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f093dd1c82be4c309182e28d3b1e9582 : SUCCESS in 13m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/46d1e5a0f425464fb28cab948ab19043 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9c977ea11a3440daec66634d278810a : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f6a2096b80f4a94bd1265a80b9e64d4 : SUCCESS in 26m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/44b9d7bd7e434873995597df73ff5052 : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b9be1cded8394a65b9eba732da605953 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0dafcac747bf4591a252544d2294dd35 : SUCCESS in 5m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3071f29609ad4a58916b512cc115f2c7 : FAILURE in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ceb427aa1664c8e9bfeb96326fa2d53 : SUCCESS in 1h 09m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fef5b6c94434f438cc5b728147435a4 : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":1},{"id":"21ab23510c51087ca454555373a26526372d1bfe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-31 14:10:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cf156d5226e2c5719fd4ee22e4e8c235a74b68b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 14:15:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef372aca921480ea2840091cbe3ed3d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1d5b65f5be04806968a0d0b030ab5a9 : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"83f289501c3bf43dd192062db306cf47501b4047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:29:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/010e380d87af49ce9967ef146c1b837b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f746695f87d4298a94c9f6360c77bf3 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be02183a21394175ba290906e3a0cb3c : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd443ec5ccb84a5ca58cde3c3c5ba653 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01d7da327e894c5fafd00085df7a9057 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/354e491272304d5cafd9e3ad26abb61d : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fce808482534073b1925703036d63f1 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/90807c7d791c432ab29853f26abebfc4 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f855c96f796747b48c246c0657e9b889 : SUCCESS in 14m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8b4b84d97c344c808806815f6c502424 : SUCCESS in 8m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8aa683f32c7c40809d141b4c1fa37431 : SUCCESS in 8m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8e551cb151814b94b54a426bfb647a4b : SUCCESS in 5m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/152dc95c030e4a94be3c7333d022db84 : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7b365c32d0104de490dbae8443f7772f : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7ed580a68d6472b8ef902bbbd9d3091 : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/339957c7be2c446088bf6dcee9c6bbc6 : SUCCESS in 4m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/30e004a9cc154705ba0b484cb6a93ba8 : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b1d6c8ae997748eeba4183c83a54c414 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b9f472fb25c4fa6ab6e903a49c748dc : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d5ed534b96e44f80a68dc221f21657fc : FAILURE in 8m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4d45d35e7f344ef6b4f9cc54d5604288 : SUCCESS in 13m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e5c09a07ed1a4785a0f45ce5c4271e1d : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7f6fefdbbe964ea4b81334bcbb7f72fd : SUCCESS in 10m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b9a7341d14a646e4a1aa20f8c22aec1b : SUCCESS in 3m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6bac7a7da4e34c76aad1137203f07120 : SUCCESS in 10m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/89c2e9565ea946df859d906aae0a6dfa : SUCCESS in 23m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a0ece2f1b6ae4f819a76fda1eaf1e5fd : SUCCESS in 12m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/be5f0c798bbb49ff91b282e7492d0c84 : SUCCESS in 11m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c75c937154ff4485889b6fdd96e5381c : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86592ddb794948d19fdc565de0076154 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/361585c59aa641a59afb0899a60a1ac6 : SUCCESS in 24m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/003ea3f5dfa84815919c3118bff0c675 : SUCCESS in 5m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2d885118df4946169f8ae3427752871e : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93d15a0341094eb59c43ae3f5dca0d3f : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e0e540c1b7dd4ff29c86a42b03725a51 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db229f541dc4432d937950ea527d7d01 : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf45a67becbf47b99bf9fd2af2e3f4c8 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6efb06a0087783da0d0870f4bbdb262dd349199d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:50:19.000000000","message":"Abandoned\n\nupdated in https://review.opendev.org/c/openstack/requirements/+/879218","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19e283c18d4302809b9df4b2a8e157b3c9e9c431","revisions":{"19e283c18d4302809b9df4b2a8e157b3c9e9c431":{"kind":"REWORK","_number":1,"created":"2023-03-30 12:51:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/879044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/879044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/879044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/879044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/879044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/879044/1"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-30 12:51:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-30 12:51:39.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.9.0","message":"update constraint for validations-libs to new release 1.9.0\n\nmeta: version: 1.9.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: matbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Commit: matbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Change-Id: I8b61365e2670370bdf0a3d1d46d4d5a6b0df0f4d\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+1: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie2649e26a8fc8a635e0276a6690967f8e65f853d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19e283c18d4302809b9df4b2a8e157b3c9e9c431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19e283c18d4302809b9df4b2a8e157b3c9e9c431"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
